)]}'
{"id":"openstack%2Frequirements~936097","triplet_id":"openstack%2Frequirements~master~Ia86ec40f5f1163b34746e4ef31a6d2e193ae4da0","project":"openstack/requirements","branch":"master","topic":"reqs-blocker","hashtags":[],"change_id":"Ia86ec40f5f1163b34746e4ef31a6d2e193ae4da0","subject":"Bump cryptography to latest version","status":"MERGED","created":"2024-11-24 13:49:19.000000000","updated":"2024-11-25 10:25:51.000000000","submitted":"2024-11-25 10:24:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936097-reqs-blocker","meta_rev_id":"8253795de41a7ae681ec5ce00038c14e3a6de4c9","_number":936097,"virtual_id_number":936097,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-25 10:24:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-11-25 08:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-11-25 08:32:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-24 15:30:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1c699b80aa25ec418e7a65ebe9e5ba3979d496c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-24 13:49:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf6c0e51d7d7b9eda5ee5f584a62ee4f556f7848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 15:30:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5d4c4798e2048a79188477113b31d37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c9ddf36cd243febc5d74ddd1fce83f : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16db663553c849aba6fb953e89478381 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23531472f15b4412bdacee86e92ddaab : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3c0d23e2d914a0399b413052808121d : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16b3f9eb9a8f49ebbf15c4f4dde9303c : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a2c71c845f84db6808ae7af2b717086 : SUCCESS in 3m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/260fd03252214d5a8df1bc8bf9b6d483 : SUCCESS in 14m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f412eb75e98c46679587f2964e131284 : SUCCESS in 6m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2b038527c21944e8ab1909c83a06385d : SUCCESS in 17m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8aca13c3504e4b77b4ee2f5453bfb22c : SUCCESS in 24m 54s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f68d033a0ec24dfaa2d358c05cb230bc : SUCCESS in 5m 46s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d6c865c3c061490cb4529cfc1f7c472f : SUCCESS in 5m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/56715a60cc8e4e898f33c4645a1b4653 : SUCCESS in 3m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/729e651f04ec4b4abd1d80576f4c94e2 : SUCCESS in 11m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/163d027514ee48eab6bf19ad60c89f78 : SUCCESS in 5m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/dfadefd993b24c16b02c1833f58d82e8 : SUCCESS in 6m 41s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1dbaa374f850464ab5790a794b149901 : SUCCESS in 7m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0fa852d0ff7345d7957ee5a8c932b927 : SUCCESS in 15m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e49cfbf2433941b28d05f618345a35a3 : SUCCESS in 13m 18s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c2a87a1842ac4f45955d5bfb84c044c6 : SUCCESS in 9m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/751e47b0c35c43af8aa5703c9fd8dfc3 : SUCCESS in 11m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/30d299c3fa74414d8f43743da8ef5228 : SUCCESS in 3m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c0f0d629727d4287934d538ecb3e93f7 : SUCCESS in 11m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8df89d2af6cb4e3d8eebd4aa4467d787 : SUCCESS in 5m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9f0dffaac3554487bd990d37906b3b0d : SUCCESS in 5m 19s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ad03275c46f64ac7a4e8b912c80fb273 : SUCCESS in 11m 50s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a88ee04c51d743ce8757b38cb3cfdea2 : SUCCESS in 16m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5fa460123ebe42d8a69c6f99ea27328e : SUCCESS in 9m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/acf4b7ca3fe8454b88f00de5ee4ed7f5 : SUCCESS in 12m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/117bffb5ba324932b8865ae9b2f8b9a1 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de3546211d324baca2b49da04f2db2a2 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6437066b90894859b63103147a6972b4 : SUCCESS in 19m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/83abd2a1b846464685ea8e90077ba4f9 : SUCCESS in 4m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/01599c6398964f4f9c556dbb5a37eb4c : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a4aeb7647f248aea0af05463e6f71ba : SUCCESS in 3m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/70f173efa67e4247aabff28d01b1610d : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ded892576ea849ccb4eb0c1d1a489fe3 : SUCCESS in 44m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3244767925546e2a98ed96862a2f177 : SUCCESS in 1h 35m 03s","accounts_in_message":[],"_revision_number":1},{"id":"eabd1fff670cf6060509d868b8b1bcb36595da09","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-25 08:32:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f51c431345e6ad611a97b1401b7497adbf6b745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 08:33:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be1b35deebc1fa2c4d959bf27c500987730b6125","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 10:24:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beb85545fee2459ca69d1d1ff3b170bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/109d50eab05643738af2c7cea770ef43 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aabb3c94afe24c4481226a0a46bf5321 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2962f2fc02854cf399a1b1d4c746e4c8 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0ac3830f4f44ec89d0a69989ec989e4 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc13a155abae44d29cb39957d3a1e23a : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5259f785c3c843e199011291a4da07a4 : SUCCESS in 2m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/081357f6227b45f99e14520a9cc0e871 : SUCCESS in 17m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3800cb98063643f0a0807408f92c49b9 : SUCCESS in 8m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1017e93eccb04ba69bf8ac7701678793 : SUCCESS in 17m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b587f92e9be3476c8c0da5cda314995d : SUCCESS in 22m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/52569a84bb934b2e8a926908c934f5af : SUCCESS in 12m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/becf8b8d98e74fd1b6266fd68b1c3e6c : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/78214af471fd460eb7945034dd4620df : SUCCESS in 4m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/43331c4709054696bf7e52ea1657bbf5 : SUCCESS in 5m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/30c7d6263b4f4237a9769b3e7500fe5d : SUCCESS in 14m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6868e043e20428286c345b62e756a66 : SUCCESS in 5m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d03029f7858d48dfbe51779d96f446fc : SUCCESS in 8m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/118d36dcedb04b9db3694986328265cc : SUCCESS in 8m 26s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5410667f2a9b431eb2e9ed2209d05e89 : SUCCESS in 3m 31s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c43860f81690451a9d66ac049d4a29df : SUCCESS in 13m 33s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/58803a2f7aa4479cbe264af5c087e650 : SUCCESS in 15m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bb699fcb72aa404d80ea7a7e35b5c684 : SUCCESS in 7m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/89092cfee70d4b2293b94189dad2a74e : SUCCESS in 11m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/db6ea5867fa2460cb9b84318862f8f07 : SUCCESS in 3m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/40b479c7d50848d8b1d419aa5abbd422 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c76fd7e5f8e74e05b74db935701b21df : SUCCESS in 24m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e6db732366ad494aaa06a9328936c329 : SUCCESS in 5m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/02e792c777ca4f95af8782a602286b14 : SUCCESS in 3m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c88ad5c7ae146bf825bd037a88b78a8 : SUCCESS in 5m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/50956a3d0c114b2c988d80bbe00f4442 : SUCCESS in 18m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fc8d925a84d4e4184a3fc4f7e43f9ea : SUCCESS in 1h 00m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80fa014ff72b49779a6d31192d9eb343 : SUCCESS in 1h 41m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4ddeabbc969e0f9b37d851f55a76788301cb8ba4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 10:24:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8253795de41a7ae681ec5ce00038c14e3a6de4c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 10:25:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecadb7ed5cd7439c828e134b44a83519\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f70cf08f284e4d229afb1b5b8b14ea09 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83e9bb9726be0b5b8c49ef8947f6aefe5582e0f3","revisions":{"83e9bb9726be0b5b8c49ef8947f6aefe5582e0f3":{"kind":"REWORK","_number":1,"created":"2024-11-24 13:49:19.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/97/936097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/936097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/936097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/936097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/936097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/936097/1"}}},"commit":{"parents":[{"commit":"52d27d85156ae6f0b3b4a4cfa0aec2cf66367f78","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52d27d85156ae6f0b3b4a4cfa0aec2cf66367f78"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-24 13:49:03.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-24 13:49:03.000000000","tz":60},"subject":"Bump cryptography to latest version","message":"Bump cryptography to latest version\n\nChange-Id: Ia86ec40f5f1163b34746e4ef31a6d2e193ae4da0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83e9bb9726be0b5b8c49ef8947f6aefe5582e0f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83e9bb9726be0b5b8c49ef8947f6aefe5582e0f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
