)]}'
{"id":"openstack%2Frequirements~936267","triplet_id":"openstack%2Frequirements~master~Ia88f8b8af84e74cfb87d1e02830fd12717377ed9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia88f8b8af84e74cfb87d1e02830fd12717377ed9","subject":"update constraint for keystoneauth1 to new release 5.9.1","status":"MERGED","created":"2024-11-26 14:35:30.000000000","updated":"2024-12-19 04:03:59.000000000","submitted":"2024-12-19 04:03:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936267-new-release","meta_rev_id":"82e449de999322b5bdd6f65b1e061f8335975e30","_number":936267,"virtual_id_number":936267,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 04:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-12-19 01:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-12-19 01:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 16:34:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-28 13:39:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2024-12-19 01:43:49.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"56b9b8a0a08710d3c639c495018895becb4cc69b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-26 14:35:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f345fba965807057ceab2330c7454d6d741a641c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:34:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc12846184ae4d55880add20ae827436\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1086212d887e40c2a5ba44cf070fc1f1 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d6e3bdd535d485ab88859fcefc91bf5 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adeaa1ce30f44def8ad41673e3ad5b08 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d65cb9086a94f78a6a0c89bf8b179fb : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3962af0ab3bb4ff9bdef806a2c208cf9 : SUCCESS in 2m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/661c7f2f1d1e4ef7a71c3b34dcdd1797 : SUCCESS in 2m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b455f31da2ac4c9bb58047a7b482928d : SUCCESS in 12m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/405d047a9fdd4bf2a8bb0396770d731d : SUCCESS in 8m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/31f66b8cfb484c38be6baea78ebf63f7 : SUCCESS in 17m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/56352311b21248be838c7e984f14f8f4 : SUCCESS in 11m 30s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e7a1ac6cf16d4ed59160b947450f4214 : SUCCESS in 6m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5fe933a7b8024e238b0a7e38b5620778 : SUCCESS in 5m 15s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/137f40f74aba4735b8a8912b6406159a : SUCCESS in 3m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ad732edd77f745ed8713f08dc1e2ed8c : SUCCESS in 13m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/33080b0179be43ad96dc01c37d134d80 : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/58459939d6ab4ebeb884f29bc4f9381f : SUCCESS in 7m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d7d12e90de7a4917953f31bec503cc66 : SUCCESS in 9m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fabcf39cc611462488546bb9517b6266 : FAILURE in 14m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fce0cd54569b4f5496e75dd7d6fe3a82 : SUCCESS in 7m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/01c7a79d82da427a9c139ada80e96017 : SUCCESS in 8m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/080fc55365a44bcfa2619538e2be004d : SUCCESS in 11m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8a1473aecfda47a5b66d157edd1e84d0 : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/dc9510a519e341f0900319a8203fd3b5 : SUCCESS in 9m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fcb9b025103f47b5bf2b94e69a372da5 : SUCCESS in 4m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0585380381044803a199ce639a7fc17b : SUCCESS in 6m 03s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/de2ab9975bbc4f41b86bcfbab5a261d8 : SUCCESS in 9m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/235e7f4f4c57426ab0664a32ba5f14d5 : SUCCESS in 17m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/17b3cf1e58504b189249287f08ac208e : SUCCESS in 7m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7c655bff43b64b3498ce9b36d9f7bbd8 : SUCCESS in 12m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5ef2eadb49824c33ae9282db378396fd : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a31a5af51b8f4fcd9bdc2708bdbfd692 : SUCCESS in 2m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b9a757004044f6ab23669be853f63b3 : SUCCESS in 28m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7079620bcfbb4e12b8253a7386334fc2 : SUCCESS in 6m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bc59ae5352684e448ad0b28413383592 : SUCCESS in 3m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8888ffb4ba9f448a8a039c8276617442 : SUCCESS in 5m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3713c635b7fd4e34a5d779205ce0b8e4 : SUCCESS in 15m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db03c6eba29047279525e1101a5764df : SUCCESS in 40m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/689ea56014234fbfabfa7bca7c85beaf : SUCCESS in 1h 49m 14s","accounts_in_message":[],"_revision_number":1},{"id":"731ad7b0276b6c9b099ece5ab60dfa7e9480a130","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-11-28 13:39:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4ff1ecc258ee7a2e355e556e189e759ff5e256b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-12-18 13:43:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de8b6140ed3004d880c44bad56aab5374a103e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 15:38:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31f26caabfa546ac953f1ab93c20e301\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ce533276c7435685aa23eac5d64ea1 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8859f564c618463aae701c30a32b8672 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fec168fa57164555beff6722a3b45c73 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb99fdd63d664ba092889a767903dbd9 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a3e736851214914be808b171e7cc694 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1e2bb3971f3473e8a8ff24cfc434cdc : SUCCESS in 3m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5e689ec281c64dfcb8dacbe843a9c5ae : SUCCESS in 15m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/181688beb94c48f1a20a3eb4b45d8918 : SUCCESS in 8m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/52c9d19d5fa84f48acfa0cdcf62e8dbe : SUCCESS in 15m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d70bbe9db6ac460e94826cfb044ba668 : SUCCESS in 32m 45s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/554c09bbe4f244aa964a5aeeb3ad5f4f : SUCCESS in 5m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c3f3650060f74eb5babc4a1d45faaa6a : SUCCESS in 3m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/56116bfd789642fa9c3054710efe5ef7 : SUCCESS in 3m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/458328e7c53747cb84d6cee09013181a : SUCCESS in 25m 37s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f4b3c17c52404b7f981169ceb30faf79 : SUCCESS in 4m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bef43130b5d04eb9acf186f0694f62f9 : SUCCESS in 4m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e52fd20306a244799eaac637923cafed : SUCCESS in 7m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/01b1bd96679440fb8ff3da20085b9e13 : SUCCESS in 13m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d13707ea147476897c55cedad712514 : SUCCESS in 9m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/497597eeb4d34c279ecfcf456a487100 : SUCCESS in 7m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8553efaba49849ddba7308e74810361e : SUCCESS in 11m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/24b33a69a6cd439cba292a79d53ee45a : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/945535f1af28417fbd8c489368f02d33 : SUCCESS in 8m 41s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2d610fb6f25c4acda2ab606dce8a56df : SUCCESS in 3m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/db60817014354a9986a5fa6fae576bee : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4ec3fe0919514494a928a8c215118512 : SUCCESS in 10m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fcccbaf5501343bdba92b5b97899b7cb : SUCCESS in 19m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9181819b9f3a4eb1aa60cb86330345fb : SUCCESS in 9m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fb2e39e42c514595ad56fd8ab7bd42d6 : SUCCESS in 15m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/dc8665996c9b4c4191d09f7c7dfe8e76 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/feb1846424ef4f998a9d329b41c63fe8 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a0b586e8cc664e79b3e996f88b4c4a45 : SUCCESS in 25m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9b6095f57b3342529a30fef78592e1c0 : SUCCESS in 5m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9a15e84602c84d8698061d058e323c6e : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/878f1cba67c84c078f7e66acd13da437 : SUCCESS in 5m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5a16dc22687a47cab8cdd6ed4751d778 : SUCCESS in 13m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/142f6ec77601426ea453f786ea23e7fc : SUCCESS in 1h 09m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f8aafe581704eae8a9359da188384c3 : SUCCESS in 1h 43m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b1e878968f9487430935546c186883c0d9284abf","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-12-19 01:43:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea62d05e73bc46e3c8edca22ef4fdea93da747a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 01:44:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ca7aea948f0e2cba44a05db2b0aa7c1d36a8472","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 04:03:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a1cc30d2c57416595f369ed7ed6d07d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c8e3485af148738985d8ba42b60905 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc4dc7431f2f4cdf98cbf1c2047dfc8f : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f58d1487e00a4b6dba7a0f4c99588db7 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2faa46d0d7934000ab1446f11db5f7db : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03de8feebfe141729475402d9369bd60 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b3b1ef49b9b4d5aa525a75f77a02e67 : SUCCESS in 2m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d6cbefa084c946e2bbbce9e553b9c14d : SUCCESS in 12m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d3f221676fd44f72a8ce41b5ce8593d1 : SUCCESS in 7m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c6dfa70beac941999403b5322f7a5060 : SUCCESS in 11m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3061697e92e2434daba46242193b601e : SUCCESS in 16m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/af0c9b649f874f1bb06f5e175802c1ff : SUCCESS in 9m 26s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/169f4a2913c64e15b6b8ba2365ecf70a : SUCCESS in 5m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/923bb451c9a746d19c918717da12e4d1 : SUCCESS in 5m 11s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/43be4e4fa038496085eb60e4a0ee629d : SUCCESS in 7m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/80e9a0e1a57f4dc992b633ae470e808a : SUCCESS in 15m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c31cbc45c5549779244318c333c80f4 : SUCCESS in 6m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0300b8ad78834ab688936a8e99a1cf2e : SUCCESS in 7m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a512029e3ca3462daef94c264a37b457 : SUCCESS in 13m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/efbc42623ad34324b1141cfa63306811 : SUCCESS in 1m 58s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/250b1ddcbdc24c7cb8e514c77669d3de : SUCCESS in 8m 53s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3f2e96a5e97547439966b749f3f425d2 : SUCCESS in 16m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5dc628a4a82b454c8f01b91c641e951b : SUCCESS in 7m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/55ade47996034649a73c85a14c181a68 : SUCCESS in 9m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4597a89222ea4208945a425302a0d657 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/18dee2944ed44c89a6852d425cef5489 : SUCCESS in 2m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08a70f86061c4c018c246529bb403659 : SUCCESS in 26m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d36357945ae04071947792247cf5657e : SUCCESS in 5m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/22ebaff073054bd4872cd765e4e412aa : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e57d05c35abb4cc3a5b8d9d8701580fa : SUCCESS in 5m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a157810f39c44eff994db1f78cdf6df8 : SUCCESS in 17m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b17a67dd80c4941b96e46d340d53a25 : SUCCESS in 51m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c340ea034544210b6cd6c48a5fa72aa : SUCCESS in 2h 12m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cd6935ab500832e4131444fe9f32e08337416922","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 04:03:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"82e449de999322b5bdd6f65b1e061f8335975e30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 04:03:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a84965327ad342109198056f50e4b0aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45977b9ec90444299265a979c7b64c41 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6b937316030ecb728bfa465cd4b948a1bf97579","revisions":{"f6b937316030ecb728bfa465cd4b948a1bf97579":{"kind":"REWORK","_number":1,"created":"2024-11-26 14:35:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/936267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/936267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/936267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/936267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/936267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/936267/1"}}},"commit":{"parents":[{"commit":"f2550a21f3b48c7376989b7a95500f48facf4ff7","subject":"Merge \"Update netaddr to the latest version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2550a21f3b48c7376989b7a95500f48facf4ff7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-26 14:35:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-26 14:35:29.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 5.9.1","message":"update constraint for keystoneauth1 to new release 5.9.1\n\nmeta: version: 5.9.1\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Commit: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Change-Id: Idf854390d49a444d6ffda25e42278c7a8bd761c3\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: David Wilde \u003cdwilde@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia88f8b8af84e74cfb87d1e02830fd12717377ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6b937316030ecb728bfa465cd4b948a1bf97579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6b937316030ecb728bfa465cd4b948a1bf97579"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
