)]}'
{"id":"openstack%2Frequirements~936787","triplet_id":"openstack%2Frequirements~master~I23eeba234e435ed3360ea10919e25d9109631413","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-03 18:09:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23eeba234e435ed3360ea10919e25d9109631413","subject":"Updated from generate-constraints","status":"MERGED","created":"2024-11-30 08:40:02.000000000","updated":"2024-12-03 18:10:49.000000000","submitted":"2024-12-03 18:09:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936787-openstack/requirements/constraints","meta_rev_id":"338a7ec1676d50f71de809e550a7eb29a445c866","_number":936787,"virtual_id_number":936787,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-03 18:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-12-03 15:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-12-03 15:53:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-30 09:33:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-03 15:53:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"86d3f703dfd091f482d6fac731cc066319a9c945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-30 08:40:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1885af1db8fa1728b924e81721d6ca63bfa3bffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 09:33:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/edbb8cb08d7346af95b1bf4caacfd740\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76a73e2cecee483bb5cf47aaada2edae : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b2eecfd36d04e73825c7d0a998f6ac2 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d5cf1ce287344edb998134ff4b8c5a3 : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78df3ed059e44a91b48fb5fb59fe1e43 : SUCCESS in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/615da273e5e5425fac934e4e5518b670 : SUCCESS in 2m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7da151fbcd484dd79ce10b16e251ea48 : SUCCESS in 2m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1153c91a56ba4736897c5810bb659f89 : FAILURE in 3m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1ceb5a4b52f748dba54fa1dcf0198785 : SUCCESS in 10m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/778325dc00f744b8834b6ce3e646c2a7 : SUCCESS in 17m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/003e64f1b9cb43749b73ea9e961bc3a5 : SUCCESS in 24m 17s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c5ceb587fe3d4da09e699b36606fb771 : SUCCESS in 5m 17s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ae65312bd2c54106a405af9fe31a5624 : SUCCESS in 5m 11s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8ddd4d416d3d4d28b70ed50a165ad3b7 : SUCCESS in 4m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a2fa95b5d8814cebb5c265973d4f05ff : SUCCESS in 14m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e4e14c65bc6d4cb985ea9f8fb1f67062 : SUCCESS in 6m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2af47bb76b9e45dcaaac6da78f3696b8 : SUCCESS in 8m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/dc0bd2a7b0234699a5798e546803fedf : TIMED_OUT in 41m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/10d84330869e4d898dd4b96f5db42bb9 : FAILURE in 8m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3621f781f2264b839f32c99093775d8b : SUCCESS in 9m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e61174e12a1044568a2c904ef908840f : FAILURE in 3m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/12c8644c7c6d4068bd567dcadc3996d6 : SUCCESS in 14m 44s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8151f835d68347b6a55050871d8c63fb : SUCCESS in 2m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a06e29b94a1946d98da05d867cb0d227 : SUCCESS in 10m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/94c7a22e898e4d43a399870fb30e2bc5 : SUCCESS in 4m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3f2295e9faef42b699adddef74b48039 : FAILURE in 7m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ef83a8f3cb0e478596b34a779348c014 : FAILURE in 12m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/60e40cf967ef4493a351ad2db7b4e244 : SUCCESS in 17m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/28acd9a3c38a469aa39a4020ebf7ca83 : SUCCESS in 5m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/033c7aa1b1264304bb5d2d4801f96e90 : FAILURE in 13m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/92a04a3ac3b24dec934b9491637fe03b : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ecbc03c8ea34403c968c3952daabcc56 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/576e6ade4c6648daa2afee07dc71ddd3 : FAILURE in 21m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5b3a2ac3d9dc4058a4e67df4a738f815 : FAILURE in 5m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/34c2054bc90644489eedcd16042e05b8 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33d7fc4728744af4aab2674bc310878a : SUCCESS in 5m 34s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8022b1b90d00414798866ff23584be80 : SUCCESS in 13m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9e7edf3585a4f18ae298208bc80e2b0 : SUCCESS in 52m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e42cfc3d4d04b5d99efacd483bc5860 : FAILURE in 29m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4de5610527788146ea34fecf8e41c01e1c3f350e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-03 13:05:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"49695ab15fe34ec1f701d89275aa197af007e252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 14:46:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fad769909a79433b878fe28b3cbdfcf6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/940227743e2c4593beea986e1b4fd835 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5737d3af014cd3bd4500f96d79a201 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb1b5981324d4f34b12083e81e3faa5f : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f9e3df61c2647338a9aec11bc9189bd : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60fc9a934e494ec48d5a689207b13dc6 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7adcccc96b104fe19004cdc6ce418aa2 : SUCCESS in 4m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f73b6c05950d4785842f443c7f135fc8 : SUCCESS in 17m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/762ddeff424e4c8da85ee25abcc32c3a : SUCCESS in 9m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e3c9e12fc0e94c029e4b424013871db5 : SUCCESS in 18m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ed7821f9c6664ed3a204eb20249822fa : SUCCESS in 22m 26s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3f843d71c1df496b9ce1e2691232d0e9 : SUCCESS in 5m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/05421887e6104df3ac5e6c4bc482d124 : SUCCESS in 5m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c8b9ab3f309d425da30da0c01948ad6f : SUCCESS in 4m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/696cf3994b1648f38a3d1b884d2e0a23 : SUCCESS in 11m 22s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/78531b0da450438b968969c77b964a27 : SUCCESS in 4m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/eeac353316604671a7cd172b885be938 : SUCCESS in 5m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/51a51aa45e1e4c4a8fb42622ac3887e8 : SUCCESS in 7m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/093be8415791413790aefd4c59f5846e : SUCCESS in 14m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/986da3c752fb495aab16285b851aa538 : SUCCESS in 6m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c51b92a4a7a342749db0f3a004519d32 : SUCCESS in 8m 28s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/13e2d2eec0b4402c941e17425b480bfa : SUCCESS in 11m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dcd2b22cb1e6429788fe1e1f1b05d322 : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c6e805517a5f425b88e411ca4d611e30 : SUCCESS in 9m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2bbc595d78614e7e9b2931f766629e26 : SUCCESS in 4m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a9535f02024548d28cd71ad7a2564ac7 : SUCCESS in 6m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9eba7bf4da5541b288abdba82bcac8e0 : SUCCESS in 12m 10s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/04d0ad92c4874ae580c8f99279c64c0b : SUCCESS in 20m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8a35bf95a52b48d489a436fc40b21af5 : SUCCESS in 9m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5a5fedfb999d4522885fd12df818371f : SUCCESS in 12m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c94d21b86c0f45bdad0b23cf42e57675 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/72729d808fc84dada6e2bfa56b2d61a7 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67a3987f225a40038242e3d264051ce1 : SUCCESS in 24m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/430e4104a2074d1cb192d5e84474b28b : SUCCESS in 5m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/50a6ee92c15643f6b72be7ad25bd1e5b : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/225ac3564c914c9f8eb2c1c7dad7a986 : SUCCESS in 6m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/70a64c7ccac24c139e726386d6bd91ea : SUCCESS in 13m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/464385edd0b9448d8369f14b03d438bd : SUCCESS in 1h 02m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8723952de69e481981f4836288721e2b : SUCCESS in 1h 35m 34s","accounts_in_message":[],"_revision_number":2},{"id":"40a39f06438d79bcb3bb6c7b1108fb93ed336ead","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-12-03 15:53:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a88a58fae3f5c11c9d930d1930d5bac57cc92c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 15:53:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7ddb4292ae0321269596c6287a926a5e1ad8ee75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:09:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03791873be294bd6b56f69fd791f78c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a46a8d2ea724c8bb77574ab5cd76ade : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22627a1c1f60409288b78d291c953bea : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d453f17c24340b4b5a575e6c2e7bf62 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ebdc7366d4c945ac8dd91fa5b6dcb430 : SUCCESS in 2m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae3dcf57bd114e12b36628ada788acd5 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aad7d7e2a32a4627b8ca7998afe32729 : SUCCESS in 2m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0f316c8b945b4e2bb99627e2ba38dd2e : SUCCESS in 15m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0d0b6f9ececf4fa08142227378ca97d7 : SUCCESS in 15m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7849952e1154431dbd68393f0ec35f39 : SUCCESS in 17m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/221d8caa21044ea8b822ff8479a0e1fc : SUCCESS in 28m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a96016b3ea7c4a208239447d53a0c9c4 : SUCCESS in 10m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/af346d97ee5e4d91ad94a0780bdeaaa0 : SUCCESS in 6m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5fc7d7b8dc44477b92105bd2dd2d7bf9 : SUCCESS in 5m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1327986f1971462aac70a9814933f584 : SUCCESS in 9m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/998a6601afa54f60a3a09bd7f08dc68d : SUCCESS in 12m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff02293a4d8d4ea3be47f67231227664 : SUCCESS in 5m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f72f8c9592b94624a2457598432c4f82 : SUCCESS in 6m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/425443722cd74950b59e3f2dde08f7b4 : SUCCESS in 13m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d1771e1ca963458484d4246c1eb6db21 : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d36b0de10d6d40728459667ca73a1754 : SUCCESS in 12m 06s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/520e17deea0f4840aa2d6294ce2c229d : SUCCESS in 20m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9fa580f410554ef3b1cc9cfd342ea621 : SUCCESS in 6m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3493c11b8f284c64a9fecbfa56009c07 : SUCCESS in 12m 10s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7b6f24ab0d754058bdcd02d0ede59061 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e36cdbe8cd64eca88cac0439818fc80 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbfe4a5adec04785a2de3fb2df5f6ba5 : SUCCESS in 26m 59s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/353063d48d38438c91074c152c5500d2 : SUCCESS in 4m 05s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/17a18d84e65e4e4092411f9bce00045b : SUCCESS in 3m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a24ee90c4fbb4d95bf8ffbcaf10828ab : SUCCESS in 3m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2ea276f6614945df898cb831d16a0dde : SUCCESS in 7m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23ef824471f346d1b87404695a5d2351 : SUCCESS in 1h 04m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d42fedc03494cacaf9733fde96f76a2 : SUCCESS in 1h 57m 24s","accounts_in_message":[],"_revision_number":2},{"id":"fe2394eb991bb8dde8c01d346e28f2cb56971d9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:09:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"338a7ec1676d50f71de809e550a7eb29a445c866","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:10:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e69bba2e0dfc4b4bb3bc4d209f647cac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33bba611a10f4f99a32998ed92a261a4 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f66315e932383ce1a9426ee41f241ca90444184b","revisions":{"01063b040c8cc9f4778ce974125846ac158338f2":{"kind":"REWORK","_number":1,"created":"2024-11-30 08:40:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/936787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/936787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/936787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/936787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/936787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/936787/1"}}},"commit":{"parents":[{"commit":"0831e3910a289564309cc9cded69b9a7c24e2cbf","subject":"Merge \"Move requirement-check job to ubuntu noble\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0831e3910a289564309cc9cded69b9a7c24e2cbf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-30 08:40:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-30 08:40:00.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I23eeba234e435ed3360ea10919e25d9109631413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01063b040c8cc9f4778ce974125846ac158338f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01063b040c8cc9f4778ce974125846ac158338f2"}]},"branch":"refs/heads/master"},"f66315e932383ce1a9426ee41f241ca90444184b":{"kind":"REWORK","_number":2,"created":"2024-12-03 13:05:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/87/936787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/936787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/936787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/936787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/936787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/936787/2"}}},"commit":{"parents":[{"commit":"0831e3910a289564309cc9cded69b9a7c24e2cbf","subject":"Merge \"Move requirement-check job to ubuntu noble\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0831e3910a289564309cc9cded69b9a7c24e2cbf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-30 08:40:00.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-12-03 13:02:54.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I23eeba234e435ed3360ea10919e25d9109631413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f66315e932383ce1a9426ee41f241ca90444184b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f66315e932383ce1a9426ee41f241ca90444184b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
