)]}'
{"id":"openstack%2Frequirements~938902","triplet_id":"openstack%2Frequirements~master~I7db135055f3e989c2af33b5f477439b5998941cc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7db135055f3e989c2af33b5f477439b5998941cc","subject":"update constraint for python-blazarclient to new release 4.2.0","status":"MERGED","created":"2025-01-10 14:22:52.000000000","updated":"2025-01-10 19:42:29.000000000","submitted":"2025-01-10 19:42:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938902-new-release","meta_rev_id":"115e623c5676abaf53d01067f3991e12c00ea604","_number":938902,"virtual_id_number":938902,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 19:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-01-10 15:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-01-10 17:50:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-01-10 17:50:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 15:56:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-01-10 16:31:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 17:50:04.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5475c209b71282256624a642376b59b112c15a9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-01-10 14:22:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9eedcaf7cc5efcf5e07b4ffe2cc2db17a5674f4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-01-10 15:56:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a3cd3855341ff72b0a6bb65243ef086fec3bf963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 16:31:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c400b530e934267a424eaf78ff755b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afdb3855fa5e4daa985cbf17ce617339 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042ea7d7972a496a84351ef25d55a350 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd02a260da324ca69271a8af3fa6149b : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55608ceeb5ee4754a6e8602dba4234fa : SUCCESS in 2m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c97b4fddbe04d00b95a82eef2e9cf7c : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/807c24d6521845c3989d3925c8465dc6 : SUCCESS in 4m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/09d3fead112348dd96316c748dc6f37b : SUCCESS in 15m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1d807262d34c4feda2f40b910be62502 : SUCCESS in 8m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3cdfd0696cf24c9e8d81187060d37c4c : SUCCESS in 16m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/51959c7368fd4142b9050a3cb2fb9937 : SUCCESS in 28m 55s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/11a2423154fd4922bcfddf4df8163557 : SUCCESS in 5m 29s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c0395fff39e0458a91de7b4ed098f8b0 : SUCCESS in 5m 33s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/72cb4bb7ffa04874bbbe1a9e00f9c4db : SUCCESS in 5m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/97a42f5921db4db89e5bd33ac2c15e6d : SUCCESS in 12m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0ed64c105e7d4e9aaf1e3d43e002921c : SUCCESS in 6m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/93d3530d980146499d8fa960b4ae7b1f : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/213a0df785134fd58473f4fde2713e62 : SUCCESS in 7m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2743dde1ccdc4311b2a240e7c2731ae1 : SUCCESS in 14m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3e2b56cd4dc4da69e04c85e6ecaa99a : SUCCESS in 6m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9013332f78284d8a9eb7606df778e954 : SUCCESS in 9m 24s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b0175c6253b7404a8b5692ee0ba1dd50 : SUCCESS in 11m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5f7879ab15ca4c07bb605182a9faa42e : SUCCESS in 3m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f6e50cd3799946cc9b90561cb417ffd5 : SUCCESS in 9m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6f8b95e0cddb49aba44eec79fef55bb8 : SUCCESS in 3m 37s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/518cc6d658b64125bdc8a1ba3a758d96 : SUCCESS in 3m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a6b3f4f606874cb9bcb72a18a8957e35 : SUCCESS in 11m 33s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b6cccde22d8b4f2f81e62cf0fdb3bfb5 : SUCCESS in 18m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b41382d30c224810a863ac0af1f820d3 : SUCCESS in 7m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ab1c83e642e14bd284808d1f56344483 : SUCCESS in 17m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a0c6a9099c8d4db5931439569f84b291 : SUCCESS in 6m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fb200d0ea1eb4ea98c06a310cd261910 : SUCCESS in 2m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eccce12645134998b1ff688b8c194e6b : SUCCESS in 28m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/97d2a099bb1641b29515e14a9f03c809 : SUCCESS in 6m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ef77ee89e7ca45bb93a5d2c417c7fd1d : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e9cc39f0bcec45808ab764ab3721c8b5 : SUCCESS in 5m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ffd6b5f8153549c6a0f561356548a343 : SUCCESS in 17m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c104d66b7744fe0992da60892533bf8 : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5875ee5895144302bcf38a64cdb363bb : SUCCESS in 1h 35m 31s","accounts_in_message":[],"_revision_number":1},{"id":"92f46261fc78045ec322b49c58a048ba3673ca5a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-10 17:50:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9b4f9d7e0d8f2fa8138d98ab9e91a7c24ba41d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:51:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5d553c7a07e6eaed5f9f81613c60233746b65d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 19:42:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04815047ae214fc488e99dfda9cbe383\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e834cbfd47b470a93080b00bc79f920 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ced7719bab5446b9f3ed2e9c2e44528 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1987b60625f4d6e8edea847b0b02379 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b519f0b7a0a1493a930720e9454466f9 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14dd4d4ed3564cf39c5869b3fa96af2a : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8818a396baee40d99899fceb39ea0b18 : SUCCESS in 4m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/067fb039595a462c924217d9120c1be8 : SUCCESS in 17m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4211ac0f03c9473ebf9879bcf2e36eb9 : SUCCESS in 8m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0436eb9c41c94094821e13cbdcf63f37 : SUCCESS in 14m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0dfb24a1eb28409198a33a84f62a3b73 : SUCCESS in 21m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/44cf6644960643b7884e7922b542a18c : SUCCESS in 12m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/913d5588055f4d7da5783439b4831ecb : SUCCESS in 6m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fb8ef3ef8b9849f6920cf5805bfc1396 : SUCCESS in 5m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c20aa48c601f474891cdad8db520faac : SUCCESS in 8m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/07a32896f69e4e58ba29fa20be495dec : SUCCESS in 13m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/750178bbdd204394a6ec306899fd2b18 : SUCCESS in 5m 45s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5ba13dbc4cac4ca29d182412945df907 : SUCCESS in 8m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8dc77062b41249538a362131e2247c89 : SUCCESS in 8m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/477b63ef4d5c4db5a474663c1ac376b0 : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bddb2a8018b848108fede3432f26ea19 : SUCCESS in 12m 16s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9172693197654404b0b7a98128b2ee43 : SUCCESS in 20m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5ef09f90b0c84d4aa57ff69dfb74131c : SUCCESS in 8m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8b91d5ce507b436ebd7f993c01848a62 : SUCCESS in 13m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0ed5b457d5db414da1ca76f34629ad3e : SUCCESS in 3m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c29f9732cab43758c5b2c2af96387e3 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a22eb40f8f1a4183bf5b0345efdfa83e : SUCCESS in 25m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0d48eee2633548d2a0e3e0c28e957d45 : SUCCESS in 5m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8cad1b5b824c4e7696217ca3eb35ad61 : SUCCESS in 3m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e717591493a84dc7938f5c6949087ad4 : SUCCESS in 5m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/afa251caffbb4cc4bf3b54caae913367 : SUCCESS in 16m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aec7148b2de7496ba87dd587072c2a05 : SUCCESS in 1h 07m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eeb963a1cd2b44408c3c98799d935c1b : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":1},{"id":"115e623c5676abaf53d01067f3991e12c00ea604","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 19:42:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be0061618d9a1861351dd81a3ae757deccf08ab0","revisions":{"be0061618d9a1861351dd81a3ae757deccf08ab0":{"kind":"REWORK","_number":1,"created":"2025-01-10 14:22:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/938902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/938902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/938902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/938902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/938902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/938902/1"}}},"commit":{"parents":[{"commit":"cacce0d843145c3d3209cb12db7038deff605e37","subject":"Merge \"update constraint for oslo.messaging to new release 16.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cacce0d843145c3d3209cb12db7038deff605e37"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-01-10 14:22:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-01-10 14:22:50.000000000","tz":0},"subject":"update constraint for python-blazarclient to new release 4.2.0","message":"update constraint for python-blazarclient to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: team: blazar\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Change-Id: I6eed803b6c31d222c317945966071b7be45eb2e5\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7db135055f3e989c2af33b5f477439b5998941cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be0061618d9a1861351dd81a3ae757deccf08ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be0061618d9a1861351dd81a3ae757deccf08ab0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
