)]}'
{"id":"openstack%2Frequirements~939293","triplet_id":"openstack%2Frequirements~master~I89ab88ac63679a6f7737a8e04a55a66c42344c0a","project":"openstack/requirements","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-01-15 16:23:19.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-01-15 14:43:50.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-01-15 16:23:19.000000000","reason":"Change was abandoned"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2025-01-15 16:23:19.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-01-15 16:23:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I89ab88ac63679a6f7737a8e04a55a66c42344c0a","subject":"Revert \"Bump eventlet to latest version\"","status":"ABANDONED","created":"2025-01-14 23:00:54.000000000","updated":"2025-01-15 16:23:19.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"revert_of":933257,"meta_rev_id":"7e11417fc865f8a165557cb98865a359004211cc","_number":939293,"virtual_id_number":939293,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-15 07:43:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-01-15 06:56:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-01-15 05:43:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":36603,"name":"Itamar Turner-Trauring","display_name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","username":"itamarst2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":36603,"name":"Itamar Turner-Trauring","display_name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","username":"itamarst2"},"state":"CC"},{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-01-14 23:00:54.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-15 05:43:10.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"de7c0d93e52ba87883699b8456d30366235ef9c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-14 23:00:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5ec8ae768726947a267d70500bd2fa5b0a3e24c","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-14 23:01:21.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5957deae9d0783a439a0c96be3983f1b860d8254","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-14 23:07:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1c825d9dcf755e8f11d53913efde98198a9112e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 00:38:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/749a86a53fcd49a38949fbda39e49d48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eec36f6740047268aaf5dd781f7d074 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f71c71d82e14cc9aa2c1b1b6af17d0e : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c79c9d60d77b4da29e60c7bb0fb9b474 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7913274dace8409ba9882382bfdd9557 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48f968a7ea0a4085a216626c3adbb22b : SUCCESS in 2m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae6f069b4a794725b699363a9028c7fd : SUCCESS in 3m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f8c70a2498184e489975da940190234c : SUCCESS in 16m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/21ab1d4f5f4d4ef1aab5cf5c549ae834 : SUCCESS in 8m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/42fc84adcb6d487fafc9c58ff54c374c : SUCCESS in 8m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/866c9d13deaf4d6eb9d0d5db8bc001ff : SUCCESS in 19m 10s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0682410d4aaa45d2b31a6f329d6eeabf : SUCCESS in 5m 17s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8b2628a1f6064d4daefa436deb029c00 : SUCCESS in 5m 36s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/66b80996329841d6a718d4ba8d921aa4 : SUCCESS in 4m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f575930ccf8949ae9413fa67a0d6b06b : FAILURE in 24m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a009ff560f5c4b698b7b267aa40f5e59 : SUCCESS in 6m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ad9a2e7049eb41f6a346d4c66929a5e1 : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2d91e01cbac241c4a5f427e9a3da29a3 : SUCCESS in 7m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5433f6b17cc44f6c83ba3ddfe76b4f6f : SUCCESS in 6m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dc8002c2fb894e89882ad6e0708ffb1e : SUCCESS in 7m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4216bb1567814506badd2ee2d3fd4d3b : SUCCESS in 8m 22s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/68ab1281d82e435ebc0c7d4ed5c5c960 : SUCCESS in 13m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/93de548f42dc4bbd86e19677a50a397e : SUCCESS in 2m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9a5425c4eaa5442ab756b9fcbf33baac : SUCCESS in 9m 20s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/de3e707259084a41af24b6b504715665 : SUCCESS in 5m 01s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/474f9742b4a2414ea3b9901975b38ae1 : SUCCESS in 6m 20s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4435cccbb1d94086af75982f514a5020 : SUCCESS in 12m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/068d99c4c3514d18b35d8dd81720d4e2 : SUCCESS in 18m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/58d1bec6285540d2ae71b7d091caadb4 : SUCCESS in 10m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e135dfe7d4bd4d649a6685d5ed9b7127 : SUCCESS in 13m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/55c44b0b8c1f4b439b948037c402cb1a : SUCCESS in 6m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9670a20bbb5e43eab3262d6fafa22284 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ae1ff69e88a4f48a70d4f505d19bf11 : SUCCESS in 23m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9cf053ab702f4bf58f387cadd5f677ae : SUCCESS in 8m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/31930d55b2e44d849c8ca14f530ec579 : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b1bd59a54ab49eab722e3f00c366f57 : SUCCESS in 5m 40s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/948e90f21199408daeeb434288375d38 : SUCCESS in 15m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b792f9723843404081ed32bc3c5c574f : SUCCESS in 1h 01m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a60cccb68bd440e9dd2ab42cdd2b394 : SUCCESS in 1h 35m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e754bb3b3eebdd6d23e3a72c8ceaf56611d6faa1","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-15 04:15:39.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"7b619b01550101d547770a00d0cf797bacd813cf","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-15 04:21:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75c9c0475268165e05009a1261166b5d8ea2f338","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-01-15 05:35:20.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1d26e8086f839b0fdf1657943e73ab9262f3d187","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-01-15 05:43:10.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cceda55e77ba14f6105fe900ee40b9c306e6a7a4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-15 06:56:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"749d856d6ad4cfbb6278638c63072dc6a957b62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 07:43:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db619a9206f443d4a957b176e8870a80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbb2a19bfe1e4843b9ecf3c5b8841c79 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a1c8e0b6feb4332867ebbe5feaf20fb : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1de5e280990c407799195176fa7b666d : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9007ecb29a71439f8ad8e3aa22c6d696 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab4c797d498a4a52a8fa3e62a6a3c132 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/217ffef65868452b9077ccd35ed98081 : SUCCESS in 2m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/49b8df71e19149f4b40030f6e6984af4 : SUCCESS in 8m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/26901e914c234ff684faefd9cb6b2ead : SUCCESS in 8m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ffe59f4429244689b60afeb091c75cc6 : SUCCESS in 16m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/25fa4f6f6e034ad3a42f7128eeaca501 : SUCCESS in 15m 21s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f2a681f6f67a4ea4ba5ed0a91014189f : SUCCESS in 4m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d714f1fd0f8548a6a77b1cad255d901e : SUCCESS in 4m 38s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8a694d2caea7498e9707633d744f4415 : SUCCESS in 3m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ee71fae186f240ee9c743133077bae42 : SUCCESS in 9m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d1211e696d0942fd95623effa6c2ca9e : SUCCESS in 5m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8e84e91529364f3fae7b7e03c2ee531b : SUCCESS in 4m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b2523bd09df64083a21a107b7b8b7585 : SUCCESS in 6m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/49d78001dcc641458733aeca656114a2 : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a58dda1cde043a28509793d10f4802e : SUCCESS in 5m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/155a3a0dd5c748879f8336bbde943a65 : SUCCESS in 7m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6a4217344a464cc5810ef165d66f7e16 : SUCCESS in 10m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/599d1eee584943ec9c70af10195e8e40 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a6f04e9e6e314fdd8d1b6ac1735d3e4b : SUCCESS in 8m 59s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/571bd6444d954396b0c047371cf8ad6b : SUCCESS in 4m 22s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/263b886c52bd4994b8309f4995048758 : SUCCESS in 3m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5cc5939a3dd2445ab40131ca1243254a : SUCCESS in 9m 24s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/eeb8248d1d0e40019500fef26801ab42 : SUCCESS in 14m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c70b0a7931124e6882efcaedc2dc902e : SUCCESS in 7m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/aa90121181814a2a9838792fa972ab16 : SUCCESS in 10m 34s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bf83ede347444238b6b8b9f401ac356a : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/217c66b5007744cea5fdc86d50ec30a0 : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/64509d2382474f9db69048e74b47e93d : SUCCESS in 30m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f46c17fecc8e40fb99aaec09ce9d40fb : SUCCESS in 4m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/88559ffef22544008b46fe639920d3f9 : SUCCESS in 7m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/492eea483b4b4d7198056e9f0d99172a : SUCCESS in 5m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d947ec99f3eb4b75a86aac9fbf556175 : SUCCESS in 17m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12896526207f4aeda39d7e6a4580fb6d : SUCCESS in 1h 06m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70a2c24b4242455bb02eefc37f4fac4b : SUCCESS in 1h 55m 14s","accounts_in_message":[],"_revision_number":2},{"id":"a3bd8ab1988cb35a3ea45ca503ecbd3ca15fa13c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-15 08:26:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b26003672dce32c161979209ad9673a782745bb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-01-15 09:29:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc3e5d0a5e5c55a41b132f979817869a1919bd11","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-01-15 14:43:50.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7e11417fc865f8a165557cb98865a359004211cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-15 16:23:19.000000000","message":"Abandoned\n\nTo be resolved by switching affected rally job to wsgi","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8915eaa5cad7f341a33e7710f69620b282bb3595","revisions":{"2f540733f045bca9d017cddc9f06744e4aee2387":{"kind":"REWORK","_number":1,"created":"2025-01-14 23:00:54.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/93/939293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/939293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/939293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/939293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/939293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/939293/1"}}},"commit":{"parents":[{"commit":"dc7eafa720146f0ff1bacfdc01d8862a7aaabd81","subject":"Bump eventlet to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7eafa720146f0ff1bacfdc01d8862a7aaabd81"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2025-01-14 23:00:54.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-01-14 23:00:54.000000000","tz":0},"subject":"Revert \"Bump eventlet to latest version\"","message":"Revert \"Bump eventlet to latest version\"\n\nThis reverts commit dc7eafa720146f0ff1bacfdc01d8862a7aaabd81.\n\nReason for revert: Seeing if it resolves issues in neutron gate\n\nChange-Id: I89ab88ac63679a6f7737a8e04a55a66c42344c0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f540733f045bca9d017cddc9f06744e4aee2387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f540733f045bca9d017cddc9f06744e4aee2387"}]},"branch":"refs/heads/master"},"8915eaa5cad7f341a33e7710f69620b282bb3595":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-15 05:35:20.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/93/939293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/939293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/939293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/939293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/939293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/939293/2"}}},"commit":{"parents":[{"commit":"dc7eafa720146f0ff1bacfdc01d8862a7aaabd81","subject":"Bump eventlet to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7eafa720146f0ff1bacfdc01d8862a7aaabd81"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2025-01-14 23:00:54.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2025-01-15 05:35:20.000000000","tz":0},"subject":"Revert \"Bump eventlet to latest version\"","message":"Revert \"Bump eventlet to latest version\"\n\nThis reverts commit dc7eafa720146f0ff1bacfdc01d8862a7aaabd81.\n\nReason for revert: Rally ovn job broken as described in the\nrelated bug.\n\nRelated-Bug: #2094970\nChange-Id: I89ab88ac63679a6f7737a8e04a55a66c42344c0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8915eaa5cad7f341a33e7710f69620b282bb3595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8915eaa5cad7f341a33e7710f69620b282bb3595"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
