)]}'
{"id":"openstack%2Frequirements~939577","triplet_id":"openstack%2Frequirements~master~Iaf139fd4492e4358a6da6dbea69cc14d4ffe1266","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-01-22 07:45:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaf139fd4492e4358a6da6dbea69cc14d4ffe1266","subject":"Updated from generate-constraints","status":"MERGED","created":"2025-01-18 08:16:51.000000000","updated":"2025-01-22 07:46:34.000000000","submitted":"2025-01-22 07:45:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939577-openstack/requirements/constraints","meta_rev_id":"d8bbc37ad2e7ad09f1047099e97113a641101117","_number":939577,"virtual_id_number":939577,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-22 07:45:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2025-01-22 05:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2025-01-22 05:48:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-18 10:02:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-22 05:48:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ea71dce80d646aa905233276990276868f47612c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-01-18 08:16:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe92d034d74242735a3f928ed1a6a0671281c67c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 10:02:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dde12e7777a04e10b326436cd76f3452\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9fc4f638684a9cb3003e5ac53678a8 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5869feb4d9964bc49935484c0f52c2a9 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68617b0a16074c3a8a9e7f260e2a939e : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81b8b7a3e43644d492488e0a2776806d : SUCCESS in 3m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/010f3f8ef59d44508cad496383e4dfb4 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/523cb94e192a4aa7b60c4e74c9972ae1 : SUCCESS in 2m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d3c0c71025174f0ea5f32ad216f578a6 : FAILURE in 2m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ace04d172a57413a9f51c117c8624031 : SUCCESS in 9m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4bf149d392c44573b63dc952e70558df : SUCCESS in 18m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0ceff3d763a240d1bf294c7b2271c681 : SUCCESS in 23m 20s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e18a6c2518624ff98851780a619b01b4 : SUCCESS in 5m 36s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b34598055ff74a9bb726dc0b168a69eb : SUCCESS in 6m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e6f352ddcdb34fb694e081063f7d1fb8 : SUCCESS in 4m 11s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/577ab3f04a024485be44a59328a815c0 : SUCCESS in 12m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8141aee9788b40e592a3acbffd18e887 : FAILURE in 4m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e69c2bb606fe46398b55908e1dce8781 : SUCCESS in 5m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2e1bdf5d401d4df884269af5ff4ebaf1 : SUCCESS in 7m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2dc255f98fa045708441e512d5485375 : FAILURE in 8m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/634106d459a84ee18463c6ba226942dc : SUCCESS in 7m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0c2f86da95d44010bc653e053406cb55 : FAILURE in 3m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e88416b032f94e8c99860d7c837e9d32 : SUCCESS in 13m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/85ed2ab6e8c244799e69665ad323cddc : SUCCESS in 2m 46s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/69f9edac875d47d9959284e802406782 : SUCCESS in 9m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/788cdd2c43194e8cb252bda47ea5c25f : SUCCESS in 4m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6e53d0b5c66c4843a11955e3181ff947 : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7e0104bc829e4d05ac15aa9d54561ec5 : SUCCESS in 13m 38s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4d970d226cc54ede8f4845057267e107 : SUCCESS in 20m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ccceb9a3e357423b9f8c4201b7e95c6d : SUCCESS in 8m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f86d681e9bf24977bdcdbfff35c440ef : SUCCESS in 16m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/048a190a4df24238b6974bc15fc66d79 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d525d0274aa42d79ea4e5909814ada7 : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c2d51ee01dc4f7e8a802f4b0f82c8ac : SUCCESS in 27m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/61a9b8f721944b25a4ecfa845ab3665c : SUCCESS in 6m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/39de4cbfb29b4ca2b8633355c64e193c : SUCCESS in 3m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d108167c6230476c8a6d1c9affcb95d9 : SUCCESS in 5m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d64847a9c883434ababc49b1604e47c2 : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b12fbe0362c640ac9206fe7962717c33 : FAILURE in 48m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4695829123c6431ebef0669b2c8b131f : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":1},{"id":"496f4551155faa387587308098bb51f518d613e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-21 20:04:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fce1565660fac6b4b24f1692826198188a247d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 22:22:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fe29229bc52431d83f20a77675fd57c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632cee92d04b46b8ad308647b307770d : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb974de1c5524cbc958c82cc70530fdd : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/180699f489be4c22b89278b3c1881122 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/944f0e1d82024fe087402a58bd2b952a : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/831de81bd1ed42f88e407be6f8be5345 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ceb49d327704de6a91cb3fb0d3f11e7 : SUCCESS in 3m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9429b9b643174f488207a17676f2ac09 : SUCCESS in 18m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cca2808d4880413fb4937442c9a30a63 : SUCCESS in 7m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c78d55a58c164599981d4a564a832336 : SUCCESS in 8m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b340a11333b84e308e6aca5a759f8d66 : SUCCESS in 21m 04s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/328b39032d314cb789ae260ff841ed1c : SUCCESS in 4m 32s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/332a0bc4d1264222aba4132225f3466b : SUCCESS in 3m 42s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e22db4cda6264465ae61d25fbcf915d6 : SUCCESS in 3m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0b1a9c9bcba54aff913b81ebf017b844 : SUCCESS in 11m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/204eba94442d494d8ad9e47e1c467734 : SUCCESS in 4m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1785086a98a74a96844b7017531687f1 : SUCCESS in 4m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/52493c0aa9f24fe2af61cffef13a2915 : SUCCESS in 5m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/da4c92cde431455d8735ed98d940f860 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1839fc69d9b743398b09af6ab85a225d : SUCCESS in 6m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/74b3f9c005c44e48854c87cad44c5d23 : SUCCESS in 8m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c9791e296e134a8ab361d2f938c1a6d4 : SUCCESS in 10m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c95151843bbd4438ab6899a4dfd69934 : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5990d203b95b441391fd47e015ba3463 : SUCCESS in 11m 06s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c7127a96387343f7a7cf6383f066bdbe : SUCCESS in 4m 51s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ff8643ad4e8e46a4b74e21de9e95292f : SUCCESS in 4m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8ec5f49f48b644698a6e11554c678c4d : SUCCESS in 9m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/821c5da48c814eaeacca2398f1d49db4 : SUCCESS in 14m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b400cde7954248089b8f34b41dc0f554 : SUCCESS in 8m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/45212c75157446e294a9af1bbe908d24 : SUCCESS in 13m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/26a2ba750f094f58863b90597e5fa726 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/288c3b2db423461a8b6e4de46d4f686b : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a24e1d4d2b774298be8fcc4a31984257 : SUCCESS in 26m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ea37b45c989d4c879f0ba1c69103f5db : SUCCESS in 4m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ca831ae3c4054edcaf68421720cf2312 : SUCCESS in 4m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c774479a1c754d6390c66e8fefd3a826 : SUCCESS in 6m 01s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b12448b7bccb4acd934a631336fe3ee6 : SUCCESS in 17m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63a5795452a246e1a6328517abc4a5a0 : SUCCESS in 1h 06m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35f521ccea6241329a1af5de0957f32c : SUCCESS in 1h 58m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d4fca6721cc2cd1ff59fe540eec863dba0f6773c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2025-01-22 05:48:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d5d45538fcaccdbf8a3d0ba48230b782fe26d7b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 05:49:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"56a4f6488d39e25e6dc248dc84eadc65ea8cb60f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 07:45:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f9eb26afd36494b889862250608b9fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b0f0497e0f48fd955e005464fa7cc8 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bad725342654464bea9af11e457175d : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0b15674283d401ba18c122f93ac8a74 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/732dec5537a74897b1714394513313cc : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0039c53d01f4e199fa45e2d9833e818 : SUCCESS in 2m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68a5d6adeb0a467aa4a8beef2607f909 : SUCCESS in 3m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/deca2f2f72624ae7bbb3b40f34698173 : SUCCESS in 13m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1032f96e23a94a8696a27c1547c1933e : SUCCESS in 10m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cbcc260c1cd64201a59dd695b9de48ea : SUCCESS in 10m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dbab1ac80d894ad8b322cb355d4caf54 : SUCCESS in 18m 11s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6c91bbdd53ed46018f2d95032124be76 : SUCCESS in 13m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cccadf115f0045548b730cdec1195247 : SUCCESS in 4m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/42d43933f9b3404496e059e8d55b03e1 : SUCCESS in 5m 40s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d660cfb754384a69af0d44397e994494 : SUCCESS in 7m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3aa8301df39e4ab4938423da1ecc6205 : SUCCESS in 14m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0e848f299614ac7bee736840babad17 : SUCCESS in 6m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4ae9af058d7644e8aa27044f68c28c6e : SUCCESS in 7m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/545bb1c6ded9495b86c5d08eebbb715b : SUCCESS in 10m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/bb0a367ac74a4fec9c7d57d4878cda24 : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d2cd66624fb5457786c7168a0b0f2594 : SUCCESS in 9m 02s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/96aeaf751a1344ee8b3584ae8994e5e8 : SUCCESS in 19m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c4dea4dd3bea488aa0f4bf08f629a4c4 : SUCCESS in 7m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1ebd3e9deeaa49e5b182316a2ccaa68e : SUCCESS in 14m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/604dc0874341481e91cf7acc908bf4a6 : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/abd8d548e52148f5a15d2e469a85726e : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b7d72da68854985a888877ec5acc296 : SUCCESS in 27m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2a0f75233df24cfd993badbdca5befba : SUCCESS in 6m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c06dd197882b47ec8c052e1e01a0035d : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aea1888b63994b6793098f420de91105 : SUCCESS in 6m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9eeeb74377a54fb3b4ac6c5878b48360 : SUCCESS in 23m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd33abc88dbc4bf282fce50211dc57c5 : SUCCESS in 1h 08m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8ab80b8b6154669be5bb8cc94fda8ed : SUCCESS in 1h 49m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e3f35f908e1ada2bafa9e7e47c1c0605a0dc387d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 07:45:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d8bbc37ad2e7ad09f1047099e97113a641101117","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 07:46:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be96541d8b4477086bcc1ed6d3ad08d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3944537e58c346d8a7f38dbc4fc3e2cb : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"42d8dcc4c74f54f338ff540c8053690d317dce80","revisions":{"4af102224975d05a854c4816068182d7a4241698":{"kind":"REWORK","_number":1,"created":"2025-01-18 08:16:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/939577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/939577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/939577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/939577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/939577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/939577/1"}}},"commit":{"parents":[{"commit":"7012e0bad5b704f95e652f70c737cdbc84d43f71","subject":"Merge \"update constraint for ovsdbapp to new release 2.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7012e0bad5b704f95e652f70c737cdbc84d43f71"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-01-18 08:16:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-01-18 08:16:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iaf139fd4492e4358a6da6dbea69cc14d4ffe1266\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4af102224975d05a854c4816068182d7a4241698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4af102224975d05a854c4816068182d7a4241698"}]},"branch":"refs/heads/master"},"42d8dcc4c74f54f338ff540c8053690d317dce80":{"kind":"REWORK","_number":2,"created":"2025-01-21 20:04:28.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/77/939577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/939577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/939577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/939577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/939577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/939577/2"}}},"commit":{"parents":[{"commit":"7012e0bad5b704f95e652f70c737cdbc84d43f71","subject":"Merge \"update constraint for ovsdbapp to new release 2.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7012e0bad5b704f95e652f70c737cdbc84d43f71"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-01-18 08:16:51.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-21 20:02:36.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied for automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: Iaf139fd4492e4358a6da6dbea69cc14d4ffe1266\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42d8dcc4c74f54f338ff540c8053690d317dce80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42d8dcc4c74f54f338ff540c8053690d317dce80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
