)]}'
{"id":"openstack%2Frequirements~940134","triplet_id":"openstack%2Frequirements~master~I29936e8ba38ebc445608b832aad381c2bbf1321f","project":"openstack/requirements","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-01-25 06:00:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I29936e8ba38ebc445608b832aad381c2bbf1321f","subject":"bump to eventlet 0.39.0","status":"MERGED","created":"2025-01-24 13:40:36.000000000","updated":"2025-01-25 06:01:28.000000000","submitted":"2025-01-25 06:00:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940134-eventlet-removal","meta_rev_id":"d69208b5f488bfe4a90d38503a81a27395cda2d3","_number":940134,"virtual_id_number":940134,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-25 06:00:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-01-24 13:50:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-01-25 04:29:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-01-25 04:29:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-24 13:50:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-01-24 15:41:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-25 04:29:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"20a27edd3697a74b7636ec1da2d77a2c3d529015","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-01-24 13:40:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a066ba859c114a0b9f3515cf4511fb516b37f3b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-01-24 13:50:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd8c98f6e9a82ecaec8b352f5d91213c39dbbd35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 15:41:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03bfb4a91f4540c7a362069899c9e67e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca210011852467695dad8128e4f5bb5 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b56f36cea94578aaa282fbf349897a : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2432fa15fc73484ba348d5c8c3fbc88c : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4703180a4010465aafd801136a84e0ac : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1358a4819c7c49edb3facb0cc215ae5f : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7ac7df8ed124b78bf13aa3956314b37 : SUCCESS in 3m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e5d46929781447d58197bb07537a9a87 : SUCCESS in 9m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2e3636aee1e64a7bb936c6cd6fca4a8d : SUCCESS in 5m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0b1b478935174d358e21aa400c04c1d2 : SUCCESS in 19m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c7a15707e03f48d7acdcb431a606c4a9 : SUCCESS in 10m 21s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e9ac6a3adc2b4ea4923e200ecb07c603 : SUCCESS in 18m 10s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c6afad1532c747589e9decbb89341bf3 : SUCCESS in 12m 37s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6301e65484a74189ba2d3f1431733b6e : SUCCESS in 11m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/854c77fbc987427595194a73af506582 : SUCCESS in 10m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6c2be4aadc2347efa1530291fc148cf7 : SUCCESS in 4m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f0070ec78bcf4a11b6f8ead717266754 : SUCCESS in 7m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d5af465498f14536b4c4935db09b5bfa : SUCCESS in 27m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a0bf973c34754992914dad1c6c987b87 : SUCCESS in 13m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c137b2c6b2854a7884837460b3324863 : SUCCESS in 6m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/27b42f10bc0a4546b153e27b76ae5ea7 : SUCCESS in 8m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/645a5883172c41528d735dafa47745e4 : SUCCESS in 12m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cf788ac5a52c4eff888b086e4dbaaea0 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/dd0355458555425a9ce2965698931f1e : SUCCESS in 12m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/167287a575e94dc7ad05b161cbf361bf : SUCCESS in 4m 29s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/eea16a18c98d48e28ae78d120d93fcd7 : SUCCESS in 4m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c17f0bf372ac4a2ca1c54739f2e8df5c : SUCCESS in 11m 31s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4a74bd2959d94313baa2e9e06cfab058 : SUCCESS in 18m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dd0d7d08d4a14d8984755713eadec9b0 : SUCCESS in 8m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ec9b26a3307e4986a594e52d38ed3597 : SUCCESS in 15m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c661fd47250b4504b8ef3e80d8369072 : SUCCESS in 24m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c1ac4985f9244817920cef76b2fe6fe9 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6cb91e579324400bec8211d79a972f0 : SUCCESS in 24m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bc013ff13c2c46d0ba3cab1d59661ae2 : SUCCESS in 6m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/602ca481cd89459f9084413398132e65 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88cfa7f0b2d843e788b6210b5416b565 : SUCCESS in 5m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1df20454c0d94d69a772dc39646bdb30 : SUCCESS in 17m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bbc04f3f35842eab20ce86b6def8fe1 : SUCCESS in 1h 05m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea1e7fd972dd466bb0499e9c2b4d7fe4 : SUCCESS in 1h 52m 42s","accounts_in_message":[],"_revision_number":1},{"id":"56bf5c6cf3889cb082f54e1a78ba573e0a069898","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2025-01-25 04:29:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9a09c1f992b3ca38bbc7054e723bddc8cf98373","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 04:29:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b4b717e21c8501b213b55b10196d1843b6a5029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 06:00:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fb31218d31845c9aa5d9bd2149fe8ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c5d945450e349d89b5846572a52babd : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d74c36a6c18c4d53830f15c84e926833 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62d785889b8d4f028b60602597653b1d : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4eeb42debfdc4184aee67e82ff59b294 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/448d41adb1b041b09afb3764de5f388a : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e44fe62df784013a1ec9390e24fddbd : SUCCESS in 3m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ec52f2902f764d8d89bfd269f8c7646f : SUCCESS in 21m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/87ee778281f241ecbddbb3594c327fd0 : SUCCESS in 12m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b765278527f94110b931a702473c082c : SUCCESS in 17m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ba4a47e5e2b34bc08f5bdccb5fbc0770 : SUCCESS in 10m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/001752ea84ba4fa1864f4e3339b62ccd : SUCCESS in 11m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/60ee22103eaf4aaa9ce27e32b2607795 : SUCCESS in 5m 22s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2c84e6e88cc74367b565137c4e899aa1 : SUCCESS in 3m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/73d2180812764486bf75357d679380d8 : SUCCESS in 7m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/69fffa3a8ab94e928e75ec831b8df64b : SUCCESS in 6m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6abf25824b9d4ef58b79be6ec4c5e393 : SUCCESS in 6m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3f1ec0ff391d4dc7843115071c628396 : SUCCESS in 5m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ae779998af65458e962f6fd6603bc2bf : SUCCESS in 9m 59s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/738db2233d624dc0a7f2e80259fa388a : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1c8375b153164ba695769fd00c9b600d : SUCCESS in 11m 47s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b24ec6487e534c898854e50aa73c9460 : SUCCESS in 21m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2560885faac244d4bb4203c7f04d7506 : SUCCESS in 7m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/29c6003376c24185b9faf122aad7822f : SUCCESS in 13m 16s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a66a6aafc34f47998b8bc6f43bb0d53f : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c317808ab5ac4169acd68e651caa73f4 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6332ef7129e04334bf22dfe88877567d : SUCCESS in 20m 31s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/869b1693f5334e46aa07de9d27a1de25 : SUCCESS in 7m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/24432fa1157a479ba5ea6a31eefff68a : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2446cec441db4fe087cccaecb6d696bb : SUCCESS in 5m 18s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1e97f93c28d142158391b3967b0c48a6 : SUCCESS in 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/190f7a4fc4c2441989a5b2cbb412dd33 : SUCCESS in 46m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbd9a535044c4794b9183d5b734b552b : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":1},{"id":"77abfa5280ba0922b25f41423bbc50a606002e93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 06:00:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d69208b5f488bfe4a90d38503a81a27395cda2d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 06:01:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97807aacfa254579a54e168ae5c699b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3561b5c0b8884d38b7d7c962954b5072 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85bdde267a48fdaf53e6933a5f81e2bb95c1010b","revisions":{"85bdde267a48fdaf53e6933a5f81e2bb95c1010b":{"kind":"REWORK","_number":1,"created":"2025-01-24 13:40:36.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/34/940134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/940134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/940134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/940134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/940134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/940134/1"}}},"commit":{"parents":[{"commit":"42d8dcc4c74f54f338ff540c8053690d317dce80","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42d8dcc4c74f54f338ff540c8053690d317dce80"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-01-24 13:39:20.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-01-24 13:39:23.000000000","tz":60},"subject":"bump to eventlet 0.39.0","message":"bump to eventlet 0.39.0\n\nMostly to introduce:\n- https://github.com/eventlet/eventlet/pull/1015\n- https://github.com/eventlet/eventlet/pull/999\n\nChange-Id: I29936e8ba38ebc445608b832aad381c2bbf1321f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85bdde267a48fdaf53e6933a5f81e2bb95c1010b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85bdde267a48fdaf53e6933a5f81e2bb95c1010b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
