)]}'
{"id":"openstack%2Frequirements~940448","triplet_id":"openstack%2Frequirements~master~I6fc11038de2f4297908802a331c303cd05565719","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{"37702":{"account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"last_update":"2025-01-31 11:38:08.000000000","reason":"\u003cGERRIT_ACCOUNT_37702\u003e replied on the change","reason_account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}}},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-01-31 13:29:33.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I6fc11038de2f4297908802a331c303cd05565719","subject":"update constraint for os-traits to new release 3.3.0","status":"MERGED","created":"2025-01-30 14:09:32.000000000","updated":"2025-01-31 13:29:33.000000000","submitted":"2025-01-30 20:30:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940448-new-release","meta_rev_id":"03dcd27f488c81f2c2699bf4b94a40f995b7b1b3","_number":940448,"virtual_id_number":940448,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-30 20:30:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-01-30 18:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-01-30 18:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-30 15:06:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-30 16:50:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-01-30 18:55:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-01-31 11:38:08.000000000","updated_by":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"reviewer":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"state":"CC"}],"messages":[{"id":"92577b179b633810d3be2402fadabbeeaae3fc2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-01-30 14:09:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6abdf474b31d9613a672759c534d239cc8047e1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 15:06:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfefb800b5984740b5fb61cb134cf4d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d62eba186a34a8199babcc009a49885 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b243d565923147bf9a61a69593e791f2 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38fad2b5e7bd4f6e8460283c5119c27c : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e4ca8c45da741d3b410cc7970a611ec : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/648823fdb4004f97a2845eb2505a59be : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dcd2a3f5d1d349208942e2939b337068 : SUCCESS in 4m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d92b77aba31741aea875791727a04935 : SUCCESS in 16m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/13fcd91e40504db393063ccebd3a5fb1 : SUCCESS in 12m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d46eb989e5af4e94a58fb490e27f09b8 : SUCCESS in 16m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/09520d69e03d4625a23f9f93778624e0 : SUCCESS in 21m 14s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/201fd6c3c99d445ea0eec11e32caa6ad : SUCCESS in 7m 10s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5263ee90d70449499ad49042ebec712d : SUCCESS in 7m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/88d3d3dec103463080d589bd6c72b6bd : SUCCESS in 4m 17s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3fab51a1603d4037944bb938d7f9ff19 : SUCCESS in 10m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/dd09ecdf4771491387ea71a12be89530 : SUCCESS in 5m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/124b7280c67a43ff9a0b681df2c0882b : SUCCESS in 6m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c5a8580e1ade470ebcedff5ce86fe555 : SUCCESS in 9m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dfebb35eb0b84cdab5227d9794499a66 : SUCCESS in 14m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49a0c86b6d2947aca0bc88a4ad211a33 : SUCCESS in 7m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cb3f00e0ba45471e88ef74f8696673b6 : SUCCESS in 20m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/09bc8baecb0a40e4845a480894982bb8 : SUCCESS in 12m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/10e9401a44594a60bf496a7e0726d0ac : SUCCESS in 3m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d15321ed412d4531bb2d0ee23b20eeff : SUCCESS in 10m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7e4b496815424b97bd50888063e387d1 : SUCCESS in 4m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5a72b035a6f040048e686cc34058b4bd : SUCCESS in 6m 12s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/95b2c449fffd428d9177866c611d5b8f : SUCCESS in 14m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/33ceae4444ef4c1f92e8afd52478c389 : SUCCESS in 20m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4efbf90cfcdc4250a5fc8ac15768c6d6 : SUCCESS in 7m 40s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/419bb0a84a89426ca9414fa3ac0b5bd0 : SUCCESS in 12m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/228a86819fbb4ab39a71ba45e3be8856 : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3f2cc1e2f2d42a798e1f4f01a306dc0 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b882b41545a451c99846ac452e6591c : SUCCESS in 21m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d5038828079947f8a63ebc3cb8471458 : SUCCESS in 10m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5b624dda80aa46f1b10a0d6a3b209f2e : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5d4c423cb7b4b9688a09900db95bce9 : SUCCESS in 4m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7a2fce0c27ec4a27bcaa0c0649b8f01d : SUCCESS in 17m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40f644a6bbe241a9a4cc2b61b8c2dc38 : SUCCESS in 52m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d8148292a944cd7904f7f68f62e191b : FAILURE in 22m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c7c010af54eb114822d4c9da7da7d41bff251752","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-30 16:50:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be4acd21c0a2f2f7fa5ac0bee8a7150ec89f08e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 18:45:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2df03c389784f348d93678bc3ce05c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c05fe5163e84ed585c7bf11ac8e9f35 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb9264bd9794234bdcd639f0c36cdd3 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f2f840744ee4e37ad661fc93fc390ad : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/009b1dabbe8b4b75a4d8c592e8f410c5 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d725530044c43168290f3255de9c6b7 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78b3d266d02a46d1b22c77bce2ecb1db : SUCCESS in 4m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1a34d953f7124a7c8ce52b4f09bb78c2 : SUCCESS in 19m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/063901b5face4c56a2104a6f63d9c374 : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/71235fa3db184c309ea09f187cfe3f72 : SUCCESS in 19m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1bb8930b13fc4581822df037b4e9d8a7 : SUCCESS in 29m 27s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d5d7ac2e6ed7401992282a19b91cf836 : SUCCESS in 24m 29s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2592fcdaf17d4ad7a420b3f8282961bd : SUCCESS in 7m 09s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/59688677fd8f426db792c269f328c8d8 : SUCCESS in 17m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2350d43129224b88b47f34e540e17828 : SUCCESS in 11m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/df25d7e1093c44f695a66a3fcfe29d35 : SUCCESS in 6m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/28f8b24e9b1d40ceb039567db362c71b : SUCCESS in 5m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6ece8ef4a51c4a44ad42730e3a991c66 : SUCCESS in 8m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1e22aaab419c43ba9e6a10fb2f98654d : SUCCESS in 13m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96e937d0f5354230ab9d28e5f1d422d7 : SUCCESS in 8m 45s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e7b6fa1a55be49c593dfd7e9b91e5cdd : SUCCESS in 7m 36s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b97c7174ed21427c935dacadc4db51f6 : SUCCESS in 12m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/10cda55cbb74466089d6948184bb3f56 : SUCCESS in 2m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/121e8986f25f4d68860447576e3f2d6b : SUCCESS in 10m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/975d34567fec47f8a427edea2624af1a : SUCCESS in 5m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2155981ee0d8497cbd1e2d2dda559837 : SUCCESS in 5m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/688aed67532d402a957f45da93b002b9 : SUCCESS in 12m 24s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9cea8dee58e94d438d21b272314db171 : SUCCESS in 19m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c75fff7d4bb240118ee44b398793fef5 : SUCCESS in 7m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8aac2d8e6bf74061af96552fbe03d3e3 : SUCCESS in 13m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/71ea1f8d78b6420d8b72020e011231e0 : SUCCESS in 6m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d80a35054b14b16a8ed902078274402 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f6b97b259634bb19b01d1ef36622563 : SUCCESS in 27m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/78230e5c7a244253bc4005f06891efbf : SUCCESS in 7m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/81b520b2cb244ad88865ab0717ce714a : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/21a8fa9ad7f440d189c6b59f8f6e21b7 : SUCCESS in 5m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f93f8bcf06ca42358a99e4326eb1469e : SUCCESS in 15m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b9bc723307a486d966fa7199d7b2a34 : SUCCESS in 52m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4d563ad79f147588a332f22225673e7 : SUCCESS in 1h 23m 51s","accounts_in_message":[],"_revision_number":1},{"id":"39768aede53e2f8d6dc50ab3141c245905aa77a6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-30 18:55:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca6ad862a48bd9d85b24476fe27f37e4b4a23a3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 18:56:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8ba60e24e0c36cb81749743b8c0795ecac22801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 20:30:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7956f9d7f5e9474899c17e34140a5730\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa6809cc66c4ba78fef2b93955aa8e2 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba1c167257e04135ac8d11e84d896978 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7600886fd2a2498fa4b02e2011c12171 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/335e3237e3584334bb9a2bc63e79b926 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/88cb16efca5b4cba9adc43d7991fc80d : SUCCESS in 5m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a2eef48f5c8c45688980180560712196 : SUCCESS in 6m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/28b8c65ed50743da8a420178f5e47023 : SUCCESS in 13m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/829e6984ac26469bbeff64e37f9a3cd5 : SUCCESS in 9m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8ae92b75751b45018b92f87af9315003 : SUCCESS in 29m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/51cbbd115dc9465a9451c656ad3dd175 : SUCCESS in 25m 27s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/91c27ea8a4ad4b85ba3fc7655dc1b837 : SUCCESS in 9m 34s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c402ffa4459f4e3397a35b908c00320f : SUCCESS in 6m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bf65c61ca4b44179a4ed2b75d660811c : SUCCESS in 4m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ee394492dc1f4d5588b4778657781fdd : SUCCESS in 7m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/349434790e0845b7ae99f6af14751ce4 : SUCCESS in 15m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/404bf53b2ca345e5990b8977e9752922 : SUCCESS in 7m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fa0fd41b331d47bcb7140902eec0a2a8 : SUCCESS in 9m 24s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/58a5efacd8f949a7a993759817baa7ee : SUCCESS in 11m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e278e1346b4a433caf8daf16c0ad8829 : SUCCESS in 3m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/843a1c1e6268459b967c5037c7ac56dd : SUCCESS in 12m 46s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/abd3b3432842441b86b264d37f810dce : SUCCESS in 22m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/26b243f9d2ed453f99140a0b1a8e7b4d : SUCCESS in 7m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5c4b03fc8b1641cb995eebf2f5906f26 : SUCCESS in 10m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a7f674ba25e644c689a29d67dde891d7 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9b0b4f2ad79443ba8e17910cda37f1e : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8fddb3d27e148c290138ed520536117 : SUCCESS in 26m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0470580c276a4d54ad7fa7622d5fbc2a : SUCCESS in 11m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/89b5e0d6855140a19be29ac85c44356d : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc84dde00ca84c1893b78f25552ca49a : SUCCESS in 3m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9fc3caf343e6472985b6552e802dd116 : SUCCESS in 19m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f9113e9c8ff4689aa596e685a328639 : SUCCESS in 57m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9dfe37dca28244ddb11a4cd3b581d29c : SUCCESS in 1h 25m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a2cafd430c2d61e19f713453cc8fe37396b0e548","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 20:30:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e514f737a913f1cca58d5dd859b9607035acff5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 20:31:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc7ea30d14164290bd5950982d37828d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d633ce4858914569b4fdd47f503acac0 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1},{"id":"7a1ac4ff9f80e4bd3e3467f49e80b028bcc08cde","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-01-31 11:38:08.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9efb564b520008091152d9f7f0ce0033cfb9470c","revisions":{"9efb564b520008091152d9f7f0ce0033cfb9470c":{"kind":"REWORK","_number":1,"created":"2025-01-30 14:09:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/940448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/940448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/940448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/940448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/940448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/940448/1"}}},"commit":{"parents":[{"commit":"471671f6ea4a1102c4e3143b9a1b6bae4a8c0e61","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/471671f6ea4a1102c4e3143b9a1b6bae4a8c0e61"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-01-30 14:09:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-01-30 14:09:31.000000000","tz":0},"subject":"update constraint for os-traits to new release 3.3.0","message":"update constraint for os-traits to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: team: nova\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: René Ribaud \u003crene.ribaud@gmail.com\u003e\nmeta: release:Commit: René Ribaud \u003crene.ribaud@gmail.com\u003e\nmeta: release:Change-Id: I62df064a9a0db7a0500ac80ee2cbe0f3b373c0dd\nmeta: release:Code-Review+1: Amit Uniyal \u003cauniyal@redhat.com\u003e\nmeta: release:Code-Review+1: Balazs Gibizer \u003cgibizer@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6fc11038de2f4297908802a331c303cd05565719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9efb564b520008091152d9f7f0ce0033cfb9470c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9efb564b520008091152d9f7f0ce0033cfb9470c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
