)]}'
{"id":"openstack%2Frequirements~940569","triplet_id":"openstack%2Frequirements~master~I34a126282b4131c8167f08c925734c4b4251bfb8","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-02-21 15:44:42.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I34a126282b4131c8167f08c925734c4b4251bfb8","subject":"Updated from generate-constraints","status":"MERGED","created":"2025-02-01 08:09:50.000000000","updated":"2025-02-21 17:26:39.000000000","submitted":"2025-02-21 17:25:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940569-openstack/requirements/constraints","meta_rev_id":"b7992173f898395cf6b926b4b35329d5cf9be570","_number":940569,"virtual_id_number":940569,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 17:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-02-21 12:37:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-02-21 15:44:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-02-21 15:44:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-01 09:14:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 12:37:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-02-21 15:44:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"58a040922fdba9d3d4aa95945a730fc1b198b8a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-01 08:09:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86ee24fb5ae3d6e7b6553e73ab0d524abcb7186d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 09:14:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e22dbee07a24381a1a5e763efbee01f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b383bacc720741a48f3e544d4ceef7a7 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85b87ea2ebd14e3d98d72a0a311dc8b6 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60939cfdc87b443b92dd5fa9f5f38a5f : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb13bd1e771742b8be22b7618e78aea0 : SUCCESS in 2m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83a0ee18d83f4eccbd635e11d01666ba : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ca07f0acdad4526981c9e6163e286a4 : SUCCESS in 2m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2166f8b4755142a29d8cbb4d2ebcc59a : FAILURE in 3m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5bff6ce077df445d8dec4288c079f8fb : SUCCESS in 5m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9dc780d8b1994be3b86d3900d27804a6 : SUCCESS in 14m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/124816e303db4e45bb3243e399c056e4 : SUCCESS in 10m 21s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e58a281aa7174d0ca892c9e4274b9cda : SUCCESS in 5m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/75ea19e0afbe49858bf0466a44dc7350 : SUCCESS in 5m 31s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/577603f70e954e989b31bdd308eca8b7 : SUCCESS in 4m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bdd4cda85bee47cca5b00f7eb6a622ee : FAILURE in 5m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/78660384f6494457a74d49f608d97ddb : FAILURE in 4m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cbaf13982e884f9cbed86d1125910ed1 : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5b526c6866f343d084c04cc451a9d13c : SUCCESS in 10m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dcb58082c35a4c839fcec7feae1c0046 : FAILURE in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f1d9c22ca8349c2a469e35ad132a0e6 : SUCCESS in 7m 11s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/469ecef4f00f4ca1a8c009eac533ba68 : FAILURE in 5m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8839bf4e5ac541199b648137ffcda8ea : SUCCESS in 12m 44s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e0d9ead9c3cf41c4919bcbb650077d3a : SUCCESS in 2m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cf17b5d5d65146f9949ed72b76a4e99f : SUCCESS in 11m 07s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f0f48cbde5514734a579a646f8d4e510 : SUCCESS in 5m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b0be9fd0c3204acc8417d3e205f500b1 : SUCCESS in 8m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/29e916bdcab74cde9b6cc1043f3d0576 : SUCCESS in 12m 07s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6636679bb30c41d99d0f8e98eee06396 : SUCCESS in 24m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2a0ff19a52e74ca69c4a0fe4c02b4c71 : SUCCESS in 8m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/226e365f99ad40ecb7e81efea390a21c : SUCCESS in 14m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2f4fa83f50a04549b31651ed046ce1d1 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/913d74674efa447c8db953e85fd14582 : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed8d107090414e6a842f548eb004e528 : SUCCESS in 24m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b2128a1de10e4cf08addc07fa1f9ae82 : SUCCESS in 4m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fa536680f6a845d289b9ca01b116e8f8 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d7a4359d8c14a7db1456feaa52264bf : SUCCESS in 5m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/91d5380f487044d186a3eddb1cde65c0 : SUCCESS in 16m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d687b0d26b7c4b65844f9663f9d3f6f7 : POST_FAILURE in 28m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/363f593d6d674406ba53c09ff24988fd : FAILURE in 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"be5e4a89bcb03bca9e0e2582bb26bde8faaa0d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-08 08:14:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3a2acd043aa7a024f426bac3f79c36258b0b7be7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 10:01:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53618b3466ec48b99d814bb6eee1ba60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8fd09d8132048edb2c1ccbe9f8a0e35 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57cb7efdf9bf438db7f5602c173d09d6 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6f7fe8cedd94575a67f2155a88cb2fc : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b4a46578a8744d293d90fac0a8f9378 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d338c5a1dd86445f8609839df905f55c : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad0a3cb91810452e95198a2c0dc3fd49 : SUCCESS in 3m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/60c69fd44da546d2a5381af31d141380 : FAILURE in 3m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d21cf07960184d4c91406b83f874d3b7 : SUCCESS in 8m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c8164b9f9e674b87922029b496abf6fb : SUCCESS in 17m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1ed1117d6a9d479b8ceb841ac8371a32 : SUCCESS in 15m 00s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/87e301e906e04d5b9cdca159105e77d0 : SUCCESS in 5m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6ae3154d36704e3e9c31b20c8a320cc0 : SUCCESS in 6m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b308fee0470249cfa341b92908024370 : SUCCESS in 5m 23s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e25d5972bc52446f9fd26893030b8ca7 : SUCCESS in 10m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c7d41fe11a0342c691b68151c4d16a0a : FAILURE in 5m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/178da0a218924663a4558b37cbf8b624 : SUCCESS in 5m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c33e605b27cc465399111491e92499b4 : SUCCESS in 6m 12s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8fab34cfa136448797dbaa1a9e41c015 : FAILURE in 8m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/485b4bff3fbf4de08b69ca98e6582c5d : SUCCESS in 7m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/db285244cba34c058df89a45f4b15f6f : FAILURE in 4m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/151e2326df264c3c80373f918900420c : SUCCESS in 11m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cbbf05125b8541748bc7c7d3e09b8c3f : SUCCESS in 2m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1cad327fb86348038da1c0355bf9e929 : SUCCESS in 10m 18s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/48c04344e40e4dbe9c44acd7d79a16f1 : SUCCESS in 4m 22s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2d9e3cf220834cc08385d9651e6e1eef : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b904a6ed2888420cbd234871372dc7d3 : SUCCESS in 10m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ef935d8e5fda456da0771bdb9c0d4314 : SUCCESS in 19m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5fd053a4c4c84d82b540b2bc9ee9fa7a : SUCCESS in 10m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cffce75772e14b138fae81aba0924f1b : SUCCESS in 13m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d545d14fa8dd457db9f841616d2a876c : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f766999b4362433fb2f153fc55e7a208 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a31edc507ccd4d31bb64ae17faa15606 : SUCCESS in 24m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/31fb732abd0e45dba2e558fa072b6625 : SUCCESS in 8m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8cec04fb1dc64df4b785bb414d021b7a : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61577b8e2d58441185b92b22b64a75fb : SUCCESS in 5m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/03bb24d9d4204ee188f02ff5ebfdcd75 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/409a210b435940e99807294a30829092 : SUCCESS in 1h 04m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17c88b1fa3bd40eea159148218d257fb : SUCCESS in 1h 40m 42s","accounts_in_message":[],"_revision_number":2},{"id":"a02e4544ecfd6352bbb9f2996c0df64e70865d13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-15 08:11:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"54f9efa96488471978553a7909a08f36c7da93ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 09:23:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/825c58712c5a498290c7c4b7ddb4a16b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70668a3038394cbca938bcc0e29813bf : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b314451dc06440b9b0b571fbfe7e516 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c29c459fa84843d8ba7e6b3abdcd169b : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b93c4d3c7ed8433e8975e3ae25df8110 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1346139b2b3469595b63c577d399a79 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4bec4d0b4dc04a4f9bf95493960a5a94 : SUCCESS in 3m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4f7d27f09fd14fd485dcc78730d8c54f : FAILURE in 3m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e981f331894f406cb9d45b35a106c8c6 : SUCCESS in 10m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66b49836404843dda6a55d29adde27d0 : SUCCESS in 17m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/84da250e9fbf46b5a844b1e9583cef2e : SUCCESS in 19m 44s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5775b46a5ba24739bdadba82c6d02778 : SUCCESS in 3m 33s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ae27de3e110d4102a7e260c00b17b399 : SUCCESS in 7m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b7ad04f4739644109b32a95bd889f224 : SUCCESS in 4m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9ce915af38ac456bbab0ac0e9fcc5bea : SUCCESS in 10m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/36db7cc42c6c48fd92527905e191266b : FAILURE in 4m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b23d6de816574ad585cb7d6da16045c0 : SUCCESS in 5m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/451baf5462cb40ecac1cbc7fd32b9d03 : SUCCESS in 7m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ae9b2d97e7b5415faaa94ef164161d06 : FAILURE in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5008b88a3c4f4c88a2a5354262fd8c61 : SUCCESS in 6m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cf9bfef189eb42f5b81a8a1c9ea50e84 : FAILURE in 5m 01s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f7b555661fd44bbfabe6f1123c78188d : SUCCESS in 11m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2a453ee8010a42c1a2ed8040309c6667 : SUCCESS in 3m 57s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c34e121dfa884740aaa976139b3864de : SUCCESS in 9m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d27c42e0d6ff4ff99429deb954b2daaa : SUCCESS in 5m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/eb5eff1d80e94e6f9982baa7cf001fe1 : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0df86d890e5d464ba3792bd9ace51a06 : SUCCESS in 10m 41s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/70c79b19b8334d96a7d93d56c8ff6476 : SUCCESS in 20m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ba45e639b00e45cd836cdb9ebb8516ca : SUCCESS in 8m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/44e666bcb87f4ac595027a17bbbc18fe : SUCCESS in 11m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ed41e52be4404089a275ab55230aedbd : SUCCESS in 6m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0c026d2fd70a4cbdb7924f0a0e232f14 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/028fe798c7334c57bad8d6a095db65ca : SUCCESS in 28m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d00fd1493f1243edb2353367b86f0c5a : SUCCESS in 7m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aa991e1db2ce4291a685122c4b232a04 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9144aecf7d584117b8c7c6900c27f7bc : SUCCESS in 5m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e699600339bb42338f698a45b5365979 : SUCCESS in 17m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/693dfa11f45249e6a3afc44ce204770d : SUCCESS in 55m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8736a35e44114e5e8f48b6367220d03d : SUCCESS in 1h 11m 04s","accounts_in_message":[],"_revision_number":3},{"id":"46b3ead18c71f1e0d359715aa9681eae56b5db54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-19 12:19:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b047a63f49d9aada16200d157ed16bea90c59c46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 14:02:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74c5e4dcf33a46ed97936a3cea2f02f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5059ba08c5ff464b8f5807ebf7b90bf4 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2514571ea5248ba830a68735f142307 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a9e78ea7b0949209ffb7dea4e830ec3 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edfa11c71b68486bb6c38518f30f0581 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2147600a2414e82aa9bffc971891fa2 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bdc91592c2f04113a343b9bb0f48b5eb : SUCCESS in 4m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0c9b63eb0e1b49de90a7737b108322bd : SUCCESS in 8m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c636e20350954213b9fbb6d53bf3ba95 : SUCCESS in 9m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c6a8d67bda6843aaa2ee982c80b6d58c : SUCCESS in 9m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6d5717e5955a43319d8c3942da4cff09 : SUCCESS in 22m 44s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a1357a1159ba4509bde7a91ca413e61d : SUCCESS in 5m 34s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/80a99006c9e6411ea959cefd0d4ac610 : SUCCESS in 4m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f155577d597b40439890b47ca91c7aaa : SUCCESS in 3m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/935f73996987465187cc53c4c1509f6c : SUCCESS in 11m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7087633fe4fb44f9beea74f201b6904d : SUCCESS in 3m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/02c5351cd8f2449082a9ffdd7bf59d13 : SUCCESS in 5m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d76a5a5db8254b8aba5b26653094cd44 : SUCCESS in 9m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f7521bd273a8475ba6560a5aad0ef44c : SUCCESS in 13m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/68ed2b51372c4854a2834cbbe1ecd691 : SUCCESS in 6m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5e5b373fac7f4b9e991f4d71a7dd736e : SUCCESS in 8m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c49ae0dc79f4418a822e87c7ba0d298b : SUCCESS in 11m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a948db652b014ed68d299c6babd43ed7 : SUCCESS in 2m 40s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ee8a720c7a634981a919e3a8799bc25c : SUCCESS in 11m 01s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4a5fdb2cbb384bb4a397f68b03874fc2 : SUCCESS in 4m 24s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4fe46820c1b945efbfdbaca67afe7f40 : SUCCESS in 5m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/66e69b38e56c4e78b8a5f18b19b26542 : SUCCESS in 10m 35s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7f9fbb372ffa41f4b7944712ca584a62 : SUCCESS in 17m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fa8dd2f1f2cd40b5860611779ae5dd5e : SUCCESS in 7m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/27d5ac298e0e420797cfe3c5061d132b : SUCCESS in 15m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2725ac0533974f4d85896fd92b502fed : SUCCESS in 5m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fd7583a50cc1491389b4ef2a1ee507db : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b774500b8bb84d3e99b938fc8ebf508c : SUCCESS in 22m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/359eb11db77340f5831081f77e1a97ec : SUCCESS in 6m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/651b3151537045a1a865d8386da96157 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a04e0a0eeea41c08ba0145709d7ebc8 : SUCCESS in 5m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0f7965a5e9de40bd99187764680fce50 : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7cb0c7e9b0148759b79828581d8b707 : SUCCESS in 1h 14m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/749cca895fb7404f9692b54feeb6cf04 : SUCCESS in 1h 39m 48s","accounts_in_message":[],"_revision_number":4},{"id":"72b887dd0831d5268133ed488ac20e3b8d47e5c5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-21 12:37:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6ed18f6da038cbb8725e6d79ebabdb3162e00470","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-21 15:44:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c7b5c8e943c75ff7c6f2605b288ff5b3e1b4940","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 15:45:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"215c496c77da5d0101ab25764da7ac2ad564a80c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 17:25:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15adfe77b7bf41d4bcedca1ee5e2d542\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0bcdc98cca94df39193228a7af9fe82 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f228f5cd6f4f49d68536fad2e3f2021f : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fe8bf576a0949f2a8e7784a6e385043 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7149d8a974464cc1b1e3d9d0c850442c : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe71e62b097c4667be45f04063562103 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a9e2ae733274e248aa02e81b67e4e4e : SUCCESS in 4m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/918bc59f8aa44775ad735d080b2df37e : SUCCESS in 17m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ad67f013b9e94347a8e667e6cb8d3684 : SUCCESS in 9m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7e3234c0981e46f2bfe78d42565aa793 : SUCCESS in 17m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8baf781524a7432eb26b281f6f2e7b0e : SUCCESS in 21m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/611dfe11947d40f9a5b43e2c50c54af0 : SUCCESS in 12m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/be5dd70b7b3e4a93a9d5ee9345dd3998 : SUCCESS in 5m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f2d8f71ee9894d198c48635a64e00c7c : SUCCESS in 3m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5ae005fffbac42288739ffd46f0a2098 : SUCCESS in 8m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9561bcad2a88450ab391feef2fcf1a78 : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bbccd1c6245f4aff8a0813311328e8ba : SUCCESS in 6m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/50670c37c7314e4ca2f4b14965df17b0 : SUCCESS in 9m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/037cfaaaf49d4624acd4d5c9a29b28af : SUCCESS in 14m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f013c7e5d7c641ba85bd59dc714262e2 : SUCCESS in 3m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d4356ef719f34b178742ac6a1ceefbaa : SUCCESS in 11m 59s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7e8db65b40d64635916be0fc124332f3 : SUCCESS in 17m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0e9e101b45d842ca8f190d46d8ffb11c : SUCCESS in 7m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/794ea3297e8e488786b85dd51b94e0a4 : SUCCESS in 14m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0743efa6130b4c66964835463d832cc9 : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4529c723a5d47f293174222a55a41e5 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44d7db866ed74260b585630cdf476420 : SUCCESS in 27m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8a8ca97d53df4813b654fb2683d71784 : SUCCESS in 6m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c5642da6f9f84990afad26cd4ccd40cb : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3db2f2e59c34f54af957e04f1226190 : SUCCESS in 7m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1e43a1895dcb4731af14bf6841e8e4ae : SUCCESS in 16m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce340e44a3ad48e3b00afa032c6672d7 : SUCCESS in 1h 11m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bed084c9fc1e47a1bfa021185dec0c42 : SUCCESS in 1h 37m 26s","accounts_in_message":[],"_revision_number":4},{"id":"422e5a636889560b18870fb2fded7b8362cb7ee8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 17:25:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b7992173f898395cf6b926b4b35329d5cf9be570","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 17:26:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83a6bbf5752d43b79cbbbd8b1cb6bc57\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a74504bca9f4c38b73d1023c0b555c8 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f7314c1b3451e3e7e354c511f3a3d1a0d81aa168","revisions":{"79dec68df3a3b7df0847419e3dbcd4e9a9059cef":{"kind":"REWORK","_number":1,"created":"2025-02-01 08:09:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/940569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/940569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/940569/1"}}},"commit":{"parents":[{"commit":"fd14b6087bc74a43dbadea5d476ad769280ce2c1","subject":"Merge \"update constraint for oslo.metrics to new release 0.10.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd14b6087bc74a43dbadea5d476ad769280ce2c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-01 08:09:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-01 08:09:49.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I34a126282b4131c8167f08c925734c4b4251bfb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79dec68df3a3b7df0847419e3dbcd4e9a9059cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79dec68df3a3b7df0847419e3dbcd4e9a9059cef"}]},"branch":"refs/heads/master"},"96e71637ee6e3c005f7fadc2779b965e73b2360a":{"kind":"REWORK","_number":2,"created":"2025-02-08 08:14:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/940569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/940569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/940569/2"}}},"commit":{"parents":[{"commit":"e223942ac7c8218b3a39022432e62117d370d0a3","subject":"Merge \"Remove purestorage library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e223942ac7c8218b3a39022432e62117d370d0a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-08 08:14:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-08 08:14:57.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I34a126282b4131c8167f08c925734c4b4251bfb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96e71637ee6e3c005f7fadc2779b965e73b2360a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96e71637ee6e3c005f7fadc2779b965e73b2360a"}]},"branch":"refs/heads/master"},"6cd382e413e2498adf182aecd2b2db12b6cbcbb1":{"kind":"REWORK","_number":3,"created":"2025-02-15 08:11:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/940569/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/940569/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/940569/3"}}},"commit":{"parents":[{"commit":"94d8f03e56405fa4671aaf3b291baced32475d50","subject":"Merge \"update constraint for oslo.utils to new release 8.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94d8f03e56405fa4671aaf3b291baced32475d50"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-15 08:11:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-15 08:11:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I34a126282b4131c8167f08c925734c4b4251bfb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cd382e413e2498adf182aecd2b2db12b6cbcbb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cd382e413e2498adf182aecd2b2db12b6cbcbb1"}]},"branch":"refs/heads/master"},"f7314c1b3451e3e7e354c511f3a3d1a0d81aa168":{"kind":"REWORK","_number":4,"created":"2025-02-19 12:19:35.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/69/940569/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/940569/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/940569/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/940569/4"}}},"commit":{"parents":[{"commit":"94d8f03e56405fa4671aaf3b291baced32475d50","subject":"Merge \"update constraint for oslo.utils to new release 8.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94d8f03e56405fa4671aaf3b291baced32475d50"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-15 08:11:12.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-02-19 12:17:14.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- pytest\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I34a126282b4131c8167f08c925734c4b4251bfb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7314c1b3451e3e7e354c511f3a3d1a0d81aa168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7314c1b3451e3e7e354c511f3a3d1a0d81aa168"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
