)]}'
{"id":"openstack%2Frequirements~940864","triplet_id":"openstack%2Frequirements~master~I5e4c999d0b4c2a1b3d724ebfc55b94eaa781d1c0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5e4c999d0b4c2a1b3d724ebfc55b94eaa781d1c0","subject":"update constraint for mistral-lib to new release 3.3.0","status":"MERGED","created":"2025-02-06 09:57:17.000000000","updated":"2025-02-11 22:11:18.000000000","submitted":"2025-02-11 22:08:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940864-new-release","meta_rev_id":"4e5e969884c4497b7c090196d32ac75331379a89","_number":940864,"virtual_id_number":940864,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 22:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2025-02-11 20:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2025-02-11 20:21:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-06 11:23:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-11 09:39:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-02-11 20:21:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f22de6f8c7b17360457b514f2b56809244b571cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-06 09:57:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75e8a42e0d3f0c77dfaadbb5659555c4096fbf8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 11:23:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02061b225a6f4fdfa9656acf7e88752a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c929cc57daa841ca90dc973a8a7978e9 : SUCCESS in 16m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15ee446e024747a796b9d6b060968f56 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9efbbc1dee34252a9090c3a1303ccd0 : SUCCESS in 15m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/474671989e904deeb82a26e37c26d035 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30fb6637d6b6436a9750ba4f6ef98d9c : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e1cb64beb3e8469eaa9c06a84a4d8a52 : SUCCESS in 4m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5f0395daeb6643b39eaa262d5aa97948 : TIMED_OUT in 40m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/37da793370cf477785b70697f209e02c : SUCCESS in 9m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c49beb768f4141b39a33772c57dc359d : SUCCESS in 9m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d9a441a7007c46e9aa0ca72820c15be4 : SUCCESS in 25m 30s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/17b05afd545648fba57536fffea015be : SUCCESS in 7m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1abb3c7ff824437c8fcf1bdf9a92d163 : SUCCESS in 5m 08s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/021c8d39077843f8950a3a4171b6e172 : SUCCESS in 6m 18s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f3d89c976a8047c28a37c656083c227d : SUCCESS in 12m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f2e30ecd064b4aeda72e06984570b3e6 : SUCCESS in 6m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fe7ab4cfceed49afb4ae8b28485ec44f : SUCCESS in 4m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bb1af345ccc34de4b9eb40634807d0c9 : SUCCESS in 23m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0000c57064eb4fe59bcc9de9a25acf23 : SUCCESS in 13m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/081c333789bb4a5cbbf469c7196f9f28 : SUCCESS in 38m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3792a72f1d684f58a5ddc79e37316153 : SUCCESS in 9m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a7f2c6feaaa6426a8755e93441fa71b1 : SUCCESS in 13m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/19340c8bb09c40b9a5a6c46bbcf72361 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a812f0a531fb4f15a4090ca5dd8ba0d5 : SUCCESS in 10m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0df7fbe4a8674c64a18a0253b228f6d9 : SUCCESS in 4m 42s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/fa8073200283472097d8727712fd194a : SUCCESS in 5m 17s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f54143e6b21f4ad38f4ba35d05155622 : SUCCESS in 10m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5ad1013e26d248f0b0d470beb7b58ffc : SUCCESS in 18m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/033c3683f08045c49dd076e9c04fe861 : SUCCESS in 25m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ba2024148e5041c3b53c4adec4f66fc8 : SUCCESS in 15m 33s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8e63593a27b845bc9c8fd4136fe12713 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3817992c94ef432a98a3cd25b3fcf170 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd55531835d64f338d71cfb98e34c3c7 : SUCCESS in 40m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/96166f4e4adc470bb5242147d50795a9 : SUCCESS in 6m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a7629429a030402fb521c07f2b4fb415 : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f655ba55446448118a1733d7fe44dfd9 : SUCCESS in 4m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bc49f04a44b54e78b9f0e3fec9ee5c1d : SUCCESS in 18m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7248d1186c1f4c7295af5a3bfb645e68 : SUCCESS in 1h 10m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3584f89f558a45e792156e2b14da955d : SUCCESS in 1h 24m 12s","accounts_in_message":[],"_revision_number":1},{"id":"26337ddb290fe5324ef876639463997145c785fa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-11 09:39:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77447c8c0530d17996dfcba3f0a7353500168895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:32:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56cbd7f6a3648ce80c3b6b690f8fd09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82438b508ffb42c98de9a222c5b92af6 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533991e13a724a3bb79dca153a04ff6a : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6aa46cf75c934292b9ca1c1e23e6b135 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/faac4b0f2ebe41cda8e03608bda9d1cb : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83ec7f4bf2dc499b959ca349c0defea3 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45d0a24f56164260b264e2debaf33640 : SUCCESS in 3m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8c7162a0f14143e9b6e5bc23c3dc4fc4 : SUCCESS in 16m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4a63a76086fa493caf0cecf0b0dcbc3b : SUCCESS in 10m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5bebd8983bdd426ca579cc5679b849b9 : SUCCESS in 15m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ba8bf6f000d0428cab35663ed40ff374 : SUCCESS in 10m 39s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/04b50a9c7c3244e08b741553bc8d661e : SUCCESS in 5m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1f11f4d00f5d437f80be13fc28aecbf0 : SUCCESS in 8m 56s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7546ffc9b5504ed897d3983b5e9ce1dc : SUCCESS in 4m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a6e2de9e487b4c9f800e4b7163ac1ffa : SUCCESS in 10m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0790b193ab8b4f7da17a775d54807630 : SUCCESS in 3m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cd22ff6bba16455692438d9e745b753c : SUCCESS in 4m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5399c49a0946422c9561ef72bee0be58 : SUCCESS in 10m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/030c75e90d884f09934dcbc637e04846 : SUCCESS in 13m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e1d6ffef7494ca7807c7a111de30dd0 : SUCCESS in 8m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/367c99a5354f48ddbcae559799134b09 : SUCCESS in 8m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9f80a990c117465d92315ac8381ed39e : SUCCESS in 11m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a787abb42b414dcf91732bec2ad67377 : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/82c43cfba3cb4b5e8c5a5f20fee8a968 : SUCCESS in 10m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/303298594eb247009f3a8e57e99d21d2 : SUCCESS in 5m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bc04faf05d284817b61962a7a4f5af3f : SUCCESS in 13m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0d1c405359ae4c218c2ecd467a6fc722 : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c2ada9b4f63f4c0081e6c871d0dca6e5 : SUCCESS in 26m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0ea7a107e1a44da298bd584dee9ff7d0 : SUCCESS in 9m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4801c060d7994974a1842f2fc80e89fe : SUCCESS in 10m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/95dd0ce541f44bc99c2153c3e0fd9648 : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc5455eed8404e6d83744d253b49ffb3 : SUCCESS in 6m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8331eb27247049ee980e5484f2b52419 : SUCCESS in 23m 38s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/836feab11b254f89bcde09e23fc9827d : SUCCESS in 7m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ffaeb4ce8467436da1fe65740f9dbb1f : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04186694489f490fbfb1bf3d009773de : SUCCESS in 6m 44s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ca46c4b3018049cb97a5ba6e7314033c : SUCCESS in 13m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/222492d896b54250af5450070b93a677 : SUCCESS in 52m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14101670fab040c49a24309b15253beb : SUCCESS in 1h 28m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f5016289a709f05bab999b60c8120d478570a419","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2025-02-11 20:21:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"688d07107a1bcc166d699cdb18f897afe52f0738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 20:23:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b9c650c4e6ce81ce19cbc39085e1950ee675517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 22:08:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77923a95236a4ac89e5992c178024695\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f452007e40704a2cb233be9a3953c9ad : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea3c75f66b9347389c4a1d920171dea0 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e72e96328834bc6b6415b0e3b54c8b6 : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79394e35592f4df08ed13880df2ecb99 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d54d677b7e44bdea19df27a7d9f1835 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a6681614a2943aca05cbc8f795b0a4a : SUCCESS in 4m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bb79aeb76eee43cbb80c59fff01b173a : SUCCESS in 16m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/605b2c27c4d74373b934b56f5029fceb : SUCCESS in 8m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a8dfad7fa8ca4386a70bed281c0ddb9a : SUCCESS in 17m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/576118d42f03469da18e0a2ac4dcaea6 : SUCCESS in 20m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/68fb50127b2a44cab9748a9675f96bb5 : SUCCESS in 9m 22s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0545eb97b94b47e0a45d05fe8ecde968 : SUCCESS in 6m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ce7b73f84ab44dd8a78e5fa135b2e97b : SUCCESS in 4m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d63d25a0ace54b82a2500c8aa00ba002 : SUCCESS in 9m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/013a297b236447e0a023389ee6d6e3b6 : SUCCESS in 6m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ea8f3aaca0246c28cccf92241003e85 : SUCCESS in 8m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ee44d11a985b4530a28662e2ba862f8c : SUCCESS in 8m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ca1b5f0ee2584fc1995870249b80ec44 : SUCCESS in 11m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2f8c18dcdbaa4e929cf15439df796a74 : SUCCESS in 3m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ef983ed9402a4fd9bca990b4cea98c41 : SUCCESS in 10m 39s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6aadf9425a544e268eff93abade0d3b1 : SUCCESS in 19m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/569849dc645c4e4c9f39bedc0853a3bf : SUCCESS in 7m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4bfd15a813e143ee9404dca127a67bc0 : SUCCESS in 15m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7860faf434d54ce38511fd85c8fe4c6f : SUCCESS in 6m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c0433b955008457c90749cb969ff85e9 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5cc881eda9a342819a946ff949de0eef : SUCCESS in 26m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3ef5ce140265421fac9cdd9e6e51d527 : SUCCESS in 6m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e614e5e6282a4438ba25b53542fbcabe : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50b39d2fb0834f9fae3cfcc59c973c66 : SUCCESS in 5m 42s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9479cfb7d6ce4900b2e40cc0dceb7e16 : SUCCESS in 16m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82cf70d9fe9f410592d6876992a171e0 : SUCCESS in 1h 10m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c47ffbb093db4b58b5767e1d73bd7f96 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a9e44a1b50d90b2ee55e2420e5a43365979fae1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 22:08:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4e5e969884c4497b7c090196d32ac75331379a89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 22:11:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc64af75cb4547ababa241add362c2fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47ac628148c4ad98641a8c977750c90 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ffa524771ff094cab62e211c1fc3b80e12e62f5","revisions":{"5ffa524771ff094cab62e211c1fc3b80e12e62f5":{"kind":"REWORK","_number":1,"created":"2025-02-06 09:57:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/940864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/940864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/940864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/940864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/940864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/940864/1"}}},"commit":{"parents":[{"commit":"bf35b87d9707d4989850662dbda6e53b9012689f","subject":"update constraint for mistral-lib to new release 3.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf35b87d9707d4989850662dbda6e53b9012689f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-06 09:57:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-06 09:57:14.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 3.3.0","message":"update constraint for mistral-lib to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: team: mistral\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I6de76f2a9a7d4e9af4822b85229b53bdbf58f117\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\nmeta: release:Code-Review+1: Axel Vanzaghi \u003cavanzaghi.osf@axellink.fr\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I5e4c999d0b4c2a1b3d724ebfc55b94eaa781d1c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ffa524771ff094cab62e211c1fc3b80e12e62f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ffa524771ff094cab62e211c1fc3b80e12e62f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
