)]}'
{"id":"openstack%2Frequirements~941547","triplet_id":"openstack%2Frequirements~master~Icbe776c7d3a01bb8927cc047a4cb91951f8a26d1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icbe776c7d3a01bb8927cc047a4cb91951f8a26d1","subject":"update constraint for oslo.rootwrap to new release 7.5.0","status":"MERGED","created":"2025-02-13 12:02:16.000000000","updated":"2025-02-13 20:32:52.000000000","submitted":"2025-02-13 20:31:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941547-new-release","meta_rev_id":"420544e635ff1d1bc10164087938d4439ca48ea3","_number":941547,"virtual_id_number":941547,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-13 20:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2025-02-13 16:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2025-02-13 16:41:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 13:40:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-13 16:41:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f4418204346bb64cf77a6b53838c0bcfe3ba97da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-13 12:02:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51610e63811d5d4e5e087601efce4263fb8058ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 13:40:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6c3d9ec5fe1414abb03d174cd0eaf66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de908de029324e0db12f01451536341b : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e783ba2381490dba4fcc6b0bd39823 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfa9cdd4bd8f420792037935878f428e : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c4324aacf4f4c33908122221cef4853 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78f8036a18764800b3f817f01a389fb4 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f76c4be10a10470a9fe5ddab0d84930f : SUCCESS in 4m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d7e38eee374c4f708405d2c61bc1d58c : SUCCESS in 8m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a4f4b7a73ad54716a0f9d9c57f164fcd : SUCCESS in 8m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0a9d7d14c4a44860973f86c04bbfe51b : SUCCESS in 16m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/48c38a8084914e2c831d824ad43ce673 : SUCCESS in 10m 00s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6e23d38c445746f0a9b12c404bef1713 : SUCCESS in 5m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1d981a3acc474fb6baeba30a63841f7d : SUCCESS in 4m 56s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2b8d3e0fa7744ca9a1b053e4f9d21556 : SUCCESS in 3m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ebf51e52ae4f46318a93de96b6d68be3 : SUCCESS in 10m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3e6b6d91d09f4df79070aebf2d0085b0 : SUCCESS in 4m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/298837b98e2e45efb2efb1b8e8b1e015 : SUCCESS in 5m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7d17ad0cf05c463b981c64ff64f77eae : SUCCESS in 9m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/98f4b6a07ca844c6af869bdbe10bc4de : SUCCESS in 14m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f4b550debd804e06b3ee557f69d33ec3 : SUCCESS in 8m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a838b713ba034f0a93de7085cbbad438 : SUCCESS in 7m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a8fc0bebce4149aa9f61668a8934b300 : SUCCESS in 13m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b3b206a4d4cf4c0aa9a43d7ad0e98e6d : SUCCESS in 1m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/73416a1cc02148418498b67398a6bff0 : SUCCESS in 11m 37s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e2d4880b5f124732b244e21cbe8dc7f9 : SUCCESS in 6m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6f0a388de5284a768b2a8a26a52eb966 : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c02cacea7a2b49e5890fde1291fa159c : SUCCESS in 10m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/711de8b2b1804f3ab29f319078cc7288 : SUCCESS in 21m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4074f09ccf8c4271848ef75e67677148 : SUCCESS in 7m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/61e9fb17c597400482b61dc9258205c5 : SUCCESS in 10m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1fa6fc2f78714ced99ab4d783e26c2c5 : SUCCESS in 7m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a2785ea37304eac8a1ed6b82e696987 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1945de7acd944c269e49c7d578d3f61c : SUCCESS in 23m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/985eac3e0e5e4184b8754c487a466b94 : SUCCESS in 4m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/94f953cbd26c4e7fa84b1e78fe3eea85 : SUCCESS in 6m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c5e9ca7e71be42b18341c2af203b954a : SUCCESS in 5m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/91b4050afe474de89d2501a45dc9ac56 : SUCCESS in 18m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73282ef37cec4e26b661cd6f8febe94a : SUCCESS in 1h 06m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7388ec70b6e4571a6e6d79b9901cb27 : SUCCESS in 1h 36m 03s","accounts_in_message":[],"_revision_number":1},{"id":"55a88cf30ee037b756210c514041c92c0e0a9de3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2025-02-13 16:41:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27e5cc57afc56c844df077412aaa9920ab03ec3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:42:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"798ea397dfcdefcd4d7e5b49b34f586eeda5372f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:31:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73950673e8224dfb8786a4cd9e89487c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ceea04ac414990be4985d735a66775 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d11b2c20974e0d8dbee551c1d9460f : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b895d2c8f2fd441296f971c9d58812c9 : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85206ee25dde4b46a2ea9c80252b0e36 : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c3b0dbd1d784092be4123e6660152d7 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/62ff7469e86b4cfd952048ae3cb09c44 : SUCCESS in 5m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f40af58eb7164ad7b6c3d6a1f14979e7 : SUCCESS in 19m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b9c42293afd84369b7412f1d4f7f5a76 : SUCCESS in 6m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7bb793ac747543dabc4a299c5a8bffff : SUCCESS in 13m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/70b2b102475d421380773260332dcd7f : SUCCESS in 23m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e491b930b4eb4e85b652f8a6774ae789 : SUCCESS in 15m 14s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/198d830ec8dd43eaa831e2a7f2945c58 : SUCCESS in 4m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b5fd5cbd720f445a9c2bacc414a8676d : SUCCESS in 6m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1c1311ea9a24493b9333dc63c1ed94d4 : SUCCESS in 7m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f27618a856e049f6ab79475fcee5b4ae : SUCCESS in 13m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21ef6eafd93f40b18ab269d2d852a324 : SUCCESS in 9m 10s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c6b7394bc68c41178ac5201fc7dea787 : SUCCESS in 8m 00s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bfd796724b4e4948a932cb28906f18a4 : SUCCESS in 13m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fd4f17e1c35648ff895cbeb31c8fb939 : SUCCESS in 2m 44s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/537c845be224499b959a3b53bc841d5e : SUCCESS in 9m 00s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/789ff70fb1cf44c9a33bf39cd15c8120 : SUCCESS in 16m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc455ce3a6154054a91bfcd9c1a98b8d : SUCCESS in 10m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f17042c1ce154c0ea25e85e8ad548e6d : SUCCESS in 14m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/adfe64499cfb4b2db580a00453fe7db7 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4933836f09e744a281484f8aa39199cd : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1dd4c5af291741d2ae9ddeeb55678223 : SUCCESS in 28m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/279ffbbd6d1a4790b226887f34e5a697 : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4db203f136d74b01b052a781a446a661 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08c21fc0a6e942c18e7d13c221da4c5b : SUCCESS in 5m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/afdf2349017a4abda06c2fab3ecb314c : SUCCESS in 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0618d2192b09473ba92cdb91119ba316 : SUCCESS in 1h 11m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dde7b5b81d8a47e0abdb46ffa08b2cb2 : SUCCESS in 2h 02m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1c9ac6c3714a37841e08c32bfc64c52baa20d56e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:31:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"420544e635ff1d1bc10164087938d4439ca48ea3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:32:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/869e5a69deef47dc9d6479862542ee15\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c6c4fa0d484817b38c1aaff281e141 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03dae536be4310f79a2b988512df6cc46e846017","revisions":{"03dae536be4310f79a2b988512df6cc46e846017":{"kind":"REWORK","_number":1,"created":"2025-02-13 12:02:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/941547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/941547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/941547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/941547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/941547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/941547/1"}}},"commit":{"parents":[{"commit":"4eca7bdc5077b3e1e83f74535743f8427adff6a0","subject":"Merge \"update constraint for mistral-lib to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4eca7bdc5077b3e1e83f74535743f8427adff6a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-13 12:02:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-13 12:02:14.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 7.5.0","message":"update constraint for oslo.rootwrap to new release 7.5.0\n\nmeta: version: 7.5.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie6fd88dc341c1c9abf18dd3cf6aaca80ead10500\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Icbe776c7d3a01bb8927cc047a4cb91951f8a26d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03dae536be4310f79a2b988512df6cc46e846017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03dae536be4310f79a2b988512df6cc46e846017"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
