)]}'
{"id":"openstack%2Frequirements~941548","triplet_id":"openstack%2Frequirements~master~Ib9e34665fbc11019e65bd12be937707d971c693e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib9e34665fbc11019e65bd12be937707d971c693e","subject":"update constraint for oslo.cache to new release 3.10.0","status":"MERGED","created":"2025-02-13 12:05:13.000000000","updated":"2025-02-13 20:31:56.000000000","submitted":"2025-02-13 20:31:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941548-new-release","meta_rev_id":"ef4aeb8ffe1085b734c07522bdd36d8d2246f603","_number":941548,"virtual_id_number":941548,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-13 20:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2025-02-13 16:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2025-02-13 16:41:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 13:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-13 16:41:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9dfea4ca88590fb35abbb5778a3ab27292b72803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-13 12:05:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"230e029fb659c2ae62df132d793931782401a0a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 13:57:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a651056f149c4a6e9819d075bd1c3796\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/620a072dac614d5799e0ca746da09606 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9edd84e8ca044ceabc7fa21a8f24c815 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/763f55ae1dd74f2c978bd331f5dbdc13 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/411fcdd043e4463b98ad4fbfc49d4977 : SUCCESS in 2m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a709471b99d44e4783ac7c8d06f534e4 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f281e23d8f0e4b479e638f80122b1d9b : SUCCESS in 3m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fa7e3f01c56d45d4845a99910e0f4e97 : SUCCESS in 9m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/41cb65f4b93f4fc780bf8d71c91e5478 : SUCCESS in 8m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a454506b201045a7b3ba118fd41b9bb0 : SUCCESS in 15m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/17abb4aef82e44499a7863ae6572d3fd : SUCCESS in 27m 50s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/87af7389e8c0415f9885904e18a8f059 : SUCCESS in 5m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e6bb7f115d894dda86c1db343d161bb7 : SUCCESS in 5m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/275a1826b2e8426191238bf0e48d3c19 : SUCCESS in 5m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7a19a76dd9f84a299fff5616812c811a : SUCCESS in 10m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f6784193b3ab4960b37ea47a013ebd16 : SUCCESS in 6m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fc32a021e0cf49da9273e8cf50adfd31 : SUCCESS in 6m 51s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/248b2a2228a546c5bf5a692c0b4f6183 : SUCCESS in 7m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/36e753cdd3b740e9a8e5f48064478da2 : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98b1f4a84e224166b0b7c86e2834246b : SUCCESS in 6m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/36b014e3e7104550bcd40c555e23ce6e : SUCCESS in 7m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/175c7ef194b54840895da64f87890a90 : SUCCESS in 11m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0db3f981c8854889b4e1ff3e9642bbdb : SUCCESS in 3m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9d52988298b444d488aa27ad9e613c95 : SUCCESS in 13m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ab9f2bdf215041cf9ccf9b2a4798f701 : SUCCESS in 5m 00s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f986b98b916b487b91b7064ed901aa45 : SUCCESS in 6m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ae647a79eb1440f2bbbff141bc819c43 : SUCCESS in 11m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2d318a8160a44e04bd52b61bf4cfb20d : SUCCESS in 19m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/42ddc4b4951d4cf7b949e0d478c00d86 : SUCCESS in 8m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4f37f2c7ed1647a3b15251ea57e6f285 : SUCCESS in 16m 10s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/41be8e3e271c482cb3a182d5be50ffe8 : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b440004c70b140d7b11a87234a16ae7a : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/918c8b41287b43da96640d200b542b00 : SUCCESS in 26m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ea773ad92af440bbad1fa092e4c90e54 : SUCCESS in 7m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6e02babc768146a0b3f84a701ab4485e : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9153929196784d68a849cc8d9694268a : SUCCESS in 5m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/db52ed6dd1b0428c86812f2ab9fed02c : SUCCESS in 17m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d10d2841b1954aa8886d3763f8c62a4b : SUCCESS in 45m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52f4cd4d99f34f14acf80817482f5059 : SUCCESS in 1h 46m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ec908f114d734e448f93ef51ca0a559362b943c7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2025-02-13 16:41:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c941d4a9673bfbee241f998cf8ff0aac87c2152","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:42:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01a7d3fb85a4e26069c7e7652a98b9e57ff290d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:31:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c7ee10de2514185bdbf874e4c4bfd7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11015cb1933c46bc9d94d4f8d166a6e9 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ff4676f9c841deaacbc471cc7829a6 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fc87d45f7f0494eae59e2531cc6a128 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b45b031fd7f4654b747eca3ff2d535b : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e22096bc82db4171ba621020bcf321e8 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12b735cefaee4bf7a63b7ca4fed2b792 : SUCCESS in 4m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2e309703030b40429d1b7e3e4b1bbebc : SUCCESS in 18m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2240bf77edb644b8986e26237c0199fe : SUCCESS in 8m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4763ba0fa10b49e68c7e589876f57aff : SUCCESS in 17m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7e625d1d820f4cfdb52a72f5955910de : SUCCESS in 23m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cd4ab5f411cc489291f2b1691bd77503 : SUCCESS in 13m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d13f4e008448493c9a9860223f858bd2 : SUCCESS in 5m 34s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/13d0788fd97d4ef7ba38bcdba067e2c0 : SUCCESS in 8m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6c16608d890443a597cda9bb731dc059 : SUCCESS in 7m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3841159c7d744b0784eadfe5420f247a : SUCCESS in 13m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e089deb5afb74652be368327677845e0 : SUCCESS in 10m 23s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d16a61c3efdf41b8bdbe94fd80557532 : SUCCESS in 8m 32s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7820be3f173a4d489cbb8c83dbdf37a4 : SUCCESS in 12m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e6d8157638ef4825ba78a8171341a93c : SUCCESS in 3m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f5443d3656ce4ee1b33ae110b49ea6de : SUCCESS in 11m 27s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/683c83cefdaa48f39a49289542ce7ae2 : SUCCESS in 18m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d4603f03c2aa49ddab079ad13844bea8 : SUCCESS in 10m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4497c69051244a6892ffae22bd7b4045 : SUCCESS in 15m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/56f142c6f6f7467e931864e026c74d84 : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c04d4199ede94628bead60ae6350180d : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9102081550e42a1bd50304298eee669 : SUCCESS in 27m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/458835d4ee7549cf8d90ced26ab57c57 : SUCCESS in 6m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/24870f5fb2664310b5e5b1ce44d3b9db : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/769ec7d96ea24c5a9af97b3ce4124bbd : SUCCESS in 5m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8610c9037b784489bfcfa4f190971e3c : SUCCESS in 18m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6bc29e2b66d4e70b5db232e8035e78d : SUCCESS in 1h 13m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eeac1b200c184e99a71708cefb943f62 : SUCCESS in 1h 56m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ef4aeb8ffe1085b734c07522bdd36d8d2246f603","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:31:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b38abd382978a852019d4e91ce70250c7624b38a","revisions":{"b38abd382978a852019d4e91ce70250c7624b38a":{"kind":"REWORK","_number":1,"created":"2025-02-13 12:05:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/941548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/941548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/941548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/941548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/941548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/941548/1"}}},"commit":{"parents":[{"commit":"4eca7bdc5077b3e1e83f74535743f8427adff6a0","subject":"Merge \"update constraint for mistral-lib to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4eca7bdc5077b3e1e83f74535743f8427adff6a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-13 12:05:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-13 12:05:12.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 3.10.0","message":"update constraint for oslo.cache to new release 3.10.0\n\nmeta: version: 3.10.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic9210e37383dbd9e1f89d7bb66de1c16b8709062\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib9e34665fbc11019e65bd12be937707d971c693e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b38abd382978a852019d4e91ce70250c7624b38a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b38abd382978a852019d4e91ce70250c7624b38a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
