)]}'
{"id":"openstack%2Frequirements~941549","triplet_id":"openstack%2Frequirements~master~I35b5c7666da111ae7ccdda118aacd87825930edb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I35b5c7666da111ae7ccdda118aacd87825930edb","subject":"update constraint for oslo.utils to new release 8.1.0","status":"MERGED","created":"2025-02-13 12:09:18.000000000","updated":"2025-02-13 20:33:33.000000000","submitted":"2025-02-13 20:31:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941549-new-release","meta_rev_id":"30a23a5f0523197e7cfa2f863c801c9a05f8b2dd","_number":941549,"virtual_id_number":941549,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-13 20:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2025-02-13 16:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2025-02-13 16:41:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 13:56:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-13 16:41:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2393e45abde05cd2704dcf9787c564f4d09c7def","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-13 12:09:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82bcf9f1cf387014ad2c8c50e11971291335b50b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 13:56:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7797fb12eb224558bd59233b3a64d469\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582f1e3ecc0d4b32b9ad5027d424c762 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28684f9d3f044d2b721a8c12408969a : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c7298837dda46ffb0d0c768425c836f : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34eb7a6b39804594a4c3212b9d678430 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/273badc34dc54aadad69d8ce26d84fea : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c061ff39f164df7a61036fe725e4f71 : SUCCESS in 3m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e37a0c12f60d4de89542ee07f1de9dbf : SUCCESS in 9m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bd89ee6ab92f431bb07a1aa77965c679 : SUCCESS in 7m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cbd14957dcdf4f1e8a692301853fd5c7 : SUCCESS in 10m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7a0146c1898841409e6dc56fcfcbeb60 : SUCCESS in 20m 02s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/37f4870cd47c48338e4ceece29c1f079 : SUCCESS in 4m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/144b8001b3ca4a939efae4839cda1ba2 : SUCCESS in 5m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1cddd45cd5fd47568dc4a0fcb25de591 : SUCCESS in 3m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8498b4c835d942d3be3771e72dcf0032 : SUCCESS in 11m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3d739ddbab0e48da8acaaf6a2229c511 : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f0783159c43d4641b4cc35c94b241352 : SUCCESS in 5m 11s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0afd39f343fc4d9cb47c1f67d3a7ffad : SUCCESS in 7m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/eae86b2da634433bacc8aba6cb4bada6 : SUCCESS in 14m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56bc27f3bfb842668b08e513f642cbf2 : SUCCESS in 7m 20s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/002633949c7d47a29106cb3482b703b4 : SUCCESS in 7m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/89519652172b412f8961ac3807f85c29 : SUCCESS in 14m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/98a1aa47b51643078bbe8e65dbe033b9 : SUCCESS in 2m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/80dc71ef6a7b40bd907f7d7749b4e647 : SUCCESS in 11m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b8b1b8c06ed14bcfb03fccb610f616c6 : SUCCESS in 3m 58s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/641bca1968c64026a1b890dfe8daff6c : SUCCESS in 6m 02s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/18fedf8a81e64d7783a06e23a4c65d3b : SUCCESS in 13m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d21f699ac9b34f53a77ecc098fb24dea : SUCCESS in 20m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d34f042d52b44b2e9a1cc81f7d1f6f64 : SUCCESS in 7m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e5bf9e172abb4609b608156ae8ca794f : SUCCESS in 13m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/840e854df7614c5bae94c512a7869db4 : SUCCESS in 6m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0c9564ad44814d108b6e9a86bf48652a : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a4d2f54435db4b8b9a83abc85e75366e : SUCCESS in 25m 46s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/017cc00611ea473f898fc475cd46b6db : SUCCESS in 6m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3cbaa8e03299486aaa810d4565a95f48 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/283e1c8cb49d46e0b5036c1f14facd1d : SUCCESS in 5m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7e60342598db4cf6ab70a94030e6eeff : SUCCESS in 17m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a88138532c240ef9b76cf88b415e680 : SUCCESS in 1h 05m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4124e93652a14ccf8dab3719c735d752 : SUCCESS in 1h 41m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f9af1eae122cfb58537c13cfc57a89c2124e9809","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2025-02-13 16:41:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b4fa96538bea88f4723d2066ff5ed730d3089f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:42:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"068cb4caf0a344b63df3a41b6ab2c12dc5e236bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:31:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34ed39c8c1c043ba898ee7cb4d3eaa4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e48f691badd4aef896d1771d5b4b743 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05b4f8bd9387477c9dd4a544ffbd2338 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/399f85dce3b741b7b5631e615018de12 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/266bec9908ee4273900b8ea9161ed172 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc86aec8805b4238b39cfe92032a0e2e : SUCCESS in 5m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd2b8ad897754c7fb37e055781969f56 : SUCCESS in 3m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/76fe35c2723d4996abbf1deba6425233 : SUCCESS in 15m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f6ee6de59a04f568e293b30d37d7133 : SUCCESS in 12m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/14f066c5a66c4d0189e7c2ad4f797b6c : SUCCESS in 14m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/079096ee274b42b497ae43d44cbb132c : SUCCESS in 21m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/89e39525ff18410aae6a019c12dfd5c6 : SUCCESS in 13m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a596dfd06f5843699889c3d7c1e444d4 : SUCCESS in 5m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d54234b6627a42f0a5eaa0291910a3c8 : SUCCESS in 6m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c9ca4a9ea66848a0961346b7a39f78f0 : SUCCESS in 10m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f8b3d33f59f74f139d9ffc55e4bee692 : SUCCESS in 13m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a81db8c995784165a0561bf227e49972 : SUCCESS in 6m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2f9a9026cb8c4d5a8da2462b7074abcc : SUCCESS in 7m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/992f699cf50e4f37af01682e9eca5ab2 : SUCCESS in 14m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/aca829aa25494b528ca1517a5eb07350 : SUCCESS in 2m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/026d103210ff40d2bc2b3a7cf092acd2 : SUCCESS in 9m 33s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b4791c092ac24cef8f74b8206fe2c13e : SUCCESS in 24m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/29f4b4f06b704c618fd91adcb8930e7d : SUCCESS in 6m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/04c1205e6ec74b708598e6f135d07b4f : SUCCESS in 14m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/27258452dbb94ad5a428ae8a12484d73 : SUCCESS in 7m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f39a1ae47c1649b8bd9b7d66b912aaf9 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad0c6dc370e94382a273c46c5a3eb199 : SUCCESS in 27m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2d5c48e2c89d419db8abb12d35538645 : SUCCESS in 7m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/07c7694967544291b7b6d8e8c57c8192 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2daa13b54b86479192f3e469fadba1c7 : SUCCESS in 6m 07s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a90d875df5c84174b8a33e625d7a899a : SUCCESS in 18m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a11b598bb17d44db8ec1de4856588c69 : SUCCESS in 59m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5072bf2b568a4d639dc62efa13e095dd : SUCCESS in 1h 54m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3cc146a8adbb248f2d1138f5cb047fb1202b3664","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:31:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"30a23a5f0523197e7cfa2f863c801c9a05f8b2dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:33:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e65d9deae15f410bb2494cc091d9c503\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88e709a4a44a42fdad4635b8ac36f599 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1dd0a941920ef7927684e65d2e8884d01253234b","revisions":{"1dd0a941920ef7927684e65d2e8884d01253234b":{"kind":"REWORK","_number":1,"created":"2025-02-13 12:09:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/941549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/941549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/941549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/941549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/941549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/941549/1"}}},"commit":{"parents":[{"commit":"4eca7bdc5077b3e1e83f74535743f8427adff6a0","subject":"Merge \"update constraint for mistral-lib to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4eca7bdc5077b3e1e83f74535743f8427adff6a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-13 12:09:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-13 12:09:16.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 8.1.0","message":"update constraint for oslo.utils to new release 8.1.0\n\nmeta: version: 8.1.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I98efc9132fb4ff7a2ac8cfc73123d55984a3b403\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I35b5c7666da111ae7ccdda118aacd87825930edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1dd0a941920ef7927684e65d2e8884d01253234b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1dd0a941920ef7927684e65d2e8884d01253234b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
