)]}'
{"id":"openstack%2Frequirements~941550","triplet_id":"openstack%2Frequirements~stable%2F2023.2~I6c62a433cd542af230fd37de0d24c1ce56615af6","project":"openstack/requirements","branch":"stable/2023.2","topic":"new-release","hashtags":[],"change_id":"I6c62a433cd542af230fd37de0d24c1ce56615af6","subject":"update constraint for openstacksdk to new release 1.5.1","status":"MERGED","created":"2025-02-13 12:21:41.000000000","updated":"2025-02-13 20:32:02.000000000","submitted":"2025-02-13 20:32:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941550-new-release","meta_rev_id":"73384fb745e9d59445e31c93c6d09ff24d7213a4","_number":941550,"virtual_id_number":941550,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-13 20:32:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2025-02-13 16:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2025-02-13 16:41:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 13:44:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-13 16:41:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"79df841750efcbeb2c65d75386e15f52422ad7e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-13 12:21:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b74bfeefb5fa5d65fdd30df3ab86792d0dd45e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 13:44:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/073526e24d0f4a39b0a95b74c130c0c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd975a6ac888431fa24dd0dfaebe6d67 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0151cd195a7f4480bad1ffa991ecd43b : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bffc12f4e82b494ead770ec92642b5e1 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/540bb49295ca40b7843741e17d436cfe : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5ecc5a1f62f4c7e95ef3ae4e07efb73 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f770cd6926b74d768a75cc1cf26ada4d : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/440d217d535941749063b03a3d49b818 : SUCCESS in 2m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/beaa81a908b54ceab74b9918ab58266f : SUCCESS in 9m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d528bc24405a44d3af9f2fceca59a15f : SUCCESS in 12m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d8a3fee4cd8d43ecb2a3cd958025cb07 : SUCCESS in 10m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a51e4939920f4ef6a216cfdb4e3f56d5 : SUCCESS in 12m 32s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/12517c1a0bdf45bfb3b81d917c629619 : SUCCESS in 6m 27s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b0a1959a61454498a2d8279935438918 : SUCCESS in 5m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/83f8a9edb132423d8c9490e4a0d17af7 : SUCCESS in 5m 29s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/aebe2933e3194b67a541bf3f658ee92e : SUCCESS in 11m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1afa9285348643bb8b4a1f4793de29ef : SUCCESS in 5m 06s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9b5fd8489e07430185ca14e63b3b92d6 : SUCCESS in 5m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2055d50fa935445da070eebe6d56a2e1 : SUCCESS in 7m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6994b2a6790f41cb99251b32a21469b0 : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40ba658d52f34cedad93b1da1adc05c9 : SUCCESS in 13m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d316198066ca4b86bb995af373f0a4e3 : SUCCESS in 7m 15s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1fa5e4e989164acb90ab4d46194ec040 : SUCCESS in 13m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/deeb8e929df24b0990da50dedcee1b2a : SUCCESS in 3m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/734ec8974dbc4771b813f018b98e25b2 : SUCCESS in 12m 38s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7f79c5b44361450caff550d0ec94ebb9 : SUCCESS in 5m 19s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c52495c450554617a20f62f6109a7810 : SUCCESS in 10m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b4cf8af884534aa3b7e9ba32323b083f : SUCCESS in 27m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6596a376716749c48151ed35b106f1b3 : SUCCESS in 11m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/043467bbbe0a43098b19615f0013b6d0 : SUCCESS in 12m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/36a477b2dbf5438cb0a69026a52f584f : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/caa518eddfa74f0db0bfb1130c655434 : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/949af66510714cb8b6f586b27e0c8679 : SUCCESS in 23m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/da203a0d5aef48fe83689d34951086f1 : SUCCESS in 4m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ff6d6159c3064c8190f2c018f3b141d7 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/148ffab0a92448e7bfc6c157b3e35974 : SUCCESS in 7m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/621f16fe216b427cb17939bf7cea03a7 : SUCCESS in 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a35e25d83df6450ab3e077ec54018f1f : SUCCESS in 58m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/244a7375a3f94556b9199c3b5c644b64 : SUCCESS in 1h 21m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1b2ca277dc2d11fe5f2650ffffe5e4df0079716e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2025-02-13 16:41:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"242907efc190b4d06573d468b0c2b17b01401e6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:42:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"572c7853ec34e0200da961a747c8aecfa623fcaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:32:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/280c364765714f92857e6fd736a6bcbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/182b33a281224b68b1bccefb2da322e4 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c36ac689b27440b9922f5ac81c02f65 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/052ef1c655e14948900464fa08dac5c4 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a6058847fc4464099aa7889428922e3 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/222c4aa23ce2481caa86b2d99890a3b6 : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80044a786f8f4b19b3dcb9e3c620addc : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db32581adb764537a647d86a12d19556 : SUCCESS in 10m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/121e2f710bac47f59b1727c32f08b0af : SUCCESS in 10m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c69496e939c1467b85d20cf9fa29be79 : SUCCESS in 9m 16s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/96ba7e74a36846a7b96bfb86307b930a : SUCCESS in 14m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/13e27f6a28264a9bbb977e9c8194980b : SUCCESS in 5m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/41fa7b4bc01d4f5391e728683c9e3b5b : SUCCESS in 4m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/48f2e39592c94bf1aac5e7c96abfe274 : SUCCESS in 9m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/64dbf88f1fbf4d2fb8fc90135a9cf75c : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e36bc1af28254933a9532fb095578cf8 : SUCCESS in 7m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/617385ef373f474baf9c72bf8186848b : SUCCESS in 10m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/05c3fc1dbcaa4bed9246acb269884e4e : SUCCESS in 13m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0b64402b5937451baf97521161448392 : SUCCESS in 3m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8fab7fbe924c430ea284574a54c156a7 : SUCCESS in 12m 51s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8e767ee394c24db4ab832e9e59c84b06 : SUCCESS in 26m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/08bd09fda3074f5bab33f6652a80cc6f : SUCCESS in 10m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f29dd61d324244afa511c1c3cc7ba814 : SUCCESS in 15m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6f431d38b8384a53bf8b9364a0a387ee : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/27bf297b8adc476387f276b575e27719 : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c473d4cf9b3848dc8c77866d4c5f8835 : SUCCESS in 25m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/579b6dba22604c919cbce5c1b57cc217 : SUCCESS in 6m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6a185d3d80b84804a6b1ca64aa65ed50 : SUCCESS in 4m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08f92e7de7354cba8d6378bc252a0798 : SUCCESS in 8m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8e629277336f427c8e747b3b9e590e4a : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c05cb298ee334d10bcd322a8e31d42f0 : SUCCESS in 52m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ae410f8d448416299fadc485847a149 : SUCCESS in 1h 26m 47s","accounts_in_message":[],"_revision_number":1},{"id":"73384fb745e9d59445e31c93c6d09ff24d7213a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:32:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4fe4c0cbeb3621fe90281c78d6cb52074890218","revisions":{"d4fe4c0cbeb3621fe90281c78d6cb52074890218":{"kind":"REWORK","_number":1,"created":"2025-02-13 12:21:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/941550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/941550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/941550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/941550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/941550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/941550/1"}}},"commit":{"parents":[{"commit":"cd5f30071ea047fa4b3eac2e2d981629e67ae7d4","subject":"update constraint for os-brick to new release 6.4.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd5f30071ea047fa4b3eac2e2d981629e67ae7d4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-13 12:21:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-13 12:21:40.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 1.5.1","message":"update constraint for openstacksdk to new release 1.5.1\n\nmeta: version: 1.5.1\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: stable/2023.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id0a697a614e8566081431eb8e701e96260e4d2a0\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6c62a433cd542af230fd37de0d24c1ce56615af6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4fe4c0cbeb3621fe90281c78d6cb52074890218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4fe4c0cbeb3621fe90281c78d6cb52074890218"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
