)]}'
{"id":"openstack%2Frequirements~942363","triplet_id":"openstack%2Frequirements~master~I2047bda01206c69aab263413015cd87d51300679","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2047bda01206c69aab263413015cd87d51300679","subject":"update constraint for stevedore to new release 5.4.1","status":"MERGED","created":"2025-02-20 14:19:11.000000000","updated":"2025-02-21 23:35:49.000000000","submitted":"2025-02-21 23:34:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942363-new-release","meta_rev_id":"e995b1ab627eacf4ff65612d0bf36505343e77d5","_number":942363,"virtual_id_number":942363,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 23:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-21 20:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-02-21 20:16:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 16:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 17:26:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-02-21 20:16:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"28f71df54ced522e1596ade1d8c8c8bbf080abd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-20 14:19:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d0fa481cdf15bb48e46bd602c6a042bde5b14c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 16:07:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3d3fa8231154a408456cd2e8757df90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ddddb01811d423f9cf29be5f97f9c60 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca1bac8fcdc461c890efe9cd9a027b9 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38db1d1a99a74e0bb28b0b380fd91725 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01d9258cb9f641d6a845b40a5fe83530 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2182b0a99bd46719b033c6ba0451086 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4123d164318443f4ad25428be918fc6d : SUCCESS in 3m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/95231bbc36ea4ddca7e326beab4f8850 : SUCCESS in 16m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9e03a5d4ef0d49d9bed2808593529e8d : SUCCESS in 21m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b951643172ac48e99eb7b50c94fb8f45 : SUCCESS in 15m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2a5cd275f64e4cdba7f53036eb5d9988 : TIMED_OUT in 40m 47s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2bb8299a1c6c4122b99392fa5eab451c : SUCCESS in 13m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2cce30d629854e1385294971f7e88a06 : SUCCESS in 5m 31s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/96e354a3242847bcadc40d4c67db5f75 : SUCCESS in 4m 17s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3aa9fb9318ce4c60be058a42ae0450dc : SUCCESS in 9m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ffa2c4293aa64653be02aaa54fa0e8ab : SUCCESS in 4m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fbde90a945024819a85d328098f6ccf7 : SUCCESS in 5m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d439d22f5b9f40de9e3f322deab74fa5 : SUCCESS in 22m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3ab5373f74d945748f23cb557264e1a5 : SUCCESS in 6m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c9d8efb48a145a496cfb33d78451b65 : SUCCESS in 6m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a652af15054944d4bd74a89f12d7e7f7 : SUCCESS in 5m 44s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e6fed13555be47f5a15f01cfacd381dd : SUCCESS in 49m 28s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/93642fcc3e964ac29f6939a53d1d8250 : SUCCESS in 1m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b80cbd252c0945e588e2fcdd01000cc7 : SUCCESS in 9m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0e83c402e75f4b31ad51055efe6e1ec0 : SUCCESS in 3m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/890e59d0cbe441dfa67fa1d532a9435b : SUCCESS in 3m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/979ba7a70b734303a0f9ae1c9af9aa08 : SUCCESS in 12m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d9aac3ea8cf34e87a1cdcaf3527a044a : SUCCESS in 21m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/13262680c402407599a097af3a69a7b1 : SUCCESS in 7m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/35e8674ba56041378827a500b76950f8 : SUCCESS in 13m 33s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/270316b104b54bfaa1b9fe17516d1a29 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6f680213bce4a9bbd8f26907859320c : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75783e2a815c4c8b93949cecf90db6be : SUCCESS in 22m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/796409c28fb741ada3896826e400f211 : SUCCESS in 5m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fe4a1e9a551240aeb96ab3a76f550e74 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be06b0a8e3c2427ea0fc508166cbdf4b : SUCCESS in 15m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f708f88f212244b587bf18f77cb701db : SUCCESS in 14m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44730a0114624fa6ac3768bf27538c8d : SUCCESS in 53m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a261b9ba2d94c629538e41c6c0839df : SUCCESS in 1h 41m 12s","accounts_in_message":[],"_revision_number":1},{"id":"cc9d65faf927bc69870e607f0309080feb0c75fe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-21 17:26:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01dd7221981c8352b19191c0c0e75d31fa976934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 19:22:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/659b3471687a4e499523a40d604e6735\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d91089c7ef4189820177c8d0c4545f : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7fc67a3bc614391ad321ec4385a396e : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc094e0a381b4fd7b00d4c8397992e7f : SUCCESS in 9m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/700c048f14d74375960ec5f9fee30b2b : SUCCESS in 2m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd920dc5aa0945199d7e1b1a74d129ea : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5df15bf2c9264ce29596c98ff76eccee : SUCCESS in 5m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/95ee99ef2b0f4981a26ab5746ff869fc : SUCCESS in 14m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5575890a2a2b4873a4faba54fbb8e2fe : SUCCESS in 13m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e6cedecc9052421fa4038bc38ccc2a67 : SUCCESS in 16m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/61cfcd3b8cc645b5adbfe3c0cbfe772b : SUCCESS in 26m 01s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a7966944d9aa4085ac26d808c22fa8b1 : SUCCESS in 4m 56s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/73ec26d72a15489386a3e5724604367f : SUCCESS in 5m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a2758980713849ac887391d27fcf2254 : SUCCESS in 4m 17s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/882a47e9742444e386b3cab01e911417 : SUCCESS in 10m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9fe4e6e9b36b41d9a4af11eca00e7140 : SUCCESS in 5m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/458361dc29d24a979d3efe011a7f3e43 : SUCCESS in 5m 40s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d79efb6f719441d5bf0ebc6f5582adda : SUCCESS in 8m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4c742170c9b0413d85db240c46f11fff : SUCCESS in 14m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/307034c2531f4004af49ffbe036b2024 : SUCCESS in 7m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/df4c1abc266a4d34a2837d5a78967645 : SUCCESS in 10m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/255fee362e3b4929aedd8dc5b5758b54 : SUCCESS in 11m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c84e88ca7f2d4533bdd3f478baf3a5f4 : SUCCESS in 3m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/494bd616e1cd45dbbcec5be928f34a65 : SUCCESS in 12m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d2bbc58eea924b33bbd88f541ac8b7b4 : SUCCESS in 6m 04s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/19675af6c41346b092293fbaf10e0bad : SUCCESS in 3m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e69cc603bbdd4ce0b3bfcfe083c9a2c9 : SUCCESS in 11m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7a1d093dbc7a4351bd6f6d92f5f71a46 : SUCCESS in 16m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc7f30cca2c64a6d87609bd05e32f6a9 : SUCCESS in 14m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/51acce52bba74323afc0e982f2e33f92 : SUCCESS in 25m 16s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6fe54fbeb0964815aaef5a5bf4cce54d : SUCCESS in 14m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/14406668ed1f4759a1c7c62940dbda36 : SUCCESS in 6m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce667317ec4d45d290710db61e5eb2ca : SUCCESS in 26m 52s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6b0440fcd2414e3685058e5b89805714 : SUCCESS in 8m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bf917516610c46f0be5307ad560d089e : SUCCESS in 10m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e0b8747ddf7488d887c923027b3b8b3 : SUCCESS in 5m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9f7cb533397d497c8eaede4522cc4fef : SUCCESS in 17m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23a63a562f274e1a81b596f33e13da28 : SUCCESS in 1h 43m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29148e9b17bd4c0d8df46bc9996451c6 : SUCCESS in 1h 50m 04s","accounts_in_message":[],"_revision_number":1},{"id":"25b8716983f05b92e35706e90a6a141ee8f97291","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-21 20:16:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"636337c8b62efc52407c68cc28ca39b016280bdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 20:17:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69965eeeb1884a2ab575e9fcce29fb996c24cd8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:34:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf2349bfd5b54e439ffeebd62c52057d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8af5039f00a4ca8b290aedb80c00e6b : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49214f6be3e942a19632d86d4eb62f73 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13e446aa455545bc8a1b58baa80b940c : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c37ad8f87269403493cdfbd0677fcf0d : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b59f8ecec16f45cdac7a162786ceda51 : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59338d6d34e842158a6829a9c528412e : SUCCESS in 3m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5c4d718cd0d2426384d1c5b753422085 : SUCCESS in 7m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9b7b433b0be9402ab4a345395c5ff723 : SUCCESS in 9m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8c434342a59545b291ba92d04c58f808 : SUCCESS in 17m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d20aebaff6c94b50bcb0525471ea798b : SUCCESS in 27m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3c73a6ff7f304e3e906c570834be3fa7 : SUCCESS in 11m 37s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f009fce984544fd6a1ad14a3e6068e0d : SUCCESS in 7m 10s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/26e3f2fb51dd4043b874844b32afac7a : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1f5f9b0c3c9843c6a701d7b63a44173b : SUCCESS in 8m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2539c58d1fe5464cb36f1decee9c6aa9 : SUCCESS in 15m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6183553a6d6b4072959f7be412c8b995 : SUCCESS in 8m 47s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/788b9510a21d4f4ca9f563d56f34f270 : SUCCESS in 11m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0f21fb6f76c04bce9fa8e46b23756254 : SUCCESS in 13m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a893a4230f7042809472db3bc59b5ce8 : SUCCESS in 4m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ad705581da904ff69284e7dade2ae74e : SUCCESS in 16m 23s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d0fe5f5ac61c4e23a016e2df7600addc : SUCCESS in 19m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/004f6e46e7c7489fa1d49684e245161b : SUCCESS in 10m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7aad7f36e19d4b95b5bce27b6512fcf9 : SUCCESS in 13m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/09bea775c91446d8b252e92b5717d4ba : SUCCESS in 7m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c56ec161f54544b6bb04d6009569c398 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f74efc4da82409fbcd184ac634d67e0 : SUCCESS in 30m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c2b59eaa9722405da17c2a89b3093e5e : SUCCESS in 7m 37s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2ca81e55bb3747139558387feb3ec5d7 : SUCCESS in 7m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b2dc00586f0340a3bd87363aa8705953 : SUCCESS in 4m 01s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b2bb4f5d8d604a7a8a6cb2bf0a6a8950 : SUCCESS in 17m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5caf68cb69bf477bbd24cb390ce203bc : SUCCESS in 54m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3cca6ec65464c4c9e55041601e2b544 : SUCCESS in 1h 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ac4c3353c51fd1af2a84328997b035927b584c0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:34:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e995b1ab627eacf4ff65612d0bf36505343e77d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:35:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf40836aa9584cbca87d53691969cc05\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d51be9df0ae461db70b094da688ab06 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0d607789bff59bf08ad00653410795b94f451c0","revisions":{"e0d607789bff59bf08ad00653410795b94f451c0":{"kind":"REWORK","_number":1,"created":"2025-02-20 14:19:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/942363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/942363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/942363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/942363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/942363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/942363/1"}}},"commit":{"parents":[{"commit":"a75cb655fe22ba8410ef3a185c57c32a37a1c294","subject":"update constraint for os-ken to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75cb655fe22ba8410ef3a185c57c32a37a1c294"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-20 14:19:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-20 14:19:10.000000000","tz":0},"subject":"update constraint for stevedore to new release 5.4.1","message":"update constraint for stevedore to new release 5.4.1\n\nmeta: version: 5.4.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia16bd36c80949adad4aba8d9d5591c2c9a5f3bc4\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I2047bda01206c69aab263413015cd87d51300679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0d607789bff59bf08ad00653410795b94f451c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0d607789bff59bf08ad00653410795b94f451c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
