)]}'
{"id":"openstack%2Frequirements~942428","triplet_id":"openstack%2Frequirements~master~I0e5beddde22ab4bb6c4f23387393138722178ac4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0e5beddde22ab4bb6c4f23387393138722178ac4","subject":"update constraint for oslo.messaging to new release 16.1.0","status":"MERGED","created":"2025-02-21 11:27:00.000000000","updated":"2025-02-21 23:37:50.000000000","submitted":"2025-02-21 23:37:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942428-new-release","meta_rev_id":"b4bb7c7ad85e7f3cff9297b4e303d95222073012","_number":942428,"virtual_id_number":942428,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 23:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-02-21 14:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-02-21 20:17:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-02-21 20:17:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 13:18:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 14:33:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-02-21 20:17:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"43fe0eca3608ca6eddcf04cc5e0640d94660dbef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-21 11:27:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dd615fe725e3928a8f62dc1418f677f54935ae5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 13:18:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d90be96dbc4048e1bae2cd60007b440f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa20f00e4e04685b492dc418bca36b1 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9ac6ebb20e84a4db223368e92a769f7 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df34fba8e17445c08a3ca6062f709660 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3df92900cde84a66ba85839f5d55013e : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d1e79dc7fdc4caabdacd905a1ef32f7 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ad29fd5a9af40c3889b8aaa4a3bb695 : SUCCESS in 3m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/412aa79d923847e5b84873285c2ca2e1 : SUCCESS in 21m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/97815a1e791c496990749fe2771f8226 : SUCCESS in 9m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a9b05faba43e4052a9ada0b55243eafc : SUCCESS in 19m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ca37237ee3b34e23ab71f6e6b614ecca : SUCCESS in 22m 15s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/621d00459f044772b403e522f60ca5a9 : SUCCESS in 4m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4848de0184f342e1ac0f764a6c208c4a : SUCCESS in 6m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f9ea86927dfa4f17bcd80cfdf67c5e58 : SUCCESS in 4m 10s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b0cda0e19a2a4640874027e5d7bdb0ea : FAILURE in 20m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/00b762a761e04ec885b6af14eefdd220 : SUCCESS in 4m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cff0de46f3e94490954e6abee964e224 : SUCCESS in 6m 38s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7d77b03876f646b8a9c1d2d82bda5597 : SUCCESS in 9m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e176c4279b8943bbae3b4597c3ff997e : SUCCESS in 13m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ec504736b9d47dda7ede5dc599eb7a8 : SUCCESS in 7m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/97176c25e22a44b59dadb35f09a36f3c : SUCCESS in 8m 20s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/580a2582929f46ffac8c75c3ae684ca2 : SUCCESS in 10m 08s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/12eb22de1a7046a2995a53341218da97 : SUCCESS in 2m 43s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0448a38fb371443aa6965084b3f1e7ef : SUCCESS in 11m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6fe49b41ec68452db4bfc1ea5d23ffd2 : SUCCESS in 4m 36s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1f580451bc0f43b1b538bfc85b26cbce : SUCCESS in 5m 22s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9c7ade453f454beda6bc42ea9338c9dd : SUCCESS in 13m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/433030cfcd39419f93654a4c012bb4eb : SUCCESS in 17m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4bd7e9e298be43eaa2be63d6fdb27e27 : SUCCESS in 8m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/542c7540c6e54e46a0cc602ee63c2337 : SUCCESS in 10m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7600dd73c44f4ac3ae59baf270b51171 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46e1035639ec42218fb034eeb108ca69 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5341e86708d44ee947bc7f7f25c8cf4 : SUCCESS in 26m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cdaa4fc65cdf4ac7adaf19c0cbe8f588 : SUCCESS in 6m 37s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8b3a4d3e601d4842a305f5b01dc213e5 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eedab3f7e1ce4e6ea7216c3a870a72ab : SUCCESS in 6m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/30a62c9a46e740ba810e165dd08ba4d9 : SUCCESS in 17m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8232c99e909649e39d2fd29bdb1d97cf : SUCCESS in 52m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fffc8c7533e9404b959b62d160214856 : SUCCESS in 1h 20m 06s","accounts_in_message":[],"_revision_number":1},{"id":"040cf5183fb81d017093aed1dc206a3b03116bac","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-21 14:33:37.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7d86e5f6fbbfcbb97471932737b622f08ce38cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 16:10:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdb7bf660af8437fb67146dfedcc86cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d444c92f3d424b6ba55a48379b23486f : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf2475f15d846b0bed158fc04ed1b06 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1035b23d4f74283a22cd6761828b8e8 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b13d64fbcb1b4e08bf8bbd59c574f2b9 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4e6ecf2cf7f42739005ccac2873736b : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9096bf7319a24118836650970bcef1ae : SUCCESS in 3m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/50ad03ed3eae4ec3abe7ee06ff628ae3 : SUCCESS in 14m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/19c7ab1eaf2441029848665b19aabbcc : SUCCESS in 8m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a60dc5f2d72144da9ac1a3b37f93aa01 : SUCCESS in 16m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bfadb780f54d47cab5b5188af581ec21 : SUCCESS in 16m 42s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a6d47326237544679f1d5a34528d4373 : SUCCESS in 4m 56s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a5e6e4378a7448f6a91564021ae8868f : SUCCESS in 5m 17s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d4e49dade42e4913851aa2a2531513fd : SUCCESS in 4m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/20ca455ebe1e4be8b5ff096c4206bac0 : SUCCESS in 10m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1254dcd652e5494c94e1612349e7873b : SUCCESS in 5m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3d598cae8e3f478d9b10f2910add2994 : SUCCESS in 5m 41s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/efff44d8fb09442bacc065595f62c41c : SUCCESS in 9m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/286286c739c647a1ad6b4d18f17a6c3a : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e6a19b00cce49d59b964dc6f9ea7fd5 : SUCCESS in 8m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a01aef551282408a8a29a3b35da329f0 : SUCCESS in 7m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0615373a24a04f81b70f0f2cfb3a0cbd : SUCCESS in 14m 15s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b10a4ef8eca545fba1bf6e40dcd2f00f : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5741547ed23e45d49fdff141ccdf68e4 : SUCCESS in 11m 44s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5227ca907bea4815aaffae381c7d3bad : SUCCESS in 4m 50s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/37346df3c1ab4b2bae96a097f929cb9c : SUCCESS in 5m 22s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/455464e29aea4313b0e2df9f163ab2f8 : SUCCESS in 9m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f6a3316fe14b4de1aba8f1f9fb8a2e3d : SUCCESS in 25m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9cf86e7e46d141dba44d086f24654fc7 : SUCCESS in 8m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d5951120d3834ac4900a709c2465b54f : SUCCESS in 12m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/17c2e50dd4c5465ea417b3daa43932e2 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b471ffe641a14a7d80fc554933c147bd : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40460a4e9b2a4ff1b9aa3c633057ac72 : SUCCESS in 23m 31s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/de5aa0b912874c23b8fdacd5380b852d : SUCCESS in 6m 52s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8215b1e2df6b47519ab8a7af30860036 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aeef5d809a0f4402830ea806024a97b9 : SUCCESS in 5m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e2ef4c82e2604d1393462f2cdd3103fb : SUCCESS in 15m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8579f767c5d24cdb9f365576f2ad7198 : SUCCESS in 1h 01m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43c8774dfda948058a960da141532e30 : SUCCESS in 1h 31m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ecee05938390b4b84bab1cc8656a65ec29ef3c08","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-21 20:17:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9825e452df140b156a685d87f8fee2e956b3dedc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 20:18:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2ff9eaac8accc0b75601127df0673a4b760b44f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:37:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4983046473804197b5348cd814e74a16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02e816328fc44e2a735991a50bea525 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c50371b79f742d89737300ec72e982b : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37d0fc06b4804e4089c31505f657cc56 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3de88e1b69474913b6b56e6dfd40b4c8 : SUCCESS in 10m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56c36df85b814780bd484ae05a4760ba : SUCCESS in 9m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d696e867841a442fbc22c3410fe3ab71 : SUCCESS in 4m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5327b9580784409aa5a6e3314fc52b24 : SUCCESS in 27m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d3b47a6655c147b58f57074e55938add : SUCCESS in 13m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a22ded7aaaf64271a7e2e347e1cf4bb5 : SUCCESS in 27m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/27f5fc89f7c34ab6a991f5daa7a0639a : SUCCESS in 29m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1ab8e760e540427891fd2ba37606ed7e : SUCCESS in 9m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/76f9b726ed7544e7b27da62cfd239cba : SUCCESS in 4m 44s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/525fca87cbfe43fe8a5d7c043bc92967 : SUCCESS in 12m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bf42422624a84b3bbc7d8c8b223b8835 : SUCCESS in 12m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/aac5f0c169b846d0bb1a125ad5059c63 : SUCCESS in 16m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ed1a7871cfa462ba996eac6ad2f8f0b : SUCCESS in 7m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6df705b6825a4bb7b502429a2200ef72 : SUCCESS in 7m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f0ce8ff4651d4736bb42b73133cf6894 : SUCCESS in 22m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9676905c480b4d50b85fcc4ef7dcc85f : SUCCESS in 2m 48s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8ef86e849b6c416189703e1e9ff020ee : SUCCESS in 12m 16s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/273fa6bec07e40a195361401e14ef228 : SUCCESS in 20m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2abc43f927864d668a20258e4fe1784c : SUCCESS in 7m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4dae072551a54e61bbe53596b5a28ca9 : SUCCESS in 22m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/96a26bc3df0840afb7d19fa3325b6f1d : SUCCESS in 6m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2cfb4f431586451caf562521c05ada8e : SUCCESS in 4m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6b490e6d6544816bb97e0dc99719597 : SUCCESS in 24m 59s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/07a12ab01f98461f9a4cc8b45eee44af : SUCCESS in 8m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1e5d2ea05cab436baab0855441cdc287 : SUCCESS in 8m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe5c17cdcb4d45ca994a8250f3e2ce99 : SUCCESS in 12m 19s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b954750eae724f4a837d69836fc3c8fe : SUCCESS in 15m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/102891a4cd974d54b3d1e67c71e8a6e7 : SUCCESS in 1h 00m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39bb33679b3947289067f228d4b8eec8 : SUCCESS in 1h 40m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1283a930008a9507c57bf36024c5136b0da5df99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:37:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b4bb7c7ad85e7f3cff9297b4e303d95222073012","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:37:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/937915da1723456996cb480a15b11caf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a17363d455444cb3d37d2cc891a853 : SUCCESS in 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f92e51f592a9e688aeaf03a9d2c5c0be3ffda659","revisions":{"f92e51f592a9e688aeaf03a9d2c5c0be3ffda659":{"kind":"REWORK","_number":1,"created":"2025-02-21 11:27:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/942428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/942428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/942428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/942428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/942428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/942428/1"}}},"commit":{"parents":[{"commit":"a75cb655fe22ba8410ef3a185c57c32a37a1c294","subject":"update constraint for os-ken to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75cb655fe22ba8410ef3a185c57c32a37a1c294"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 11:26:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 11:26:59.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 16.1.0","message":"update constraint for oslo.messaging to new release 16.1.0\n\nmeta: version: 16.1.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I88a707e05ea646f4d294cb52794732a1f6c84cb5\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0e5beddde22ab4bb6c4f23387393138722178ac4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f92e51f592a9e688aeaf03a9d2c5c0be3ffda659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f92e51f592a9e688aeaf03a9d2c5c0be3ffda659"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
