)]}'
{"id":"openstack%2Frequirements~942444","triplet_id":"openstack%2Frequirements~master~I54a2987283745a7b3e0e8645c1ebf6e615f4e1c0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I54a2987283745a7b3e0e8645c1ebf6e615f4e1c0","subject":"update constraint for ovsdbapp to new release 2.10.1","status":"MERGED","created":"2025-02-21 13:45:06.000000000","updated":"2025-02-21 18:54:18.000000000","submitted":"2025-02-21 18:54:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942444-new-release","meta_rev_id":"7701e3064d0a3b6d6963572400b7927d7656be19","_number":942444,"virtual_id_number":942444,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 18:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-21 15:48:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-02-21 15:48:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 15:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 15:48:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f9731a16cb909a29f3a8f71b96c83a3d52cc5921","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-21 13:45:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bbc61a93a46bb0319064fb09ecda1d61d684a3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 15:23:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94cbf57d11fd4f02bee18d3a76e64382\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b54c6699ce84763b420954bbe0ddedd : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb012b97ce54443a7bb1ee49910991a : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85878f87be004d7d92dbd67552752edd : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/154cd79f3ca64b618c8da01947851ae6 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e80adca104845ec92745e87e307ee9f : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cecf73c5666c44f9b9833f6285f33799 : SUCCESS in 2m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c2d5a5167b1645e683f57f0914f41f6e : SUCCESS in 15m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d065aa7c58cb4c128f7c047dc640bedb : SUCCESS in 10m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e2f2f9d3d4af425785db9b81e204ab66 : SUCCESS in 18m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/38fd4434f39b492da02e8dfef6eb3bc6 : SUCCESS in 31m 04s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d3f67cb810a844a49a091c364350dc97 : SUCCESS in 4m 17s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/22ec322cb36d4314b408be78fe6432fa : SUCCESS in 4m 20s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7e640f3bdcb24f038d74234df5c30163 : SUCCESS in 3m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e99d9ae390214a128c42fe22ad14e5b6 : SUCCESS in 14m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/29c335983ad0423fb4fa3b250202b422 : SUCCESS in 5m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a5a6e0bd1d3649b8b441c00cf9ef1ab9 : SUCCESS in 4m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c566124919984611b0c4de8a2dc4214b : SUCCESS in 8m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b2def0b231a84994b7de069e406c22f2 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8bc38f2ea6e447e98e8b320ef3675b7b : SUCCESS in 7m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1db87988e7704e02b3eff02fa7d032fa : SUCCESS in 10m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3e0e4ad6c06f4a999fed8a5b7a6e5dc1 : SUCCESS in 11m 14s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1342cfcd87cc4220be0cae314240dde7 : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ed21d425765e4a1b8faf196987d2f641 : SUCCESS in 9m 17s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/57b0976737e24bf7bab79763878a50db : SUCCESS in 5m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b89483981eeb432eb4e7ef362779e662 : SUCCESS in 4m 02s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ce3fa90e13644bfb9fa67e0fb4ceb0e0 : SUCCESS in 12m 17s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c5e6f2f4cd7b46f79960d529907d42bd : SUCCESS in 18m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dd89dd1cc6bf4e08b797c7e90ae36866 : SUCCESS in 5m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/19adae13a4f64251b7abb179a2ab9d3d : SUCCESS in 10m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6de504e6658647928f20261eb039cec5 : SUCCESS in 3m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3a4990e739a048b1996fea4d15214025 : SUCCESS in 2m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dffa4c22c878410298a89176c94cc80b : SUCCESS in 27m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2ff7a3f100474e8697d092b6a12a83f5 : SUCCESS in 4m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6fbe0a6e6a3d459b9166a20c04ce6a2c : SUCCESS in 3m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c5450968d59458e85a272e182f75ae5 : SUCCESS in 3m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/15eaf0bc23d04471851652d2c7acc034 : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/463592959602477caf4a789ccd4f405a : SUCCESS in 59m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e72b22dc275c4078a367d04c994eb6d9 : SUCCESS in 1h 35m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8df056064b058c1416eb168f7f913eed104000f5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-21 15:48:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b91638237b0774b9f8e3b278d2bbf4413113567c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 15:48:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0733bb195a58016e5e1daf0b2e47cf502edc4b3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 18:54:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2adc92cbdf8a45a6b4805c7c03ad6afd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc0698934c8413b960d023129f84927 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5943055f056454ca132ecd31c30d673 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9213336288b84318a9301cad328d68f9 : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ebd233e7b30441eb93a9385304d83e19 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/935feaadfd064ab08baf29a72e9ddc3e : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c35713e00e53470191f4d5dbf687913e : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/70c4c4fe81824bc1ad1418efc714b907 : SUCCESS in 16m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4d25cb839e4349b2b1c4f244d623dc55 : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e0701f70dbf6456b9f40c45366009626 : SUCCESS in 17m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c596f426595a48df90b654838aa17928 : SUCCESS in 16m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7fb98e3354e0491dba18956299de2db5 : SUCCESS in 12m 14s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4e598880a96d4d61997bc448b8e1dc3f : SUCCESS in 5m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/967364fcdac54ed5a81c751c4a422e06 : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7f216413e09e4684b13e6744467a2f2a : SUCCESS in 9m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/51f64b2a1d3f457b8fe5175d38cd81a0 : SUCCESS in 14m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60fc56e438b04a71b02d37d8f708be39 : SUCCESS in 12m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/227a833d0e3a406589f8357a88cf5464 : SUCCESS in 8m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4f04b9fa7f0b4e829bd9316056c84dd6 : SUCCESS in 12m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/eada7a259e1c45c08056ef76e3f3fa80 : SUCCESS in 2m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/15a498b25493445d8725557d67a2c477 : SUCCESS in 11m 14s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4b06d8dee8d1452f8991409870760b32 : SUCCESS in 18m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a40a7f9454554092aaf9ad50fbc6472c : SUCCESS in 9m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/587d0f86d76e40928f8fd3d256c297c8 : SUCCESS in 13m 33s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ab955a042dc64915aed4372dbb01be66 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4252120005ab486d91c9d34b7ea22b82 : SUCCESS in 5m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e857742a1024b9cbdec8d023f439c0f : SUCCESS in 25m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6c1b97c58e454a1a863d872f10d1d330 : SUCCESS in 6m 05s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7a85d488a90a488ab8f0c45716a0d702 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de83b480d57c4632aaac9d32b1c18451 : SUCCESS in 6m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e32fe22698dd4fd48047c8d0446c9f45 : SUCCESS in 17m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da8e857de0fc4a8cbf6e5dea94c9f467 : SUCCESS in 1h 02m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3804bc00da7455ea60197200245099b : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7701e3064d0a3b6d6963572400b7927d7656be19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 18:54:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ef81e065ed7da6cfc875574edccd0506005e99a","revisions":{"5ef81e065ed7da6cfc875574edccd0506005e99a":{"kind":"REWORK","_number":1,"created":"2025-02-21 13:45:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/942444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/942444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/942444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/942444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/942444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/942444/1"}}},"commit":{"parents":[{"commit":"a75cb655fe22ba8410ef3a185c57c32a37a1c294","subject":"update constraint for os-ken to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75cb655fe22ba8410ef3a185c57c32a37a1c294"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 13:45:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 13:45:05.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.10.1","message":"update constraint for ovsdbapp to new release 2.10.1\n\nmeta: version: 2.10.1\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iaf8e7e75ecba5247b998d1877c9ab4185a6d0f46\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I54a2987283745a7b3e0e8645c1ebf6e615f4e1c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ef81e065ed7da6cfc875574edccd0506005e99a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ef81e065ed7da6cfc875574edccd0506005e99a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
