)]}'
{"id":"openstack%2Frequirements~942445","triplet_id":"openstack%2Frequirements~master~Icf604b90fa4fed6c776696bc893c459f87af80cb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icf604b90fa4fed6c776696bc893c459f87af80cb","subject":"update constraint for oslo.utils to new release 8.2.0","status":"MERGED","created":"2025-02-21 13:55:29.000000000","updated":"2025-02-21 23:10:56.000000000","submitted":"2025-02-21 23:09:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942445-new-release","meta_rev_id":"e65520855a9ad67d78b44836321c7bc96ae3ef29","_number":942445,"virtual_id_number":942445,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 23:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-02-21 15:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-02-21 19:34:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-02-21 19:34:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 15:56:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 15:57:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-02-21 19:34:47.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3ae8f010e3108cef3837ca1cd2f72e707e11e174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-21 13:55:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"649eee91267a75c5e95df3a59f386ffe45aaa0aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 15:56:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fab554a71f3c4a35b2dfa499ac61a1b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08bcef3fc20480ea1303e96ce0f962e : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d58ba583bbb545419efede187e4d959d : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c49f61969ac436180ab4cc3ceef8208 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa0fa63276304693992b7ec444ac6de1 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ec030954b8c46f4adf2cd98ab3f81a8 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab6c9c67176147c8be4d2e92bc59851b : SUCCESS in 3m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d6a7ba61e0ef4a6c86fa06e9aeaefd6e : SUCCESS in 15m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8ac5009254974bbdbf19b0347bc08b6a : SUCCESS in 8m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c01b99f159824e41a7cb69e7e336d0ed : SUCCESS in 16m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/263c31205abf42a1b667568c5f983f6d : SUCCESS in 20m 38s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e31592279c1a4b368e68496050504390 : SUCCESS in 4m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ddc162f7fbf240ee920472219c408157 : SUCCESS in 4m 42s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a493a7aba2c54b29b342f6d96ed8039d : SUCCESS in 4m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5aaf0d5451c44b23a68ea63eaff4e6d5 : SUCCESS in 10m 44s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bae709295d624142bc4b38050b7ac46e : SUCCESS in 7m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b457d372c20c45d1bc3442b743ba99ee : SUCCESS in 8m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e9576105851e42d6a1ddcbbfade89936 : SUCCESS in 6m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/03f51276715641e1939abe0a99529c1d : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0992123b71794c21976f5c76d90550bd : SUCCESS in 5m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0050c2d5a80e42dea374e2674508af4a : SUCCESS in 10m 01s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b90e81e06f1d4945b3dbb09a2878e4b6 : SUCCESS in 13m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/74b62207572741d19751413384590606 : SUCCESS in 1m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5613773587b74923a5075d9835f83790 : SUCCESS in 14m 58s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8b5099c69a894c7e9951450515da60e0 : SUCCESS in 7m 06s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/15c685a347bf4114a2514dc96076db6f : SUCCESS in 6m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/349c4e790d1847c3b84819c4ba294287 : SUCCESS in 11m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7626334dcee846879427fdfe42e25be8 : SUCCESS in 23m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/007a9d0465604bc28f0f3fe0ddcb75dc : SUCCESS in 7m 14s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/264df102a2a843ca8310a5b53849d2d9 : SUCCESS in 17m 33s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8f91a8615f04483fb9a16320d7f47982 : SUCCESS in 6m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ecaed0efda56409d9f04e79f31cd767f : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/455fa6aef4cc4538be75d96de226f15a : SUCCESS in 25m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/edd33952a2db4ca197c1facff66b17d5 : SUCCESS in 6m 14s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/560c093a898345208610215df86970c8 : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a687e242220444af82d036133ec79e7d : SUCCESS in 5m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1407c32b25c747de92fc962dab6c37cb : SUCCESS in 17m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdde2c653286474491d2a9e319575cab : SUCCESS in 1h 03m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ce9a3ae6f564486bad8919e2907e998 : SUCCESS in 1h 54m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a836c3ff7744d9f72179bd24deb0354e139d967f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-21 15:57:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"275709ce080d96028e1501c3e47e94471cf7259a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-21 19:34:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b8e1d79dfd247126ac2450ec581dfe713b947df1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 19:35:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54791880e6ef8c5c651193d8a82762e2232262a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:09:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e25f64d759640c9b599e1165b7c2fa6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c7e88bce75418f9f032e307a8968ca : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a2bdb9411443228ea46e008d82b6f5 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a50de05727ee4ee5b85fbe4e725828de : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39b8cf940f2a493c8ed236dd84a1da15 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0be39b3bf56e41769e20ca7850957d9a : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28d60258385241c796a4495246ca7a52 : SUCCESS in 2m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4ac02acf4f1a41e8adb5656bc04edd59 : SUCCESS in 8m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dbf48182d9b744d3a0dc3dfdac412d54 : SUCCESS in 6m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e8b55a8cf19e46f1b3eccc15d1c8df45 : SUCCESS in 8m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/810fb409a9534efca75fcb503c1cb3fd : SUCCESS in 12m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/684cb79835cb4317b9c29116cc006a77 : SUCCESS in 11m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/74ce627f017742e68891300ddc2f0e4f : SUCCESS in 5m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1a3f15bc02c34e4bb8a5892de9543f7c : SUCCESS in 7m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ccb8ffd85fe74b1485b8cb150af8b66c : SUCCESS in 7m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7b9fda7083f64b979ffc943fd64c6b37 : SUCCESS in 17m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aeb6fbcf59654315beb263de2f3b1a76 : SUCCESS in 6m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/90c618c3cfe24737a120b2c90081dcfd : SUCCESS in 10m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/556c87c9eedb45179bac799abffdc692 : SUCCESS in 12m 26s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/88ce889340fb40d28258a775f35247e2 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6398a12e5e77438192134aee3779918c : SUCCESS in 14m 13s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8681695ecf3a447dbe1e8ad673b54fce : SUCCESS in 19m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1398a16c824f436abc166711d7148381 : SUCCESS in 7m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8f3fc90193a74af09e495e387aa94d13 : SUCCESS in 13m 41s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e18153eb187245e68c449cad76e7933a : SUCCESS in 6m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/faa24c9f285c4d608ee6351d08a63aa1 : SUCCESS in 5m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59386b3d9c0540bea9f5ceae20f8326d : SUCCESS in 26m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/dfc7c7dc4371437cbe20f58e7140a26e : SUCCESS in 6m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3853e4fc67d3411abfb0a67d9542aa92 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6d34a4613054e9aa7feaf54fd17f564 : SUCCESS in 6m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b7d5577431c342a7815f4594e6521204 : SUCCESS in 18m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d6a41c0662a49e9a908635a159462ec : SUCCESS in 1h 06m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08e31e77ed824987ab8b4d7e67f653b6 : SUCCESS in 1h 38m 32s","accounts_in_message":[],"_revision_number":1},{"id":"faf30e4b4b147718bee51b3f42e6f89bbe07390a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:09:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e65520855a9ad67d78b44836321c7bc96ae3ef29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:10:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b1bc070ea89456093c0be1517f6bc1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdab00bc0acf454eb791c9f531d37131 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf9f949f98e624d518dfda0ce191ead5dff84024","revisions":{"bf9f949f98e624d518dfda0ce191ead5dff84024":{"kind":"REWORK","_number":1,"created":"2025-02-21 13:55:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/942445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/942445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/942445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/942445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/942445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/942445/1"}}},"commit":{"parents":[{"commit":"a75cb655fe22ba8410ef3a185c57c32a37a1c294","subject":"update constraint for os-ken to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75cb655fe22ba8410ef3a185c57c32a37a1c294"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 13:55:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 13:55:29.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 8.2.0","message":"update constraint for oslo.utils to new release 8.2.0\n\nmeta: version: 8.2.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ica78ee1fe6183f17ce795b226891d0d4e36ffe0f\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: Icf604b90fa4fed6c776696bc893c459f87af80cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf9f949f98e624d518dfda0ce191ead5dff84024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf9f949f98e624d518dfda0ce191ead5dff84024"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
