)]}'
{"id":"openstack%2Frequirements~942449","triplet_id":"openstack%2Frequirements~master~If36b9c727a5d42ecfb4e557bf4056e20ca2b1454","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2025-02-24 11:22:01.000000000","reason":"Change was submitted"},"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-02-24 11:22:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If36b9c727a5d42ecfb4e557bf4056e20ca2b1454","subject":"update constraint for oslo.rootwrap to new release 7.5.1","status":"MERGED","created":"2025-02-21 14:04:40.000000000","updated":"2025-02-24 11:22:01.000000000","submitted":"2025-02-24 11:22:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942449-new-release","meta_rev_id":"ad22ab85ef63f6e6018b04e6cb73e925a75cbc32","_number":942449,"virtual_id_number":942449,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-24 11:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2025-02-24 09:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2025-02-24 09:09:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 16:01:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 16:27:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-02-22 02:48:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2025-02-24 09:09:16.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"9e5519388dba7f60216c0229d5efbb5adbf483e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-21 14:04:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6d1636265b1af6a18543c47d485e527f7cd5bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 16:01:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df3eeb47b6784d3da1bc2d8256868984\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e71dfaf6e7b1478cbd9e2eac786b5b50 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcc57409cdf341948dd7cc2a5e835fe4 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/330e2b1459154645a3570eea9865e179 : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1e0ab6685e745608759a401ff41a37b : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96dcd11ea7e042d7a54ff5e9c15ff3db : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aaf79d32c69947d2b565914361523635 : SUCCESS in 2m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d8fac0cdd76b49e296e65b7b660e54fb : SUCCESS in 7m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4f0ec0b827694996991ba9535934e29e : SUCCESS in 4m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/963d082cb1324b81b0d638ce43d77ac7 : SUCCESS in 8m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/72f287cefc5e4847ab01063621917028 : SUCCESS in 30m 43s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/95c2c23777194665bd57733e0b0d18a3 : SUCCESS in 5m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/03a3c9720a554a988df330aba46b4b25 : SUCCESS in 8m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b78ce704f49f4c3985c4e77b5b6459dc : SUCCESS in 4m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/57391870a5734cbe8f5aa473102c10d4 : SUCCESS in 12m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bcf8539bf43d42b8a0e9895523f08228 : SUCCESS in 6m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ea40e8ba43ce4bceab51723421dc391d : SUCCESS in 8m 24s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f8a556964c9c40edadd6e1d7548f9a94 : SUCCESS in 9m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/63d82a7f0e274132af89e456c0458f8b : SUCCESS in 16m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cef0effba05145ec987224bfbc625d9e : SUCCESS in 8m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f8217d2371e842a99eddef67c6370591 : SUCCESS in 9m 20s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/70606be84640403e8bb07ed58d1cb18a : SUCCESS in 14m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/69e457dcd0aa42d6af6b58cb63c09df4 : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9d43edd27f354f889375e8a5c84e6c80 : SUCCESS in 10m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5c4e3374aad54ebd8e9df18e1654cee8 : SUCCESS in 5m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/34a8cf0fb4b94f63b56dbcd2343174b3 : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ed1cdfd760ff4c0889557e1141706689 : SUCCESS in 11m 27s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4b23a3622e3a4365b21821ad34e69cb0 : SUCCESS in 19m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0b3bf9097a9b4841a1fb1b7823f16869 : SUCCESS in 7m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/56ac09e1c7ad480c9745d78c805f9663 : SUCCESS in 12m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/39f566fe48c14419b1943b07f81ccc1b : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/37932d8aa983466c971e7c936420fea2 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2da2b745664842778dd91a0520fefb8e : SUCCESS in 24m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/701a8cb3fa2644e8932e09252c76bdfe : SUCCESS in 8m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/905eaec5855a48f3b0fc29dc441553a7 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c7330621dfa74ab7b007dedc02447346 : SUCCESS in 5m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5a042dcacd59428c88c4a762bd456640 : SUCCESS in 16m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70817b9c7a554eab860b26532eb4584f : SUCCESS in 1h 06m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3297e6b2e0bb4a6bbd0cb5731f961474 : SUCCESS in 1h 44m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0b9728c64d0ca07514646d152bf20405b1f9d24d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-21 16:27:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02b9a50aab1ea9526c805cc6a4b08480c5579d90","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2025-02-22 02:48:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d41a9898cefb9e6c276992e58251c76940c656c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 04:25:53.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 942449,1","accounts_in_message":[],"_revision_number":1},{"id":"0bb09621b5aefdf0bc4b7baf0cac9f680c6a7979","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-22 07:47:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6a937085c57e62f0224fae494b32f5e636984467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 09:26:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74e7a6f9b21f4b72b4dfb71e5bce2692\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31ab74e02a147bfa83beadd9347d53d : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df62356f6365486abb539e57934b03b1 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c68a0967e3c644c1918766035fb692c4 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e0f4eb4831b459fa86ed9df8a773f55 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd30c307704b4f58b4a385c2db5a5cb3 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0fd6c634780242c8931e422251263dbb : SUCCESS in 2m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/54981de7bbba4b63a0c15a0d2a600281 : SUCCESS in 12m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b9e074d775f34716bd850984aeef860a : SUCCESS in 7m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/438fbb5ee0d842afaf79d062d5b7a69f : SUCCESS in 16m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/15850c82f7384dc686c47ebc0055f3fd : SUCCESS in 19m 29s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9413615d7bb54df4aeb367e0448d48c8 : SUCCESS in 4m 21s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a1b62b27f5724ef2a10db469ea76da0d : SUCCESS in 3m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0fe571cfcf7345b88898208eaa678bc8 : SUCCESS in 3m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ff2f0c125a6244b6968742c3e40fe7b2 : SUCCESS in 10m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/828e7312f3834c3ba5d0a8443d17b7c0 : SUCCESS in 4m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c825d6d977ff418c8c6253e5d9e888ee : SUCCESS in 5m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cd11215463c94aaab33afde571609369 : SUCCESS in 7m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fe409bc4d87a4d07b8f122e7c784bc30 : SUCCESS in 13m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23b92b73478c4903971c87450dc60cc8 : SUCCESS in 7m 03s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7db8506dda8e4d07b402e52370661b47 : SUCCESS in 9m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ae1e01ca480a4f02bb14b17eafd50f21 : SUCCESS in 14m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/aae652e5a6054d62bc0b67156f78e335 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7621899ffb5b487984aa8214c693d3f6 : SUCCESS in 12m 03s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/dc498bdbdc3043feaf91369136b4f983 : SUCCESS in 4m 18s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/18333609c59a4466a43b12c67f813c6d : SUCCESS in 6m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/540347cd09a243d08d8bcfac6b7d5d0d : SUCCESS in 13m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/eda693eae72b4374947396452619f9aa : SUCCESS in 19m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1dfa27e6d6b2462aa7e336f335291144 : SUCCESS in 10m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b8f5890b21314e2ab5a0b5af127fb508 : SUCCESS in 14m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a9a960a15755494eb4b482fd308f1abe : SUCCESS in 6m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2e40dfccb0854995958d950e576787eb : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e77c22bc80b54efca41209fb979f165c : SUCCESS in 30m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/da15e876388e44b2a6031abcb9dac2d7 : SUCCESS in 6m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fdd04a1fadc44d4a936fa9efdf4250c8 : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aaf1187503304374932039d134dd0f4f : SUCCESS in 4m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5d196071be88400eb8fdfeb584642543 : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f6678cdaab344659d637368e4af6420 : SUCCESS in 1h 06m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4ffff9a42bc4650b84f81194ee51baf : SUCCESS in 1h 37m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d318ccaf495b692cdadcafb754cb65720902316e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-24 09:09:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2046b1cfa25ba1e23e9ec3bf89f4eb5cc067a46e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 09:10:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"02b76bda3e5539fef4d6bae47cc1a693572c7a44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 11:22:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b37d9665a8ee4bdf982440b5127126dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc296e25f5fe4599858084294a1b62ae : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0bdc70110544f85a89d782f2868d401 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ce1de8cecf94948881b1bd4e0525320 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17e7722adf354d10a4fbc09317530260 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0684e601a1614537b865b13e08890b66 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/963d15e4ef0f4923b7e1845d51c9fb2d : SUCCESS in 4m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c9b85b29614b4f0b8ccd492d9d67ab1b : SUCCESS in 15m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/98085a3af0f542f98a3cbdddec9f8108 : SUCCESS in 16m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66994e160c2644519c3d100c0d7e8c37 : SUCCESS in 17m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/79e987573a4e40efaa6cfd347f440dd1 : SUCCESS in 25m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d04c7fbf0bc14eb5994df873d6c11b28 : SUCCESS in 13m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6cf420fbd51e4869b11076eec73e2fd0 : SUCCESS in 7m 28s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/781603e91edc449f832eedda5933c7a8 : SUCCESS in 5m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c2d409e66b474d18b582d961cccb9b32 : SUCCESS in 7m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9ac6d41dc54c4c3999186f68efca6a0c : SUCCESS in 17m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8dbd27b2881404cacdcb294bcd815a0 : SUCCESS in 6m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bac8a3ec029f47f6b37010ef6fe2729f : SUCCESS in 6m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3e05358596f44002bf0f50a9bfceb60f : SUCCESS in 14m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9d2bebc84fa34603bd5c5d1518b4214d : SUCCESS in 3m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5b61957542914e229aa3dffc2292a94d : SUCCESS in 11m 48s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/674cc03a24b64072bc564a24d9d23087 : SUCCESS in 17m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ab22ca88bd2f41dbbf810cb0eb2ec9c2 : SUCCESS in 7m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/57570d6da3ed45c68652087e91132afc : SUCCESS in 11m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5235caa23dc54c5e9720f5d1d4dffa0c : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b7b60d2d5c654d3cb3f05c109320ec49 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24807413912e4ba1adae8aaf222d95b8 : SUCCESS in 30m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/be595cf7eefc4677b8be2ffedadb5058 : SUCCESS in 8m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f1d36181473e418783f4c76fb9e4d223 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/97968b2179644bd58979c5dc6bfbd99e : SUCCESS in 5m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/97774e9d70744e08b8d9c9c7f1cc275f : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1eaf4308f5764d9d99dfb789f9f0f407 : SUCCESS in 1h 04m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc5df6f132794886bbdcc51f90644fe2 : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ad22ab85ef63f6e6018b04e6cb73e925a75cbc32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 11:22:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"424e4730f14e3efcd8c9450203384311200a88f0","revisions":{"77ccfcddfd38da5891f0ea6403a3d8ac5d03285d":{"kind":"REWORK","_number":1,"created":"2025-02-21 14:04:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/942449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/942449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/942449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/942449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/942449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/942449/1"}}},"commit":{"parents":[{"commit":"a75cb655fe22ba8410ef3a185c57c32a37a1c294","subject":"update constraint for os-ken to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75cb655fe22ba8410ef3a185c57c32a37a1c294"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 14:04:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 14:04:39.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 7.5.1","message":"update constraint for oslo.rootwrap to new release 7.5.1\n\nmeta: version: 7.5.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I874c2a25a5e789b97a5fb2a8b4b66956c8caa1ee\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If36b9c727a5d42ecfb4e557bf4056e20ca2b1454\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77ccfcddfd38da5891f0ea6403a3d8ac5d03285d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77ccfcddfd38da5891f0ea6403a3d8ac5d03285d"}]},"branch":"refs/heads/master"},"424e4730f14e3efcd8c9450203384311200a88f0":{"kind":"REWORK","_number":2,"created":"2025-02-22 07:47:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/49/942449/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/942449/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/942449/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/942449/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/942449/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/942449/2"}}},"commit":{"parents":[{"commit":"38d7b2220f1752494717880f67eb4d07866092d7","subject":"Merge \"update constraint for oslo.policy to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38d7b2220f1752494717880f67eb4d07866092d7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 14:04:39.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-02-22 07:47:32.000000000","tz":60},"subject":"update constraint for oslo.rootwrap to new release 7.5.1","message":"update constraint for oslo.rootwrap to new release 7.5.1\n\nmeta: version: 7.5.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I874c2a25a5e789b97a5fb2a8b4b66956c8caa1ee\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If36b9c727a5d42ecfb4e557bf4056e20ca2b1454\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/424e4730f14e3efcd8c9450203384311200a88f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/424e4730f14e3efcd8c9450203384311200a88f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
