)]}'
{"id":"openstack%2Frequirements~942450","triplet_id":"openstack%2Frequirements~master~Ie5a227a27b4c3f3be212e1f6f72fc70448d244a1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie5a227a27b4c3f3be212e1f6f72fc70448d244a1","subject":"update constraint for oslo.upgradecheck to new release 2.5.0","status":"MERGED","created":"2025-02-21 14:07:39.000000000","updated":"2025-02-21 23:33:45.000000000","submitted":"2025-02-21 23:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942450-new-release","meta_rev_id":"091a54da8be7562bedc38f0546c2d262e5dc3bdd","_number":942450,"virtual_id_number":942450,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 23:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-02-21 15:57:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-02-21 19:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-02-21 19:38:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 15:55:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 15:57:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-02-21 19:38:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"9c06b200a156c80cfcddcf6c6c0e1fe0c419b340","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-21 14:07:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d192ded5cb54fa44e00363727c0bc350785ba71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 15:55:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1c3f329a6bd4c6ab64743689d75268b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca943226c414d8d8b57021a123600f5 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae682c8158de42528e03ddd8fb0c9f97 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2db02eebe0a94326bc843b29e7858570 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7aae90db05eb475bb83487a4f3db2fc7 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/338ae4846b6f401d8bc9aa6df107c206 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a507bd7ac1040c4bb8ed08a91d1c947 : SUCCESS in 2m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9ea31da78317416e9a7b5459f1227eb5 : SUCCESS in 15m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7eeed51d2c524ea3980b4be779bca93e : SUCCESS in 7m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3557825bd2864c94831585345ddfc916 : SUCCESS in 19m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e17e65fdd26442b3b5d500d0f17dd275 : SUCCESS in 17m 36s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/60cdb1c8093c4103a532632b72de22cd : SUCCESS in 4m 56s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6b6d51fd23ee4aa4b02ed5064167ac7d : SUCCESS in 7m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/28089979707143fd990fec6cc6a4111d : SUCCESS in 3m 54s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d5ecc7cf7738486085381f94db222425 : SUCCESS in 9m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4c9d0ad5c1994b9aad89c4d85a7c25ed : SUCCESS in 4m 53s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ac072451d8c84f918651196395b2f4db : SUCCESS in 3m 40s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a64aabd750bb48dda590947cf5374fdd : SUCCESS in 7m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ba611fce69b347afb2c76b550041171e : SUCCESS in 6m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19f10c61382d484aa1b9dc8ce0eaf71f : SUCCESS in 7m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/79d37633a4af4341890559afad01395b : SUCCESS in 7m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/33725aa131f24ab49766fdac8748d64e : SUCCESS in 9m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5354833d415441f9990b1e78078fcd79 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/23248316b9e24ddaa80b2a5db1481e7b : SUCCESS in 10m 34s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/36d3e2c4093a40daa052b28f06db06c8 : SUCCESS in 3m 08s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/930015272b1e4e618ba944746f6c9051 : SUCCESS in 4m 41s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a1e880ee5d5248c48fbea0739f7aa86f : SUCCESS in 12m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bace698ddbb749f78e98b8e2d228cd2b : SUCCESS in 17m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/956d1ca67965473f9bb8ab26b730b8c9 : SUCCESS in 7m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c8a3193be8e346dd97d5276f24c67822 : SUCCESS in 15m 42s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/580852ce835c4e0e820ec7183bc59e90 : SUCCESS in 6m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/77558818aaeb473281d7ec048a140d88 : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/64f1995984f2436381edc33c93a344a2 : SUCCESS in 24m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d2673ba030e74547b79714a6b87f1148 : SUCCESS in 8m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/47841b96acdc44c18c36e59c6cf509f6 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d2c740fe80a8473fa780f2788abcdc3c : SUCCESS in 7m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/221497a74a464862b7b1780e9c03c1bf : SUCCESS in 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb239918ce2c4a9cab073785b4257a07 : SUCCESS in 1h 10m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/413cd1994036445fabf099ffec545cd5 : SUCCESS in 1h 38m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cc1fb1f088957cff4bb8436b396216a7040c5128","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-21 15:57:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a48d94e0f362a15335b486115e0808933a159bf","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-21 19:38:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7ed51b7795499e241b90b941e798986088869ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 19:39:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ab370bb1b0ae7e57643490495e87941dc8f7249","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:32:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85c0badec00b4c36a2dfd13c1b808c68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7aeab27ffb64fa9872acc0001b5d4d4 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b5401c40ccf490285348ff599f04bab : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e38403943aa4785932b8693de2b0291 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44b59cfc84294cb6be8ca0ce217903c9 : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0edf76aab1d34fb89ce548607c60fb0f : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36bf714dc38c44f88bf028d0bce24e55 : SUCCESS in 3m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/33ecd624159a49ce8fe1fd5d692eb5af : SUCCESS in 15m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/591bb5555c39444193a580fb41f54fed : SUCCESS in 5m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7d88e7f6b9a64c82a90d840ccf5c9a4e : SUCCESS in 18m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ac64471979904bdcba6e4e46e1814fa9 : SUCCESS in 28m 09s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ba70e6bd3f2b432eaf2fdba5e975e18f : SUCCESS in 12m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fbe01bc3d29f4881a7cd49f45f673d6e : SUCCESS in 5m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/54a30b1c5b5845279aca6033a1ac080c : SUCCESS in 7m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/31ea9c70998241d9a52230a3603a60d1 : SUCCESS in 9m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4e8ad6f973db4604975db6e394d875e0 : SUCCESS in 6m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d7649c84cb946649fbe3f1758560eec : SUCCESS in 8m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/02cfe00dd1414d4f8e53091432f224e5 : SUCCESS in 14m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7629fdc7808e48e0a474e6f9241efd15 : SUCCESS in 25m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ab81d2c4fa5748078266f8280a4b8138 : SUCCESS in 1m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/70f796bc177b4f2eae0837547e786c2a : SUCCESS in 9m 56s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c3d643b40a2a4cfeab2b1dbe94af0461 : SUCCESS in 25m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c1068f7cb7d1425c898cd7fcb4757a19 : SUCCESS in 8m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c9e806738bcc4f8e9367da3307ce34da : SUCCESS in 13m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8e0d134970c94f42b2327fc2fefb826f : SUCCESS in 12m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/15ffc492c7724752ba4a4a8d707a7209 : SUCCESS in 2m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7dcbb99de2de4eaabc19873c44a31a5b : SUCCESS in 25m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/958e405b5fdb4e08aa75a1e632bb08a7 : SUCCESS in 14m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d4d2cd5e5254462b8795d02e3d57208e : SUCCESS in 3m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f56d9ae203b43b3abb454652a7a0a55 : SUCCESS in 5m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cba1e374db514b199d916a6080a21ca6 : SUCCESS in 16m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae4029a5f6a94c74a7f487f40904d86c : SUCCESS in 58m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff5a2b9a6ec84a90a6b6133b546babab : SUCCESS in 1h 34m 41s","accounts_in_message":[],"_revision_number":1},{"id":"891954fb1ca7fe09d0718ae95641fc124561ccfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:32:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"091a54da8be7562bedc38f0546c2d262e5dc3bdd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:33:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/946ac3cb54e4465f8fd6bedd74f4e318\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/340fc9339be042f4ac5c0945833b876b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d24ae758a5a38aa594f49dcf3b88140a831ea4fe","revisions":{"d24ae758a5a38aa594f49dcf3b88140a831ea4fe":{"kind":"REWORK","_number":1,"created":"2025-02-21 14:07:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/942450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/942450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/942450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/942450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/942450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/942450/1"}}},"commit":{"parents":[{"commit":"a75cb655fe22ba8410ef3a185c57c32a37a1c294","subject":"update constraint for os-ken to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75cb655fe22ba8410ef3a185c57c32a37a1c294"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 14:07:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 14:07:38.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 2.5.0","message":"update constraint for oslo.upgradecheck to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I2919c0e04539b6cc4dbf033956d9646c7c1e0f0e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie5a227a27b4c3f3be212e1f6f72fc70448d244a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d24ae758a5a38aa594f49dcf3b88140a831ea4fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d24ae758a5a38aa594f49dcf3b88140a831ea4fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
