)]}'
{"id":"openstack%2Frequirements~942453","triplet_id":"openstack%2Frequirements~master~I2050585f1ddec99e3b37fe241cdfe8bff75b4f31","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2050585f1ddec99e3b37fe241cdfe8bff75b4f31","subject":"update constraint for oslo.metrics to new release 0.11.0","status":"MERGED","created":"2025-02-21 14:17:52.000000000","updated":"2025-02-22 00:05:38.000000000","submitted":"2025-02-22 00:05:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942453-new-release","meta_rev_id":"8b3767b42feed8722d3cec0500e0c29534098ea8","_number":942453,"virtual_id_number":942453,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-22 00:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-02-21 16:27:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-02-21 20:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-02-21 20:24:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 16:06:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 16:27:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-02-21 20:24:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8423e7ea71a911d2bbba24896b5612669c66a781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-21 14:17:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c06988b970af637bd0ac9cbf6a53c2611edad95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 16:06:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81538ba4ed844077bdc24473d6c6a0ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0efbd5af983344a3986581c9fe21633a : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac7180b7ed940baa486eaed967a0c64 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d53f1bc5e2a7452bbf3760391cf863e5 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d838db741ded4b1896bb2c1764607149 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/005cfc8fcc0046a7bbffaa188bc2de2a : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a23d723b5eb4c03ba62ba85ea8f82e7 : SUCCESS in 3m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d45e7431fbfa493c867ca12e00dc79dc : SUCCESS in 15m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f651275eea184740bc91a0895a593ed5 : SUCCESS in 10m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fdc70bd31daa47fc8ceead64e4af80d0 : SUCCESS in 12m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8bf9f637fa2a4d41b6378ddb8120acf5 : SUCCESS in 19m 02s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/183c63b5b6fa4ef0923f97bb1e2c958c : SUCCESS in 4m 45s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1d1c148f25824483af071c0240714686 : SUCCESS in 5m 02s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2ff60f224561459f8c8fdf2b7d7a1237 : SUCCESS in 5m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3b4b03e6c2904611ab87448d83b5680f : SUCCESS in 12m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/461a31563d0c4086ad5eac562a06c33c : SUCCESS in 4m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8e2e6d47322f4f14a3ae38cc4d0c7226 : SUCCESS in 7m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f39368b8588a4319a92882f4ddcb2f6a : SUCCESS in 8m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/60976332b15c4a8588e2abef7bd73aa7 : SUCCESS in 14m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e44f8f34b5e74e158332b91c7f487932 : SUCCESS in 4m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e1185859ae35463eba49afcd39c9c452 : SUCCESS in 6m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e44a1b4eade5436ba852b109374f6221 : SUCCESS in 9m 21s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0dd5e92a8e8e415aa2160e169d6f8ce3 : SUCCESS in 1m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7ff8fc4185ef496c9a6faaa8ed593211 : SUCCESS in 7m 44s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/95de32755fea487f801b429df26efc61 : SUCCESS in 3m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/819385f974c54fb1ba664ab5311c197d : SUCCESS in 3m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/98c0dcfdb64a4de39da60ad1747869bb : SUCCESS in 11m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/79d880e82dca4bebbce9da4540ce1939 : SUCCESS in 21m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bd16084cb38c4502b02e89b519a4585a : SUCCESS in 8m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c10f956f57ff43a482de0d6fbd3568f0 : SUCCESS in 13m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b5340d223dd045cfb78c3b777e7ceaf4 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dac10f5a38e04a9fb806950a11be07cf : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d7b3b9ca5794339a9a31f6cda1582b9 : SUCCESS in 26m 17s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bf1e866c405d4b419e50541b2239b32a : SUCCESS in 7m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ef28870dd01549a19593443ab1c1cbd0 : SUCCESS in 3m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de3dda6b03f04acab58cbb4a6dccde86 : SUCCESS in 6m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c9daa480603c4a3db0f66b8828471cda : SUCCESS in 16m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9185a7b8fbfd46d69a709fc5bac25923 : SUCCESS in 52m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/997959f9aa4b4efaae6c6f391d2b99bf : SUCCESS in 1h 43m 26s","accounts_in_message":[],"_revision_number":1},{"id":"51be35ec7b67b84eaf5e389a9ca41b9cdf24a358","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-21 16:27:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2413dd506004ba694313d1d99ea2fb2389a6950e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-21 20:24:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f31034d1f75d0fbe6775c6a51d61720525534e1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 20:26:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c52638580e52b6debdda7d869d681fb64f297a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 00:05:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b572a5d1bdc4e519dde6dd9a6fad83d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40cad633c1e44fe99a0df67ed79add2d : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b2c054dec44e1daff9162b2b68ec32 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fa063ea8ac14f3b9397eaa08aebe6eb : SUCCESS in 8m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55d72097d4af41a69a530b6bcbae7738 : SUCCESS in 8m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f757f26a727e44c8bd8654ee0c120d82 : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a3085333d6d43379749c7dacce94f4a : SUCCESS in 2m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2a2c2aadbda94af7b6341a9104699276 : SUCCESS in 15m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b760bbdf60704e83951d82b4d7781eb2 : SUCCESS in 11m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c6c8ffc514894834b978b8c07c1a5ca8 : SUCCESS in 15m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a834c52f001f4c50b870ee4db6ebb025 : SUCCESS in 29m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fafc69d6cacb4bd5b17305af86fced04 : SUCCESS in 10m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f4fe29efb5514ab5b60ff1df48b13fba : SUCCESS in 6m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/10fe6d960b5a459db6281046a28a82d3 : SUCCESS in 3m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fb3f16a73ae24111bd55f30586e5dd22 : SUCCESS in 10m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/060e7fcee9d64fd9827074abd9c20a3b : SUCCESS in 14m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6486a8f74c4b498ba1704d21745fd418 : SUCCESS in 6m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/47676d87f0b74f3099b7a5c9aa72b7a3 : SUCCESS in 8m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c72c8d4563f049539dc08c4b07b6f821 : SUCCESS in 9m 14s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f7cb6b0f1cb042cfbd295a01042de511 : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c7a70e9085904513b981c60f39e33d85 : SUCCESS in 11m 10s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/40751cf9f7a84a5c9ab7785bd72e70c3 : SUCCESS in 23m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/86c9aa8366594f79ae276f4ec68b693c : SUCCESS in 10m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/03e9b6d64c1d43ff8103e0af70f975a0 : SUCCESS in 15m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f2c77b7ccc6b491da937c8355919b772 : SUCCESS in 7m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c42c1f7993f1496a8d37cd6461558734 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ece12f49c8245b2a19eae25d18c69cc : SUCCESS in 24m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1f821c1572d2497fbcbc8c22c8981802 : SUCCESS in 4m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8ce1376ccfe24b9784292b636fe0de7c : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f486fe1a615b4af1b8d78c885e57fa97 : SUCCESS in 7m 32s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e0752f03b4ad462a9a2cf45acf1862c7 : SUCCESS in 20m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fa5cf5967ce4978b93fbfd99960e4f0 : SUCCESS in 1h 08m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11004a4ecfbc462e9823dd286e9c8085 : SUCCESS in 1h 36m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8b3767b42feed8722d3cec0500e0c29534098ea8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 00:05:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50b8b9b3c5949555506a565b4a4997ed651e8c12","revisions":{"50b8b9b3c5949555506a565b4a4997ed651e8c12":{"kind":"REWORK","_number":1,"created":"2025-02-21 14:17:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/942453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/942453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/942453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/942453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/942453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/942453/1"}}},"commit":{"parents":[{"commit":"a75cb655fe22ba8410ef3a185c57c32a37a1c294","subject":"update constraint for os-ken to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75cb655fe22ba8410ef3a185c57c32a37a1c294"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 14:17:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 14:17:50.000000000","tz":0},"subject":"update constraint for oslo.metrics to new release 0.11.0","message":"update constraint for oslo.metrics to new release 0.11.0\n\nmeta: version: 0.11.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie360ecc1840e3469c7190cda19295ab3e2de3b6b\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: I2050585f1ddec99e3b37fe241cdfe8bff75b4f31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50b8b9b3c5949555506a565b4a4997ed651e8c12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50b8b9b3c5949555506a565b4a4997ed651e8c12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
