)]}'
{"id":"openstack%2Frequirements~942476","triplet_id":"openstack%2Frequirements~master~I8dec34082e8a451f617348ebdbf4cbb267dccabb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8dec34082e8a451f617348ebdbf4cbb267dccabb","subject":"update constraint for octavia-lib to new release 3.8.0","status":"MERGED","created":"2025-02-21 16:27:30.000000000","updated":"2025-02-22 00:01:00.000000000","submitted":"2025-02-22 00:00:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942476-new-release","meta_rev_id":"338ca8d6934d4d3eec25992549cd466d75981e51","_number":942476,"virtual_id_number":942476,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-22 00:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-21 20:22:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-02-21 20:22:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 17:49:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 20:22:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"4a4bc11bdceea84adb4f1719b591f4cf375609d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-21 16:27:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"548b36fe5e157caf9075a02638081b74163fe3e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 17:49:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/526f004026474fce974653a0bd3f55c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/126b56fb14ea4f9d91f30d148c16d552 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de4d766f09fd4d1b9b2856b99e90958b : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2ed9eec38d842f2858bd099ab9ce18f : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8a6b2edb8844f3abe81ad5f1304a898 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3612e6c8dae3447ab2530a2667a2fb3f : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd6e801cc61746618c130069a4a015e5 : SUCCESS in 3m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/38236b82dbbd4930860fca59d01c10bf : SUCCESS in 17m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ff19bec9683442afb00842bd11e1133b : SUCCESS in 8m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0792b661b1944df480344605514aee6d : SUCCESS in 20m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/450006a2c38a40a8ae64e49f95862580 : SUCCESS in 29m 51s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/51b3ed2b626042028372db32016df1c8 : SUCCESS in 6m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fd961a55b01e4182870c2071bae8c96c : SUCCESS in 6m 00s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d0e00a9822b241c2a3d538522d0fbf6e : SUCCESS in 5m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/280b58cf70654efe98bf4fe265eb00f8 : SUCCESS in 14m 37s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f1f4674edbdb49e1abaf40c14ab43dfd : SUCCESS in 6m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/652a53253edf4629a1d6efe5adbefe96 : SUCCESS in 3m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/62280f94c3414788a69050cc610a9a85 : SUCCESS in 8m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ab18fdd7875a41a3bf3a568265067c8d : SUCCESS in 12m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/02adb91c355b446ab5fcc17ec07e4a93 : SUCCESS in 6m 20s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5fe81b19f88f462fadab6300b169c355 : SUCCESS in 10m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/12b9a4f73bf84d21911c4862cf31425d : SUCCESS in 14m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e0b8c3c7cecb41bc8b5417d1b8443e02 : SUCCESS in 2m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b789b2f6b2f94a3b91ac6370360caf9f : SUCCESS in 11m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ff619bc4e49b4be4819eb0251f733f12 : SUCCESS in 6m 01s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/aabfaee1845f4c7baef42d71361acce6 : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f027d8d1bdcc46f5ab169ac260be5af6 : SUCCESS in 15m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fd8094dcbd7146c09978277673767714 : SUCCESS in 18m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4a2abe602dc44ba3b95fe92511958f3d : SUCCESS in 8m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1ea807d1d6dd4539a8c63a51f0128f38 : SUCCESS in 14m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/16af39b2dcea45d3abebf2d65ae24fca : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99947c63652f431cb542c31dc9879953 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65ed04d0894d4abaa6fff85be71fddc3 : SUCCESS in 27m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/828901083b4249f196496f7adce9d63d : SUCCESS in 8m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cade20366e7e4443974aacf818c10b84 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a84529af2d0b46e0b2ab96f2f7b69895 : SUCCESS in 5m 14s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4115293e150948659a8a69144f6e9633 : SUCCESS in 18m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/deac2db5b31d4e67b4cbfce755dcdcaf : SUCCESS in 58m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0769990613284b049083461a6ba56b03 : SUCCESS in 1h 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d1de3d97bcdaf032396c3d915888f06c04426146","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-21 20:22:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8c122007e476f717a96ed916176e1252a57b9ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 20:23:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9463156f126a0a3de8598d623ed4d02cb43f68b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 00:00:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4999d1aebb824397ba9259eaee0180e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f95883cb9f44c57bd8cbe469769902d : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/300cee9084994794a6e5bbdf84b50a95 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/189deaed85694d458a01efc673bf86d4 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e945a783ef58424ea97b0a48ab68b616 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/458096b7732548d6ae832e5a3c8c292e : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c68353eef2584845afcd68ec147cba87 : SUCCESS in 4m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/570c380cd0844674a54948e6801f7b1b : SUCCESS in 7m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d0942f59d4ca4420883d4b83c96a6b7b : SUCCESS in 9m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7601d994181346c0be0e5fe51c3930c7 : SUCCESS in 17m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2f1fc5d7690d4bffa92144a41d62f8f6 : SUCCESS in 31m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/10d0d104264140d5a125dd5fd34dd87b : SUCCESS in 11m 11s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a2dabcbd669f43039032f9355bf657f4 : SUCCESS in 9m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/85693b50895842148b8b712b0a649c02 : SUCCESS in 6m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5846de1b3c6748caa45d75fc7a59a3cf : SUCCESS in 7m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/94a3780de6bc40a69ec0f58557493383 : SUCCESS in 15m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ef0468abd5f418b9b741c071de2dddf : SUCCESS in 6m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4d9e878273cc494a85af75d0df270967 : SUCCESS in 9m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b3fc6e1334304611b89c9d72535e556d : SUCCESS in 14m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e18ae12464e24935916acf4bfb530cf9 : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/79c6fd4327b14d5fa20497669f0c4f23 : SUCCESS in 13m 37s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c1065b8dbcf148b6a77e86c8d6681dbd : SUCCESS in 20m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d00a34bf5a9c4f2dbd0d7391452fe9f7 : SUCCESS in 9m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7c0d216f354c4932b3692b02401c5f39 : SUCCESS in 15m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/71c00c2efa8f455092049074c4c021a4 : SUCCESS in 7m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f49af7cabdda41eb9f6b35cccc4c4820 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/098e353482144a9882b23ab4438723c6 : SUCCESS in 30m 46s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5074fcac95c847e9beff525a6286aa41 : SUCCESS in 6m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8cd6993c81314a27aba146cac880bbc3 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f5afc0764fa4b27b64325da0e13ca54 : SUCCESS in 8m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e68cc6eb879c4037a1735a03607fd864 : SUCCESS in 18m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e72e70cc7e1d4782a5dc344037e94c1d : SUCCESS in 1h 06m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/667631ea81c54d9aa24159a0378b0766 : SUCCESS in 1h 54m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2138675d28669f4f042494deaeeaccea9971d84f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 00:00:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"338ca8d6934d4d3eec25992549cd466d75981e51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 00:01:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71bb416f5c0344a9907f49fd22d427cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb3c381645a14ac182c18fd66446255f : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd57c18acd8d5035543e9719c666a4ee2e4c5136","revisions":{"cd57c18acd8d5035543e9719c666a4ee2e4c5136":{"kind":"REWORK","_number":1,"created":"2025-02-21 16:27:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/942476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/942476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/942476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/942476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/942476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/942476/1"}}},"commit":{"parents":[{"commit":"a75cb655fe22ba8410ef3a185c57c32a37a1c294","subject":"update constraint for os-ken to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75cb655fe22ba8410ef3a185c57c32a37a1c294"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 16:27:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-21 16:27:29.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 3.8.0","message":"update constraint for octavia-lib to new release 3.8.0\n\nmeta: version: 3.8.0\nmeta: team: octavia\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iaa8f83eac0bd7d93cd58ec69e64f0ffd00e34eb3\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8dec34082e8a451f617348ebdbf4cbb267dccabb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd57c18acd8d5035543e9719c666a4ee2e4c5136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd57c18acd8d5035543e9719c666a4ee2e4c5136"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
