)]}'
{"id":"openstack%2Frequirements~942571","triplet_id":"openstack%2Frequirements~master~I614bae4d81f34f297167439acdbf0d93a5031d33","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I614bae4d81f34f297167439acdbf0d93a5031d33","subject":"update constraint for futurist to new release 3.1.0","status":"MERGED","created":"2025-02-24 10:57:58.000000000","updated":"2025-02-24 19:27:44.000000000","submitted":"2025-02-24 19:26:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942571-new-release","meta_rev_id":"172e52143f795e0339e5a6b9a56c3af448ac8ad0","_number":942571,"virtual_id_number":942571,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-24 19:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-02-24 14:18:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-02-24 15:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-02-24 15:40:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-24 12:37:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-24 14:18:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-02-24 15:40:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f0c834d0ac573969ab6fd53c1a1bfdd1ab117507","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-24 10:57:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"617833bba1eb696d19e441de64518c02594a6581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 12:37:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f76365ac174c7a97b28263254a1655\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e83018364705493fabbd83807c0e9383 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70119d46742f416c8ed61701c5d4bbc4 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b39d1dce4214dbb84aea5497f852024 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2c6037b36ae4d6eb368e23e21c7ff7a : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5601da0042244f39b80ee6c0a40259b : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e44da1dda1bc4747835d775d2487a658 : SUCCESS in 3m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c18c743cbeb64388947d6d28168d2135 : SUCCESS in 15m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/eb76f4ce2c65423a8ad361574ce468ae : SUCCESS in 8m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b7f0274c2f6f45b68f5cf332a88e5d61 : SUCCESS in 15m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2d5c35f3722846a7af831954a2492c44 : SUCCESS in 20m 48s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e301b9bbe83f46da9c00ce808622e9e2 : SUCCESS in 4m 33s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/669f4bf6a83e49a18eb1aaa40cdb0393 : SUCCESS in 4m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9cdb96f4a5d14b45bc911291c1d0d85f : SUCCESS in 3m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2258b500f7934132813a9867a48d2f9c : SUCCESS in 10m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1dbac4555b704ef8a96ca38771157c97 : SUCCESS in 4m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/61a961eb824f4894ac630deefa8f9ee4 : SUCCESS in 5m 22s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/aa9b6b4b2978412bbe4c44f8612e27ac : SUCCESS in 7m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6a79ff5f0cd64555bde273c037186779 : SUCCESS in 13m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fdfdc3a8167b4456b2f822c745df9e93 : SUCCESS in 6m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f4b7222f4ea348f19176613364c0b231 : SUCCESS in 8m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/307f135675124cf5b664241bd0c2a11d : SUCCESS in 12m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7e43723f236c4ab8978247173d4b64f2 : SUCCESS in 2m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1779142d96c54e1a8a72b287f49eb8ed : SUCCESS in 13m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2060c4bc89314ab7bc1043602c6fde19 : SUCCESS in 4m 50s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/123d9daa46c54c10b33e740199bd0d34 : SUCCESS in 7m 30s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/26379871d1854b679429057a1514c1e1 : SUCCESS in 11m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/90e43e005ba943c78c888ba5d19a7919 : SUCCESS in 18m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b49e5da532734e028b1260810ab6019e : SUCCESS in 8m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d1184112b0c949bab5dd3456de79cdca : SUCCESS in 12m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ea426bce6b3340f4aae84cc709491b2b : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/010dbc4d2c31499a9c60fb80ce16764c : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/847cb6b0eae4461290682e4dae7a526e : SUCCESS in 21m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fc9e638001d54cc49ea0d9042d74fc8f : SUCCESS in 6m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/afccb29e8f454fa5bd8ba54a4f277877 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8642b786765147b7a1ab10bf588a2ef3 : SUCCESS in 5m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f612845dd2924f3c8ab485a68f780796 : SUCCESS in 16m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01a532d643e3430bbdba3f532c70ce34 : SUCCESS in 51m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60c2a41252994189b2332159c74c6300 : SUCCESS in 1h 34m 01s","accounts_in_message":[],"_revision_number":1},{"id":"228ce268ab9172136a88776cd197da02feeb067d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-24 14:18:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b320f548e2d39e90cf749ed2e1ac65ffde23cbd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-24 15:40:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a90995ba4b908c8d2193a97ec785a493c5eaee40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 15:41:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33b8efeb5b69de9ce0e193fef53778916e8e8e8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 19:26:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1bc62117da64e85b0fb141ee2c5e13f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eafb71d127694f798dfb094bc42dd69a : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60d68f27bad14cf897d01fc54a12becc : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a1a51df28904e828913677b25539dd8 : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c4331206c994d5f9b66c74a3a4b825a : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86eb3ad8ee2f429998c0cae65fd573cc : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f005403665e42a0bf3496cecbc1b073 : SUCCESS in 6m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/139980873bd1497cac93e9263753ef50 : SUCCESS in 14m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/45024d72a05745c2a5446aa81f048a40 : SUCCESS in 11m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c47697ec8e564ece81f19892080b161e : SUCCESS in 8m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7ac32c0f2f7445fbbdb7903098d2c6dc : SUCCESS in 27m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/88d0a617ab574d8ea133fa22d2da4468 : SUCCESS in 10m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bbe411566fb14645ac2d38f9118af36f : SUCCESS in 6m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/de5d293724c5411192ba8b98d18f5777 : SUCCESS in 5m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e215b8cd4b1848ea867e30a54ddbdb09 : SUCCESS in 9m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3f65747c63314de289e93a882e298217 : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b710a7216aa4f63bf8f47c86c09171d : SUCCESS in 14m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b239f50803604260bfc34039eaa70be6 : SUCCESS in 8m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/64cabdfc67684ac79d88c63d73fa1dbc : SUCCESS in 12m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1e614b0fbea044c68d532142cba16ae2 : SUCCESS in 2m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f430b1ddb1db403586f384f8987ff3f0 : SUCCESS in 6m 37s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/07f572a26abe4382997af61e2f609bdf : SUCCESS in 19m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e2c036e8ae9b495c847909e9d788d8c2 : SUCCESS in 7m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ef387658f8e9408a8500587a4de7adfd : SUCCESS in 10m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7115d15db4954870bc63ec7a8148e5cb : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc013c6180cb4ab78de66d91d2152399 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7794fa91a1e4e48a28bfbf81f2d540e : SUCCESS in 25m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/619ecbcec67944699a6e4ca2a98f0ca3 : SUCCESS in 6m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7c6f0ea9c68b48afb777f591d6072618 : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ded7b766a0b4a65a9b75a4aae3b8023 : SUCCESS in 6m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fd769b12596b47aca65e5526dff2983e : SUCCESS in 17m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d207e802860742b490486ed042807f60 : SUCCESS in 54m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2fe2a386955d46d189dc5508b5d1a186 : SUCCESS in 1h 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"40e8209f87edd411c53e6ccb128b808c647bdb37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 19:26:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"172e52143f795e0339e5a6b9a56c3af448ac8ad0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 19:27:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f7b3d4dda443fcb27da909a1dede49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2adbf75cca4548d5bfa273c8f22611dc : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"548204007547c09ed0fddb3b84bceda159a91f1c","revisions":{"548204007547c09ed0fddb3b84bceda159a91f1c":{"kind":"REWORK","_number":1,"created":"2025-02-24 10:57:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/942571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/942571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/942571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/942571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/942571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/942571/1"}}},"commit":{"parents":[{"commit":"96cf3ab8431adaed7b1075ce6c5fd6b276c06e20","subject":"Merge \"update constraint for keystoneauth1 to new release 5.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96cf3ab8431adaed7b1075ce6c5fd6b276c06e20"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-24 10:57:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-24 10:57:56.000000000","tz":0},"subject":"update constraint for futurist to new release 3.1.0","message":"update constraint for futurist to new release 3.1.0\n\nmeta: version: 3.1.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie9376ed6df24d779c51b6c0ed65a3664558ebe15\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I614bae4d81f34f297167439acdbf0d93a5031d33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/548204007547c09ed0fddb3b84bceda159a91f1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/548204007547c09ed0fddb3b84bceda159a91f1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
