)]}'
{"id":"openstack%2Frequirements~942845","triplet_id":"openstack%2Frequirements~master~I00bf38e6d6537ac12c98c5867092f2e1379b1e6e","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-02-28 09:32:43.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-28 05:47:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I00bf38e6d6537ac12c98c5867092f2e1379b1e6e","subject":"update constraint for python-openstackclient to new release 7.4.0","status":"MERGED","created":"2025-02-26 18:36:51.000000000","updated":"2025-02-28 11:46:33.000000000","submitted":"2025-02-28 11:45:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942845-new-release","meta_rev_id":"a2a1e3b1090065fc9bfcb4ad8f14913bd132fed7","_number":942845,"virtual_id_number":942845,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-28 11:45:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-02-27 14:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-02-28 09:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-02-28 09:32:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 20:34:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-02-26 21:30:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-27 14:18:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-02-28 09:32:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3e35b287284e6890a8f7701d6f8c0ff1f9f28b26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-26 18:36:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b9b03ea30d00544cf90f7bf711d239ea0ae6ca8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 20:34:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c626b773afb79e242288c75d50853d8184c1ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 21:30:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f58bd92b9c7140f8bd32b6ef28b47e88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/777596773bd64630897f56d1aa64267c : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70009743667145d89f20f75a98aa25b1 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e97113c2bdd496cb257153ce6ff23e1 : SUCCESS in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/605cd2395f7b42ec8fa9a8fbc47679b9 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8588031aadb34edca9a7a473f67bedff : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55da8c780bd74e5c97f3dfc3e927dccb : SUCCESS in 4m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6a8f2920e57b4364bd6c32068e572023 : SUCCESS in 19m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/851d2fc48f304f56acd4a3a55d9cd616 : SUCCESS in 11m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/32b8e6cd335541888ddc013ab52205fd : SUCCESS in 16m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a1675a3c5833415cb38d547cd736fe3d : FAILURE in 27m 18s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fa5f3723f1134428856107ee3ce23632 : SUCCESS in 7m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ab75b7f39d2b45bcb9f361245b2b6965 : SUCCESS in 11m 09s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8dd94ad479c24663ac4ffe18b94d733d : SUCCESS in 5m 36s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8cb1131169b34100ac90f630b5de3333 : FAILURE in 12m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2a7ec25b1d4a4c7e837cf2a600269976 : SUCCESS in 5m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/aba68f1d1d62487a8d6512ca23423ea7 : SUCCESS in 7m 27s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bd8313609d6c4a10992fd046f1a3e810 : SUCCESS in 5m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e404ff57272e4a8b9f9f7623f7489329 : SUCCESS in 14m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/701632909926427d833110767a052f56 : SUCCESS in 25m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fcb7b05cfd8540b58521924f8f121902 : SUCCESS in 8m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3b8f859f0d054a08be5e1e2b7ca7557f : SUCCESS in 12m 26s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ef37b3c53332456c8f4337ad9af91fb8 : SUCCESS in 3m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d5b8eb800d3d447bbd4cbe1c3935b292 : SUCCESS in 7m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/03cb88e6b98240f8ba1301fe1ac2a5db : SUCCESS in 4m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1069fcd3fe5a4a3db45ddee5bb4442c9 : SUCCESS in 13m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3bbb21e099d349b18c3f1d34ace2050d : SUCCESS in 12m 27s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b3d9147945fa470fa06ae525b77b4a51 : SUCCESS in 20m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a9d2997fbd694ae9923ffff7293bf165 : SUCCESS in 9m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/04ec61492d284a888a24540a3d004e22 : SUCCESS in 15m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e0c7aff7d65f4d2dbe2ea8eda5a88a72 : SUCCESS in 3m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a9ab696f922461582f75186d0c31b77 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a716411a77424a3a8e3595787cd10184 : SUCCESS in 35m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4eb43c486cee4da599f7d469fd3f8e15 : SUCCESS in 7m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/068df7cc386e4a04877cd05098f8f8c5 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/85c4531c836a4dcd917712c66859884e : SUCCESS in 6m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c2d93d9fc54e4d8b957a108d6b731ff3 : SUCCESS in 7m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fd2e69e5c764672b855ffaf4c18cdd7 : SUCCESS in 1h 04m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78231b0660544c1c80b46fe7ac72f96e : SUCCESS in 2h 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f24d9200560e3706aa9cc38f97a96d501f7e1a3c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 22:15:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46299a2b4d841c3841e0b120fd2a101d80e6a331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 00:04:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2a5cfda09264dd3a16554255c6a8e97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22746e6646854fc1a12d2d9618f5db8f : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae8c29589b044be9bbe52e0037f3b512 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6fde9a51036473cb2e175c09dc9b119 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/160b2cfbd76c4e83af9ee0ed7359d583 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df9a00e361c449fdb0aeca17983b2c36 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9fb8ac0874394d85864805dab7e4daa1 : SUCCESS in 3m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d2d4e350963848b3b41eadf8e38716c1 : SUCCESS in 17m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/79d28f36b34a46e186919f457e652457 : SUCCESS in 8m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6ea88698892f449d81315d6fd4dbf7b3 : SUCCESS in 18m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/764bd59583f14af7ba12f852709535a1 : SUCCESS in 25m 59s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9a8d1983a232487f9fd8da467ea9d20d : SUCCESS in 4m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6e483e8f48c943d9bfdcfefac70e4bf9 : SUCCESS in 5m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/670523fb223a4b058197ec03940b2985 : SUCCESS in 3m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2be12d9f33ac4fe29aa9daab312e38a9 : SUCCESS in 10m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/55aed4b5681c4f038524e1c1d1e5e1c4 : SUCCESS in 6m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/10d7f13a803645c485e08f049cd8575b : SUCCESS in 5m 05s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/883e4f60f1b349b9a6314b600ffcb925 : SUCCESS in 8m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cca995fd10b440d0b158e7ddac544719 : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c1c66b7e6049408381186a16e6255e35 : SUCCESS in 6m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a1703a970cb34dfc819c3b2cac38bb24 : SUCCESS in 8m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/54bca57d01f246269837c10edde6caf7 : SUCCESS in 13m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a312945e6b5f42429663db76c11f5a7a : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2837300dc0584a3c9bf6a4a51349932f : SUCCESS in 12m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8b136583c5f94f848eb6e16dbb92fa99 : SUCCESS in 5m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6171225d0b8c4392bea954e17f4d92c5 : SUCCESS in 5m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d50e40752cc34c638dc7a832b3959394 : SUCCESS in 12m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4259c08da82a4b9292afc0b142e5904b : SUCCESS in 20m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/320a955d288d46e6bc73d9705572f750 : SUCCESS in 9m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5a04d92f70014daca1dd8d54d48978bf : SUCCESS in 13m 16s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/406d3fbad3a9438ab4bd45c42d7ec3e3 : SUCCESS in 3m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/32c9c983754742f8a121e75d3f7232a1 : SUCCESS in 2m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7b936249a5749cbb2f2bb42aa97da35 : SUCCESS in 26m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3c4b2cd9f30442d29fd7c5b2777f4a91 : SUCCESS in 4m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/989faa86d2e74d7b9a21a60b314d1e2e : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71cd1f8db2514c3f8e3db2e9f40d6536 : SUCCESS in 3m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4606fb0ad00e437ca42818a5458d8b46 : SUCCESS in 15m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98f0a9e239c54702bd0501d6f8e0a661 : SUCCESS in 1h 01m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c061d6ccf8d4052976490d1e886a72d : SUCCESS in 1h 44m 47s","accounts_in_message":[],"_revision_number":1},{"id":"36fc31d6ea303b27e7d2bab18c0d7c726dd17873","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-27 10:20:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9377b55c273a1f54ee61fa438f0e3e51317d3361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 12:16:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d601030f8bef46c3b52882bb8378e1c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6933583e8bc648c5b7779fbdb37de4b8 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a700faa21c04dc184288127a97947a5 : SUCCESS in 14m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c814b797f184f4eb9208ae76e1a6b1b : SUCCESS in 17m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72a7066323fd455f99947fce2d58d5df : SUCCESS in 13m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b6f2c365c5b44b09fe777a828588319 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00e8c2a710894650b37249f6ff353e14 : SUCCESS in 3m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/31f1b014c469454ca13a1cb35cf5d347 : SUCCESS in 13m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c98e8aabcfe6497184f8694b6c375dcb : SUCCESS in 7m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5a20819acd9e461591290b150d71ede0 : SUCCESS in 16m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d359f4d6830c462a803071f42563aad5 : TIMED_OUT in 40m 37s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/21f8195bce804d39badb98e9790a174e : SUCCESS in 4m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5e6daa0e412348629ab6acb5e013d15b : SUCCESS in 4m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7fd0845edc514883b1ef77323bf46c5e : SUCCESS in 4m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/03f1d2b4141e4f208eb5e3b3053ba8b8 : SUCCESS in 10m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/32d7f58803df46beb95ca3974f2f86d0 : SUCCESS in 4m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f2e384a040f24a11844c27414fba9176 : SUCCESS in 6m 22s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0b25acc361e04659a54b9a1b1460b6f7 : SUCCESS in 7m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/18a30aacb3c74bf1a3b508172eccb4e8 : SUCCESS in 14m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7828e9d375d54596b7e0d380a05b29dc : SUCCESS in 6m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/63416ce931014e26974159216e3ad6fc : SUCCESS in 8m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/21ec2bc8313343979122b1c1f63dc22a : SUCCESS in 15m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/484890bb0a774d37bf1970a2c39712c0 : SUCCESS in 2m 43s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/40cb5e18990a47e7a7a3caa0167ee483 : SUCCESS in 8m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5fb72403f946410abfddff71c17d2a2c : SUCCESS in 5m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4f6caeed0c8841d59c543835f50ceef7 : SUCCESS in 3m 41s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/26b29d5938874b70961c91876496f36a : SUCCESS in 10m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/869ff3c387fa4144bf5a0364b2903495 : SUCCESS in 18m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8f739ea2a8714514938991942679cb03 : SUCCESS in 7m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d04387ab5682421681f57d9fd52138b3 : SUCCESS in 13m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f7ceabb553b841369a1240dec103920e : SUCCESS in 15m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2e44592bcd2a450599f7e9b293c94e6a : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e30a16baa2454367a5b6465019a9ee45 : SUCCESS in 24m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/055e57ebac2745e7ba474c5b5aa3611b : SUCCESS in 7m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5b0a4ebac9ad4d3bba44c8599a2648c8 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ea82e504f704686a6b04a4aadec9535 : SUCCESS in 5m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c1cef8229ff04bc59353be352f375271 : SUCCESS in 16m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6063c3b0c189447eb06e9b706393828a : SUCCESS in 39m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b2e19b9399e461fa2d832646102a262 : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5cf65b18787824f41f9b07c0780ce715946d437b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-27 14:18:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af179faaef03378946ace5b73e87004b57d1fee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 00:57:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c74660b3798e4c25bd1931849802fba6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f26f402d534c98b1f94225f197fe0e : SUCCESS in 16m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab6714cfe2141e2ab4e76c17198bde2 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/804e72e18cfa4d07848e56456f0741ee : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cbf9a7bb2ec45f091782f23faf85770 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c89d3cec72334cfc9b55a1236ec93cdd : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1926d36d8d2b42bfbf1c71bad8ec9c5b : SUCCESS in 15m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d471180a4b1d418697d9f15c1c936367 : TIMED_OUT in 40m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f8d46545fe04701ae6cc1e313392778 : SUCCESS in 9m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/888465a5866b43ff939ec91699f7819a : SUCCESS in 17m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6db722fbad144f28948c5a94ada12b80 : SUCCESS in 21m 01s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3796cc6e64ae4de2aa1a93ba4e69b9c2 : SUCCESS in 4m 40s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/464989f6a4f14f468c2566bba83b964d : SUCCESS in 3m 36s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b8f5e1e168ae4f2a8c4de30c45ff516d : SUCCESS in 7m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/dae621d848fb4c0cbe3c7bcdb6b807f2 : SUCCESS in 11m 16s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2854ae0083ed45c6bbafd2baf037c47c : SUCCESS in 5m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a868835857f44994923ecd32119e127f : SUCCESS in 4m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/16258e6c4f284131899a9b29033b20d2 : SUCCESS in 8m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5e9ada52af3d420c93de4224559f36c0 : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9bda79d7455c4547ae8b912a2dd1eedd : TIMED_OUT in 40m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cd09726bca6642958c0784de3bd64661 : SUCCESS in 10m 01s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c56d8a14e22345efbb793bf2d5f3dd84 : SUCCESS in 13m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/525b83be681d4b48a0175813001222f7 : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3a62824b07ac4a24ac699259e7864a90 : SUCCESS in 10m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3b1fd451ffeb47d09c38c3dd591c6a88 : SUCCESS in 10m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ca65572e4f864a31a740ac4951d5b34e : SUCCESS in 19m 44s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3e6e2d8299674e929b388ad584fba32f : SUCCESS in 11m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bbe93125032c4358a5b49b1b8b75e713 : SUCCESS in 22m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b5d06c13609747708aff3b54ad3dee26 : SUCCESS in 24m 40s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3e47c15963e94240afeec63a623be0e0 : SUCCESS in 29m 50s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/836353d913a34cb49c18b0701f924f91 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c6497a6f9e174a16bb50e2786e101272 : SUCCESS in 5m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54b04eb2cd8744dbab2c89083559c5ae : SUCCESS in 32m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a6ddab45f267415097249ef7c611740c : SUCCESS in 8m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/57ab7478ae3240609f59352fa9541428 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52a06cda6524439eb393bcde5bed0a7c : SUCCESS in 5m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e321bbeb66ba435da228c93cb58c62d8 : SUCCESS in 15m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81f052caf7b244949fa5805732221176 : SUCCESS in 53m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82d9c4b45cc84040a3d80b34766cda55 : TIMED_OUT in 2h 35m 33s","accounts_in_message":[],"_revision_number":2},{"id":"6b9b11e0c35339864939dd6334dc536d34d3a91a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-28 05:47:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8867d0d804ce9f911b8ba3bceed74f3d492ebf5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 07:55:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b54fd214d3874d88890ba03b5457938c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26ca5a78d49409d9dc9dfe057a03945 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6936b28ff72420f915130716fa691ae : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46b60a60971c46b280716726e2c03cf0 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a866563bf34b4dc68c4667ffb9c72d95 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e421a9116efc44bcb470dc24243a5cdd : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b39115b720845ae82393191986bbc96 : SUCCESS in 3m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8b84c74b7a794792abc50d644e88bab4 : SUCCESS in 16m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/33e8f73c5ff4404d8d2a493db1417189 : SUCCESS in 9m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4ac1bd0c745b478ea34d59cbb2a6cb88 : SUCCESS in 8m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8316d45325614036be2828482adbe380 : SUCCESS in 19m 51s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/546879efe48b435489039b136348a7e9 : SUCCESS in 4m 49s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2e54bc0897d24bee9c50a08299d6c371 : SUCCESS in 6m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/277bd26f42954b5b8958c0bb5574151f : SUCCESS in 3m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/85e0341b672b48ea969273353af55b39 : SUCCESS in 10m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a5c3ee30d45e40d1ab2dd8e6740e4282 : SUCCESS in 4m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cb36a825bc124f4ea4bfb958ca86a857 : SUCCESS in 6m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/aa16713468ef4135a1c66b1a2c9b96be : SUCCESS in 7m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d14ffa6146e4459c89ca9aea3808f1e7 : SUCCESS in 15m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/874b5d83eb004cb081010dab9c428e09 : SUCCESS in 4m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/633d4ae4d98d450982d18f3dfd4f4af8 : SUCCESS in 7m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b7962b63028a4d3698859912e431113c : SUCCESS in 12m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/62f430f8c90f4786bd60156c1507077f : SUCCESS in 2m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d048958432d949a29d05b19c93ba20f6 : SUCCESS in 9m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4d00dc2c3b0f4e87a12b5619e80c2188 : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/001d84526a2042b38162fb56f9433b67 : SUCCESS in 4m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e7cb64ea98b34d10b7278bf19f71159d : SUCCESS in 11m 31s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3bf0e8244f7d41f1bec60e25e375bdec : SUCCESS in 19m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/153366651dcb4c1fbc19eeae61d4d786 : SUCCESS in 8m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/52a845f2ff42498f94218c925eb7d712 : SUCCESS in 13m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3c28bff7ccab4ac69925ad399a638e38 : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2408f79fa2e4a46b6e69386aa1d5781 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22b4a04bbcf1438799dc54be1f836924 : SUCCESS in 23m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6ae7137fe6c7462cb17b34b127552e62 : SUCCESS in 6m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b6dbb97355ff4bcfa44bbeba8bef4572 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24b7514c10e345f2a6ae5fdac4c8c042 : SUCCESS in 5m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/39e4575e97e94f18916f7f35ddbbe3d9 : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89bf3edc7b374fdb8c00e8d864830052 : SUCCESS in 50m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8a49815a90a4e23ad65a30e44294023 : SUCCESS in 1h 50m 48s","accounts_in_message":[],"_revision_number":2},{"id":"b78fb64072402293be50f42d76fd92f86e1bb5cb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-28 09:32:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"035126d1c039d3b1d04cc717d790c2c43cdb176f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 09:33:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"686bd53c47ca8df97e4667151529bd0ab21a5daa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 11:45:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d162c6a437be4dcdac47151de8d28e3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69274f370874972b4e0229d4fc022cc : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8ec8192be640e1b0da74a3b070697a : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb63a7f7d7c4495180961fd59b8d157d : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4166b45d550a4c6ca29b16377be28c3a : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c7da1efc79f48a1b80642b7d8e12e78 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53f209e258084eb8971dd05e782d27f7 : SUCCESS in 3m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a6457eece25043c98983472f959e5a72 : SUCCESS in 19m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4dcc009114e64d1aaa4813f302b8a7f8 : SUCCESS in 11m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/33a21014ef8547b997fc4cb19ae27e8c : SUCCESS in 21m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1274fbfa5f274975a3b1bad2c61d8e29 : SUCCESS in 22m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b98c755233144983bcb5be89b1b5de65 : SUCCESS in 11m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b5687c9340934b6cbd33d4d89c390a73 : SUCCESS in 7m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0b2f0e193f3448bfb5053733afbc67d7 : SUCCESS in 6m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/de29cca98a614d68a0c9e528ab9ab756 : SUCCESS in 7m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c8ded1711a644723895be9faf42e422f : SUCCESS in 21m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4466d0b6295244beb5e4f6bf20fff4a7 : SUCCESS in 6m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c6d61c31d5684e399f0d7926babf1e2c : SUCCESS in 7m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dde92446ca00463eba916d0006a3c694 : SUCCESS in 11m 28s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/05d26fbc01d846bc8c6b8eedc83d794c : SUCCESS in 3m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/90d815629ab64231aa2943380aed0983 : SUCCESS in 9m 40s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8e34a154960c4409b7974fa4cc29a33d : SUCCESS in 17m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7a896abfb1d94482b9c96592b6096f6e : SUCCESS in 11m 06s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7fed9f5cdc0f438fa839527829214846 : SUCCESS in 9m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/55a1d0f1943d4a0daacc71ab62c67e18 : SUCCESS in 3m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c4e9fc6a340432e9b2b41bed125f0bf : SUCCESS in 2m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ef2d99a3d4340768f9c8ff449e9aa1a : SUCCESS in 25m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/16f6ee3d01d24275993c1f1fc3dfb46a : SUCCESS in 8m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c89968cee7924da9989e49a67c3cf144 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af322e320c614a0b983ac17f518788bd : SUCCESS in 6m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cccdb0c89a3943b497ea97ce048d049e : SUCCESS in 19m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c30b23719b04fd2826ed99ec0a7646e : SUCCESS in 1h 01m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3480e1f6d8a49d1811f39282015d14d : SUCCESS in 1h 49m 51s","accounts_in_message":[],"_revision_number":2},{"id":"86646b0e5fdf4ab6020070b17932fa7db569bad3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 11:45:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a2a1e3b1090065fc9bfcb4ad8f14913bd132fed7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 11:46:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec763e6cdb8842d88bfe7d13602758c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2d82ef56fd4ac2be45e1662ecca9db : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f91cd35279c4943e924480108e53ff0d62f2f096","revisions":{"2b15d89fec400afab6bcb878580f890d3fabb16e":{"kind":"REWORK","_number":1,"created":"2025-02-26 18:36:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/942845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/942845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/942845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/942845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/942845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/942845/1"}}},"commit":{"parents":[{"commit":"bc5c68ccddc8e4ced7abb4361a7af5d2cb8c1f5c","subject":"Merge \"Bump pytest to version 8.3.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc5c68ccddc8e4ced7abb4361a7af5d2cb8c1f5c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-26 18:36:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-26 18:36:49.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 7.4.0","message":"update constraint for python-openstackclient to new release 7.4.0\n\nmeta: version: 7.4.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: I385de6c58ddebc8c0f13c982eb6574d26aec93aa\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I00bf38e6d6537ac12c98c5867092f2e1379b1e6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b15d89fec400afab6bcb878580f890d3fabb16e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b15d89fec400afab6bcb878580f890d3fabb16e"}]},"branch":"refs/heads/master"},"f91cd35279c4943e924480108e53ff0d62f2f096":{"kind":"REWORK","_number":2,"created":"2025-02-27 10:20:49.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/45/942845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/942845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/942845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/942845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/942845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/942845/2"}}},"commit":{"parents":[{"commit":"bc5c68ccddc8e4ced7abb4361a7af5d2cb8c1f5c","subject":"Merge \"Bump pytest to version 8.3.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc5c68ccddc8e4ced7abb4361a7af5d2cb8c1f5c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-26 18:36:49.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-02-27 10:19:27.000000000","tz":60},"subject":"update constraint for python-openstackclient to new release 7.4.0","message":"update constraint for python-openstackclient to new release 7.4.0\n\nmeta: version: 7.4.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: I385de6c58ddebc8c0f13c982eb6574d26aec93aa\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\n\nAlso exclude the previous release 7.3.1 in g-r\n\nChange-Id: I00bf38e6d6537ac12c98c5867092f2e1379b1e6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f91cd35279c4943e924480108e53ff0d62f2f096"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f91cd35279c4943e924480108e53ff0d62f2f096"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
