)]}'
{"id":"openstack%2Frequirements~942899","triplet_id":"openstack%2Frequirements~master~Ie6f8b395f1154475a283d42830a62bc4b2ac2481","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie6f8b395f1154475a283d42830a62bc4b2ac2481","subject":"update constraint for cliff to new release 4.9.0","status":"ABANDONED","created":"2025-02-27 14:14:08.000000000","updated":"2025-02-28 17:27:31.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0c316560fb67399e2affd68cef9c047584a80cf2","_number":942899,"virtual_id_number":942899,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-28 12:01:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-27 23:33:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-28 09:33:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"614781c9719d91f9b78489bd825729c75a31f765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-27 14:14:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33c0b2bc01e0b6d4131a4d059579d15f3fd983aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 23:33:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/326b026d5aa044da90c7ca555b7acbef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad1e74865ae41ea974c1b692c5d5edb : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f183766e984cbcb99a1da987e22cec : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/823a41abd38d4d06a1eb865dba462e97 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad631d1189e04d3ca012ed43e01a3dda : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68cd0e4ff5aa42a4b8aa29fae8ec2517 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a986348c4304d3f930bbd60a1cbaaa8 : SUCCESS in 3m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9d507960a196450d881134e8472cc7d8 : TIMED_OUT in 40m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2acabc5baddf448ab09d6139190ea7fc : SUCCESS in 12m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1c84173993e644d7ba91acc4fb2a95ae : SUCCESS in 14m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/611988f6e4d843d280124fa8ac67b887 : SUCCESS in 17m 35s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/37b7f8184c614036823dec8d3a084b10 : SUCCESS in 18m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/003413db60ac40d1818d2fcacbe8ff4d : SUCCESS in 5m 13s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c85eadfe7b2b401fbe6c6445e0de5a02 : SUCCESS in 5m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5d7bc517e0f7426aaa59b6d78be4ce7a : SUCCESS in 11m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/408619f7c47f48ee87e45e4af3001050 : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4d7e341397524527bc6fb6bd6a5cda77 : SUCCESS in 4m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/50f3c15a7e3d41ad9a5caf37f0445196 : SUCCESS in 9m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bebe45b64f114a10be805bc540436fb4 : SUCCESS in 6m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef4ad285c04442b584f7451cefd809ee : TIMED_OUT in 40m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a5460cc19d2d4f54b8acfed366c34123 : SUCCESS in 10m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/211689c7df554e7983fa4d879a75492c : SUCCESS in 12m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9f55ff8bef404d339f03001b6dd740a5 : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/67823544d2be45e1867f251d56a6d79c : SUCCESS in 10m 09s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7cca40f5660846ddb7cb29224d8d5921 : SUCCESS in 5m 01s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7d9a4de9335a4562bd3efeb8f5e5bebd : SUCCESS in 6m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/70fbfb40162944d58a8737194b749c1c : SUCCESS in 12m 26s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6d576526834046f9aea4e9a1a5add0ca : SUCCESS in 18m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5779ecc8b26744f6b7108536da9ff3cc : SUCCESS in 7m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/39685b923dc84e488f75fa7092c79284 : SUCCESS in 13m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/070fce01868d45f98d36cb9f6f411a9a : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e89bf78e2f9f4e349c75433163fcc974 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/adef758a167345f0aead1ee593444a6a : SUCCESS in 25m 55s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3aa40d9c109c4b5aa020757ae1a319e2 : SUCCESS in 4m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/57e1309803c2448e8256bfe9bf7313cb : SUCCESS in 3m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eaf3b00af78a45deb9c5f61071da1786 : SUCCESS in 23m 18s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/341ac90d71ec47ec8af4f30ce84906a9 : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eca43d502b404da180981a870a1381c7 : SUCCESS in 58m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a5e37224d8944abbe315a92d55a87d0 : SUCCESS in 1h 44m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a44a17c5c3a3b78eaba4c4011d2f1e85bf5ade34","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-28 09:33:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a4bba091fa28c31f7986fe08eb32edcb02d0d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 12:01:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53a14301b10e4c9784adf317eedaaed7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ff9fd678d34dd9879d2615f5289a65 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc930ed64414a91bbf1f84cc26d2c45 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71fa05af1da244c88a23150deeba6ce4 : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea76c3070b4345abba39f370d993d79e : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a5a012a211b4bc4a45b331ad8e03b33 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c625ff39753343dd87be1b14fe5304e1 : SUCCESS in 3m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/434512cc6928498dbd0c14db3eb6dc51 : SUCCESS in 16m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e83a8c24027540c1b071661ed4ab5403 : SUCCESS in 10m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ba69d3cc66ba4eb7a391b518c938b5ae : SUCCESS in 16m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7665a84b9d88498f847698ea08c8980f : SUCCESS in 19m 46s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/32c5ddbd181b403ba99837c756e8e147 : SUCCESS in 13m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/30c6fdb7be10405eb9dad54df1398c88 : SUCCESS in 4m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3cc90bbabf71425686e45633f5fa98e9 : SUCCESS in 6m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/429fdda445534b1485677d07f5ab9b74 : SUCCESS in 13m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e28006519bbf4db69d77544db64e61e3 : SUCCESS in 10m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/446e14018e134afc969f7bb51c13ca04 : SUCCESS in 7m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/31f04bb35b1a4e39afdab83361c6642c : SUCCESS in 9m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5505d617336f4616aaf2bb024fba3f80 : SUCCESS in 14m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/51567b1e08be414eb2f77ba840e88e86 : SUCCESS in 6m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/da2fc1367b884351a144f36226e93142 : SUCCESS in 8m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c253880a19ec460b9d5ff4caf623c3af : SUCCESS in 46m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d4cdb55993234bca841ba189f13833e0 : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d99f485181424c169fd069c7c1b2df0a : SUCCESS in 10m 17s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2fb1fe98702040cd8883e22694b88ab0 : SUCCESS in 7m 06s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/734ef570dd654035bce9f5d293d9bc9e : SUCCESS in 13m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ab33964e82e14ff9b914b4b22b42b345 : TIMED_OUT in 40m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/db097d0e7e6e4b3b80e3df829c28f521 : SUCCESS in 19m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a3bb18f871644ebbbfabc7a55626338e : SUCCESS in 7m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a516c3330cd049ed80406c4791f80959 : SUCCESS in 26m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/38c5b40e76044f3da0327f590f058e83 : SUCCESS in 8m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f70e3f7094fb48a5b496391a75f61f78 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e121ff0c77ed4f4b9b3601ab5b1028ea : SUCCESS in 38m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/171b196ebf8742abb6c640b0e3a5c1c3 : SUCCESS in 7m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/84551880fef74823ac9d38d490debdab : SUCCESS in 11m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2189371184014d88b1b052c268bbaa97 : SUCCESS in 5m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fb63ce08a74f4d65b82e16fb82a47816 : SUCCESS in 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffb6ed17136a4f8fbee0f7f38ef8bd78 : SUCCESS in 1h 03m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/228e89c1e5884c6ea306cd0e933cca48 : SUCCESS in 2h 21m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0c316560fb67399e2affd68cef9c047584a80cf2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-28 17:27:31.000000000","message":"Abandoned\n\nsuperceded by 4.9.1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a9283dff363322d649ade5c132208db7475584e","revisions":{"0a9283dff363322d649ade5c132208db7475584e":{"kind":"REWORK","_number":1,"created":"2025-02-27 14:14:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/942899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/942899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/942899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/942899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/942899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/942899/1"}}},"commit":{"parents":[{"commit":"ec3a73771a1cc048e6210b602ce96652059238ad","subject":"Merge \"update constraint for oslo.service to new release 4.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec3a73771a1cc048e6210b602ce96652059238ad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-27 14:14:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-27 14:14:07.000000000","tz":0},"subject":"update constraint for cliff to new release 4.9.0","message":"update constraint for cliff to new release 4.9.0\n\nmeta: version: 4.9.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Iaff32f3be72409fc1339337b06f6c6c9deb33074\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie6f8b395f1154475a283d42830a62bc4b2ac2481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a9283dff363322d649ade5c132208db7475584e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a9283dff363322d649ade5c132208db7475584e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
