)]}'
{"id":"openstack%2Frequirements~942963","triplet_id":"openstack%2Frequirements~master~I9c18f9cb38a279dbab17318cadbe1d3482778b15","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9c18f9cb38a279dbab17318cadbe1d3482778b15","subject":"update constraint for python-zunclient to new release 5.2.1","status":"MERGED","created":"2025-02-28 00:35:52.000000000","updated":"2025-03-01 06:37:15.000000000","submitted":"2025-03-01 06:37:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942963-new-release","meta_rev_id":"5dd7bd2d10302de4a705786fe8b78c2c7bf6eeaf","_number":942963,"virtual_id_number":942963,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-01 06:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-01 04:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-01 04:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-28 01:47:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-28 09:34:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-03-01 04:35:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"7b0279e660be93041ab9d8ce11b01e1bc8e56eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-28 00:35:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad2549710896a7ceabce984c8bd607d2878a7ee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 01:47:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/439073d4f8214bd5ae0c46f49482b376\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67709459bb5b4fdfa3311058dcbbe998 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f80ca9ef09249c2a93bce0b21629242 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e8207bbfb6a4aa5b5e37c9c8a70ec2d : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7bed1cbadd1444b0b9324c74b204dd4b : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a12912afdcb54492b549103005ba9653 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/35b71619f1b642b18b7fa63cc735a36e : SUCCESS in 2m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6892a1ba2da8429c813c78e1cba1804d : SUCCESS in 13m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cef94239e7354cb3b56ba61bf525aece : SUCCESS in 5m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3e9e42f32a10425b9c1360a49955fe64 : SUCCESS in 17m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cee96cca2fff43a9b1c88c6412dc4ff4 : SUCCESS in 10m 39s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/92b277a5ce704cf6b412099b4b6e7720 : SUCCESS in 6m 41s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f652474313c9421ab7b3b612dac7081d : SUCCESS in 3m 41s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f1c7625418a947b1b1434dacf4caf369 : SUCCESS in 4m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ec1d674f8ab142a087d4c3513c82ab6f : SUCCESS in 11m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/baa4b8c6ed974e83b9b74d065d5db5d0 : SUCCESS in 6m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/96b981f0404f4c60a48d97b819c2edb7 : SUCCESS in 7m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d3b3f254b415458192ddffbc8520c898 : SUCCESS in 10m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/def2e197affe41fdaf3bb610515e394f : SUCCESS in 18m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/993521a74b814cfc8d7bc5ba8999826d : SUCCESS in 6m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/65094d0f709f4b65b2539880f34682f9 : SUCCESS in 8m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/490a80e95e2b419ba2055b774920c502 : SUCCESS in 14m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/97fa3bc3176143fc9e0f6e2b07ae960c : SUCCESS in 2m 42s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a7aa2dce6dd14bb682137ca99df7e909 : SUCCESS in 10m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bb5c8d5c8aa64dffae0e621f8b711305 : SUCCESS in 5m 13s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6874bbf923544336b961f00ef3a8bf67 : SUCCESS in 4m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f6a71c9cc9e24d139676c8c95c570bd9 : SUCCESS in 12m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3f6fbd9d80794b3baac935471145e533 : POST_FAILURE in 21m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d33cf0861a6b48eda14a3b78ab6f1b7b : SUCCESS in 11m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8727f89adf414f98a534ebce92f07962 : SUCCESS in 23m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/87026bd7d5b24d998e81a129745c7b2e : SUCCESS in 6m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79fa34c1a0c9472f917e76a7841d2af2 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/745fab23407e43c6be0a97c255915cbf : SUCCESS in 25m 28s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1d55afa40ece4a88a810bf79c5d89ec9 : SUCCESS in 4m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8a596d3e6ca64d1ab5a0a796b1a79095 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ef25025a90c94c869d7e53366399bade : SUCCESS in 7m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d80825c39e714a98bdd3acdd271879ee : SUCCESS in 16m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59a829d38a124001973d3dc502aa6043 : SUCCESS in 51m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ada3795f6fa3442baf67354e0f5b92e8 : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e2e52f588c76e5283b2d319ee01575eebc260e74","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-28 09:34:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e0a37572efe28e73f779cd197f08c646926d66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 11:15:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bee5129ad93f44d28de82830fba6cdaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d024580e634ac8b0934f6e68efd0a6 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a88c1e70534d6f92f6fa98964da372 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6abe60d8a6c141c084414eae10ec4196 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1f4084ec5cc4b0393e95888b9d6f10b : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/069691af5a44409586cea8df7f3413b0 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af19f6daf4af4eae97242d60a9faa172 : SUCCESS in 4m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b3af5ef8bded48feb392e04d156d28e5 : SUCCESS in 16m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/183db2d3cf624b48b30712c6985067b1 : SUCCESS in 12m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/25a9e687c52f4e96b27ee6151421ec3c : SUCCESS in 17m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/aa7585a657df44c996cfb54f2c9fa080 : SUCCESS in 12m 07s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c3ffdbf6ae304341825ba32e8174c003 : SUCCESS in 3m 08s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f5d9b094337d418d8dd2576a728796a2 : SUCCESS in 6m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/edaf0d2258794735a9499b47d2d6cb5f : SUCCESS in 4m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7159b6927d6648f7810d106dacda037b : SUCCESS in 11m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/de4cbf59044d4b87985fc95dd5f7148a : SUCCESS in 4m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/53d1fc4e026f46c988602e5ce4594c40 : SUCCESS in 5m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/58d8bcde8519476298304ede3e70ab68 : SUCCESS in 7m 49s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e82d35cde26b47e293c69248d99e57cc : SUCCESS in 17m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/47a1876d98fa4aa6890a868cbf0cbcbf : SUCCESS in 8m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bb6771407ae14958a282d3c8f87077b8 : SUCCESS in 7m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9d819b8e1f304c4180598dcc046169a3 : SUCCESS in 22m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/89365960a8f1402db629dadcdaced539 : SUCCESS in 2m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7a205a3f51924cd5ab755547c754a01e : SUCCESS in 16m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/40cb740719a347949d6cd9367a7247fc : SUCCESS in 7m 11s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2c64817575cf495d9acc3df7baf78d14 : SUCCESS in 6m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2eef19b1325247d9b472b5e14e2f6227 : SUCCESS in 13m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/77a101b7f37c443786cb11e3f27f09c4 : SUCCESS in 21m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2766bc023b2348f0a2833ed8419a961e : SUCCESS in 9m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/62f430509d46448c92c4b0295d3c31d6 : SUCCESS in 16m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3c493614b3df407ebafa8df150f6e48e : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/690ca176f0d146468f35c2ce74e8feb5 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/207423c9103c47f496fb57acf663d595 : SUCCESS in 26m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/540bacbf66394f7e8d0a1495252d8212 : SUCCESS in 7m 52s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3a3393681ebd43089079f19ed79cbcd4 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4779f725dce3405db472f922f9b6448e : SUCCESS in 5m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5643806ce07b4a9683989a842c09fffc : SUCCESS in 16m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0e0408c295548179639d43fd1504994 : SUCCESS in 59m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4155e9c22ec4b8eaed5b60d0addfcea : SUCCESS in 1h 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0246ffd51560275e3706d23321bf2396aef87bec","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-01 04:35:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a86374da51aea98d263205248fcb246a21d7c7f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 04:35:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a95215cf27e68b89548609802bcfdfd3b6237807","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 06:37:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fb881d5089143d7af9b97dedd257a03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9429a2e0bb9b4bad9c770fe49d8028cc : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b94646cf03fb425e9f3448022eaf20d6 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4534070f747f4d3b85b504e67e09b5ac : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d31a14b4bc3a433482cbc98672939f39 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8266a31419104ca3a86f8137be115715 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b4b1623cc1f49dcabc595326a4bbbe3 : SUCCESS in 3m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2cce4dabf2f54d299981d9877e265f30 : SUCCESS in 14m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/936b025aa68f4215945a3046b520cc83 : SUCCESS in 8m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/15eb7a25f5d7467896b839bfd9de85f0 : SUCCESS in 17m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5964cc7256834867a5696e9c1a75ecd8 : SUCCESS in 21m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fbd74a59fc7b49648bc86d1f57fac465 : SUCCESS in 12m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/faeadf1ab4224dcdac6143bc1fb1c015 : SUCCESS in 6m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/03f5e8cbb7b140f6a2bdd08beab04e48 : SUCCESS in 7m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/57a4e975cdc444eeb340851597c0a286 : SUCCESS in 7m 12s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a5c13bc224f145d899014e1208062485 : SUCCESS in 14m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5ede778fea349cd81ec9c24730366b8 : SUCCESS in 6m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b95f99c06063400c8114fe42f2e869df : SUCCESS in 7m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/25fbb74b27324f2d9ca2017cf99b23c1 : SUCCESS in 13m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/332e1315dff04e31856133470d4cc817 : SUCCESS in 3m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f7df721d1caf4a7fa7edabbfc568ed6e : SUCCESS in 9m 36s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d2d6d9063b7144b78a2cee45f0263851 : SUCCESS in 18m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5da352a175a141baa6c12d48e69a7b7c : SUCCESS in 7m 29s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1f2588e0f8b04dcaa877bfb01b38531f : SUCCESS in 10m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/225159b477304e6b920e6bf4d408cb9e : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/578a398866464b24ba8e819f7541dd3c : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eea83ef420c34bfca3edf14e296938ab : SUCCESS in 25m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9ffc9b86bb1a40d1add1cb643ab524a7 : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/509ff4665a2e4f1390384e1ee9c9b63d : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86d3d01395b5435d9dda84eb8934e69e : SUCCESS in 6m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8623d9481d264d4db42671022ab62fc8 : SUCCESS in 22m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab6975c634114ab9b0af6c89fcffd23c : SUCCESS in 1h 05m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45f81264675b44e9a4dcc534ee9d7ab1 : SUCCESS in 1h 54m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5dd7bd2d10302de4a705786fe8b78c2c7bf6eeaf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 06:37:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb5ea9ebc11da0659da27973b809e8587c42cbf6","revisions":{"cb5ea9ebc11da0659da27973b809e8587c42cbf6":{"kind":"REWORK","_number":1,"created":"2025-02-28 00:35:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/942963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/942963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/942963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/942963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/942963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/942963/1"}}},"commit":{"parents":[{"commit":"ec3a73771a1cc048e6210b602ce96652059238ad","subject":"Merge \"update constraint for oslo.service to new release 4.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec3a73771a1cc048e6210b602ce96652059238ad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-28 00:35:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-28 00:35:52.000000000","tz":0},"subject":"update constraint for python-zunclient to new release 5.2.1","message":"update constraint for python-zunclient to new release 5.2.1\n\nmeta: version: 5.2.1\nmeta: team: zun\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I6eebacb69192008f85f33c67e61a69aae32a3b17\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: hongbin \u003chongbin034@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I9c18f9cb38a279dbab17318cadbe1d3482778b15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb5ea9ebc11da0659da27973b809e8587c42cbf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb5ea9ebc11da0659da27973b809e8587c42cbf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
