)]}'
{"id":"openstack%2Frequirements~942965","triplet_id":"openstack%2Frequirements~master~I7f4f96cd762b38144a50d0ea5364e082ac8ef882","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7f4f96cd762b38144a50d0ea5364e082ac8ef882","subject":"update constraint for python-cyborgclient to new release 2.5.0","status":"MERGED","created":"2025-02-28 01:05:01.000000000","updated":"2025-03-01 06:39:07.000000000","submitted":"2025-03-01 06:37:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942965-new-release","meta_rev_id":"18e04a8705df7a8484e1b13dc013052f1f059b65","_number":942965,"virtual_id_number":942965,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-01 06:37:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-01 04:37:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-01 04:37:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-28 02:49:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-28 09:41:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-03-01 04:37:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"341e5602e64d1f3e85f1633da2d0232ed16d65ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-28 01:05:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"543258d5f23b35fd5a9bb0e4a61b27ce76fe8610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 02:49:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05b8f620f7e2442da26a5ee24b0257ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bdd2b9667bf423cb45fdfe8600ee958 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdcfc5c705df4806bdfa1af928eb3366 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/369d9c5556924705bac0043de3ab16bf : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30b9b882b6a5415c9ad531510a38b7e1 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a065cc583fc47039f1c68222409f035 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c9a8a7d8ef544f5b86f4b489ea8a272 : SUCCESS in 4m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/46d1837f44d9486e8b4e4aa40f6e21e8 : SUCCESS in 17m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/55b8cfb9192549feb4d9b7943f8a2530 : SUCCESS in 10m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/41a140358c364857a3aaffd8df449122 : SUCCESS in 18m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d9441237143e436b9aca5f0efca5a454 : SUCCESS in 28m 32s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/256f2c1573a544bb981e5096f41d0c6a : SUCCESS in 7m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/eb72b55c64e542b3947f1bd0d04af35b : SUCCESS in 6m 38s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a5678144ea0942f5b76e05b0c435571a : SUCCESS in 5m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f399df1b1c5148acb79c112d91c9b434 : SUCCESS in 8m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1976119493c849ad89965be899e0f585 : SUCCESS in 5m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8fb2b83de2e649e0ad2fe930929f5e34 : SUCCESS in 3m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d4cccd9d98584a13a1912e0657a31e38 : SUCCESS in 9m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e6ba33d834cf4f73a4839a8750d73cbd : SUCCESS in 6m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f2841e4c2624e2982a0ae6490454d5b : SUCCESS in 6m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/02fd55269ede466096d1a1395bded4f6 : SUCCESS in 5m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f4ba0faa05c742a09c4c31a722f4e76f : SUCCESS in 8m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0caea7e4bc5346a3a3bd89b6a21b8378 : SUCCESS in 2m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2acf2397494d4a278a08b8d5faaa1f30 : SUCCESS in 10m 55s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d2acce7dff124d35b471a51556c45867 : SUCCESS in 6m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2a986d0845c34b2c8ae81faad7a7ae8d : SUCCESS in 5m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4cb82c6afc064281a6b0254165f185ef : SUCCESS in 11m 20s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/79f2f82ba4b947e59e30e2f1cd43d872 : SUCCESS in 17m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0b1c66359efd4491b16f5063c3e59d6c : SUCCESS in 8m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/727c072cc5ec47ccb8e621f163b403f0 : SUCCESS in 12m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/eb1aacb8baed4d62948f5a86a3c3cc90 : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dcb93185437048a99289da5ff7a0deff : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/287a66e3357a4c60bb48a7bfb63d7bbf : SUCCESS in 25m 28s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/162dec086d664cefae7cee28c2dd48f6 : SUCCESS in 7m 30s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b3e13b4482454d1d8e8dd8e613369cda : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5fa9d5bfd7242a3a9cdce88c56865b4 : SUCCESS in 5m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f9868488bfeb47b8bb843904279c2766 : SUCCESS in 15m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f4d46c9d7454689b502957f21a611a8 : SUCCESS in 1h 12m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f32fa04772a412fbf2f5f2d73ebf5ce : FAILURE in 1h 41m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7a74cef23da9e12e5a251896d4520b38f01ef07d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-28 09:41:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c42f2ab16d6a58d90110f5d21e9bcde6cf85aef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 11:49:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54019be6ee0c4d57a3ce0ce58a78af3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc96f8d95a6471b9e70401402448da3 : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0adb29e2504f0f99f7aa807c9832b0 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dea6c91a95d45c9abb98773ae8de3f9 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aff9f61402af4cca8066f5bdde57ae42 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a235c8cec8be428d8e9ccda03439fe79 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/799c19430b7e40c396639179daeece1a : SUCCESS in 13m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ff1f83837c694af4b4b2600127aa239c : SUCCESS in 18m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ee9832d9c5024bd68fa53d19dea4418c : SUCCESS in 10m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c0359a791387452791097ff10028563b : SUCCESS in 19m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b04107b6ca0649a0a4274d3ca6c485e9 : SUCCESS in 26m 59s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b087e36e6cf04959a5416bf52fe41cef : SUCCESS in 6m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fa087758db7f468fa6b8b1fef777f9e3 : SUCCESS in 12m 25s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/bbebd55579994dc2838134b4b243a2ab : SUCCESS in 4m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5f8c5d6288ec417c9a13187845f99e58 : SUCCESS in 12m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/654b4d544ce347438729927317286de8 : SUCCESS in 5m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3a8f017dabf14dde8dab5e0ecdd6f9d7 : SUCCESS in 12m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6d67aac71f404871afb8881681956db7 : SUCCESS in 5m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a1a4a75d7d394123bf01e0d64f62da94 : SUCCESS in 7m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b92c50be23c4783bc5f846215046796 : SUCCESS in 7m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9fa5e84023f249fa9cfa0a213fa4f507 : SUCCESS in 7m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/96c048d25718477fbe675aae0b50d3e5 : SUCCESS in 10m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4e755ccae1f540a794a48514cef64c6a : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7033818163284107a1a02def6e880c0f : SUCCESS in 11m 03s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/085adc8edba44fc3928a03a0d6a7bd02 : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dbabcb6a2ec140e995a72fe0da7953e8 : SUCCESS in 7m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/24c490ea53ab425eb76b1a9ac3ac86bf : SUCCESS in 11m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5d47fc528e4641c89168791fc07a4985 : SUCCESS in 19m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/98e50ae6efe9413680e4fc0990045eef : SUCCESS in 10m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c9f0a3673e684d9682ffc1a078814673 : SUCCESS in 13m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4af413a58fc34e34acf0716d930f1c4d : SUCCESS in 7m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c26bb71f508246bcb306f6b85a3e25a3 : SUCCESS in 6m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7749e9e0d6024957889d5e12be95287d : SUCCESS in 26m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d3edd771201641659e2c7f69e1e5534e : SUCCESS in 6m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e1ba2492034c45b796d382832ddc92af : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d3698fb39764c019cf8fcffc6fadeed : SUCCESS in 7m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/37aafc219abd4fcd95600d0f0f352ec0 : SUCCESS in 16m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c776d191d9b45bfa399acc59eea1d32 : SUCCESS in 39m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79fb4b16fc3f4a73b97ec838b7dfcf14 : SUCCESS in 1h 56m 41s","accounts_in_message":[],"_revision_number":1},{"id":"8ceea470e34e7186114dde716d533109a0bdf747","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-01 04:37:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71d007c82bd3c71f3b313f76903dd3bc04416c6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 04:38:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0d312b999e56ca1e273dabbe951e8a2642c5bfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 06:37:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43b0768a50d04c648bad42195fd3048d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9968856db94454aa405c314a8d5593 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32d3da9470c6493ca9e9bf8e68b1fe55 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b36d4a486c9046d7aa7f922736cee2c5 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e049fdf865a04839b241e4bb11c819f9 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52c8938344dd4fc59e521b8f209294f7 : SUCCESS in 2m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5fa046bdd5414e8aa7a254771c44cc20 : SUCCESS in 2m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/47527fdb2600419d91a237e41c2e28e4 : SUCCESS in 15m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8c5bbe44ade64c61bd14f9672df63d90 : SUCCESS in 8m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f43e7abb4ef840c0b8b293e87dcfcaeb : SUCCESS in 16m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/09efdf63085947aa93a8a92b7bdb53d9 : SUCCESS in 26m 10s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f535dd82fe7045eab052b184799280c4 : SUCCESS in 9m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/83a044c5d8e94a039290158bc44002d1 : SUCCESS in 4m 53s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0b915805a5024c299ae33eb158b04ccc : SUCCESS in 6m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6b572e58e7814886955442536c2ea374 : SUCCESS in 7m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/147f31a1f83743608c85921707255f1a : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e734c09bcd04abca4949a30046cbd7b : SUCCESS in 6m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3338134d846e4d9eaeeef8686d98db77 : SUCCESS in 6m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e9204f25f8ac4137b0ef134497fbaf01 : SUCCESS in 13m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ef0e28463d6d49c1a318bf5b10336ef2 : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5504b57f8b784b75b4332cb20e2d50fe : SUCCESS in 10m 03s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b5ac32481c1746c6af31d0aa025c6e05 : SUCCESS in 16m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7a8cee8ca9b245e3a3dd12575f4b48ba : SUCCESS in 7m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/86337f5b331140158ea71e52d4e91963 : SUCCESS in 12m 41s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7c3be2007e40422f8d9e2f4cda44eaa5 : SUCCESS in 6m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f750dee8c2824ee282fcdedcfc02af76 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/884f3ba0836f459382af03496c41ebf8 : SUCCESS in 30m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3bbcd8ede57a48aea665959bee24e1a7 : SUCCESS in 6m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b5cf7c3358154e50801ee19302faf4c1 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a50d4e043e442e89e93224adc3dc523 : SUCCESS in 5m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/32df5300cfa54ed3af606e4ee25b14e0 : SUCCESS in 18m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64740a3d80f24bf28b69d5748ae4a4cf : SUCCESS in 47m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eff50d7b0ab549efbfa0e7f9edf2ef2e : SUCCESS in 1h 43m 42s","accounts_in_message":[],"_revision_number":1},{"id":"991686b7610a638244c5f1ff47c0543a88a7a4bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 06:37:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"18e04a8705df7a8484e1b13dc013052f1f059b65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 06:39:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62bd2986887a40aea05281e561eca841\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f8a8c2b7226471b93ddbe0931d97255 : SUCCESS in 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29fc1db059c9d8b62daa3f750d2a1f71b9a2f321","revisions":{"29fc1db059c9d8b62daa3f750d2a1f71b9a2f321":{"kind":"REWORK","_number":1,"created":"2025-02-28 01:05:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/942965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/942965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/942965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/942965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/942965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/942965/1"}}},"commit":{"parents":[{"commit":"ec3a73771a1cc048e6210b602ce96652059238ad","subject":"Merge \"update constraint for oslo.service to new release 4.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec3a73771a1cc048e6210b602ce96652059238ad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-28 01:05:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-28 01:05:00.000000000","tz":0},"subject":"update constraint for python-cyborgclient to new release 2.5.0","message":"update constraint for python-cyborgclient to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: team: cyborg\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib78c4b8019333e6b4ea35396c40385804002730e\nmeta: release:Code-Review+1: Wenping Song \u003csongwenping@inspur.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7f4f96cd762b38144a50d0ea5364e082ac8ef882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29fc1db059c9d8b62daa3f750d2a1f71b9a2f321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29fc1db059c9d8b62daa3f750d2a1f71b9a2f321"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
