)]}'
{"id":"openstack%2Frequirements~943033","triplet_id":"openstack%2Frequirements~master~I3e472b09282b62045e4fe5cdfb6e680ab862f90f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3e472b09282b62045e4fe5cdfb6e680ab862f90f","subject":"update constraint for cliff to new release 4.9.1","status":"MERGED","created":"2025-02-28 16:48:31.000000000","updated":"2025-03-01 00:06:23.000000000","submitted":"2025-03-01 00:05:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943033-new-release","meta_rev_id":"d2c7f67d6964845e5ae12db0ad8335dde256e388","_number":943033,"virtual_id_number":943033,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-01 00:05:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-28 17:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-02-28 17:28:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-28 17:28:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-02-28 19:06:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22548582ac9a1318fa057962ab6f99e4e3aa7363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-28 16:48:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"822ef20248e1628616bfbe9db86af8c16b800910","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-28 17:28:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a76237bd51dea92cdf3c9c959abca2cd363e826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 19:06:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fa992997a8a461785753d8c464b5601\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfa2ceca9c0467c86beb832cd0b286b : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4544b423ea8e404f9ab20c2f16933197 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55486aee0fec4bb89e1b735f4ee139a5 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1df21cd8d0cd4f15a9a97c3842ba8364 : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e786212e380e4020b91ca0f66c1952a0 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ee83accfba740849477e27ac4bf633d : SUCCESS in 4m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dc9be19804794b3d88cf2b65e6adac4a : SUCCESS in 21m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1f421a94eecf4d5a97a593f599a38bc0 : SUCCESS in 16m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4e5ce83546f8401db87bb859c57e8277 : SUCCESS in 19m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/949a38c83f754c6f81d440bd112a6737 : SUCCESS in 31m 05s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e9c110cfb96f4a1387faa42db7969d6c : SUCCESS in 3m 26s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1a46e6494e4c4233903d8b4b47f2e179 : SUCCESS in 6m 27s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/44095ba12e8c466d89193636d154c84d : SUCCESS in 5m 15s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/73563740c4f949ef9ed4ca72723b7135 : FAILURE in 12m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1519bb05242f41748a5fe268e1d0210b : SUCCESS in 6m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/734aaed146c042ce92027f4099fe4c46 : SUCCESS in 6m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9d19d8ba1faa4f2c96963f1e7d90a07b : SUCCESS in 7m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/49dcb3358b974a97ac4b844996ee842d : SUCCESS in 12m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d839b864081144b2bde559ea8ca4254f : SUCCESS in 8m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/671a5aa3190049c6a332062c0a799746 : SUCCESS in 8m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/baf5b9834d9e4d8a8be71fa08d082348 : SUCCESS in 11m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c824a4bdb7b84859bba8ff4d13be64aa : SUCCESS in 5m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/dabef7ea34794846bd812cfced772d25 : SUCCESS in 12m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9da3d23a6b3a4e51aa5f813acaa59f94 : SUCCESS in 7m 49s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/36421872c743422f83d3073fbf89fde0 : SUCCESS in 6m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b33dfff42bd741e68340e60959ecb882 : SUCCESS in 11m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a8346159f26e40f9af89b8124278021f : SUCCESS in 20m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0081747bcb5e4e249c4bcc515fef1083 : SUCCESS in 7m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7a569b9341be48a6b80537bf8eb76096 : SUCCESS in 13m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d75e600b507b41c0a62a8cc9394c7bd9 : SUCCESS in 8m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b3620f9d565e4a3e813b1ad924bba3d6 : SUCCESS in 5m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f8d976b2bd14d05993cb935622e2683 : SUCCESS in 26m 38s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0deba6bc7405458ca731c3bb3bdd6c1a : SUCCESS in 4m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4488b051f0af450e935bcd7afe814765 : SUCCESS in 8m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/badc928666d24a8e80877ee2b80ac61c : SUCCESS in 6m 52s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cbb2d41f2d2e45dcaa63383f18a6e56a : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1357243362b40809d396ac0db22b0e1 : SUCCESS in 40m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9db202286a9747d48148d9ad8f416104 : SUCCESS in 1h 53m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d72fec136945386eb262e00194506b7df4946994","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-28 20:12:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5efae51bf3ec097fd916fdf3cfbbf7509d693471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 21:52:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff90379fbd8a4d75b46083da024a405f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33205b0d2e65442cb8a9e0ccd5a6d94e : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce6dd194c1842469d2e3b573981a229 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21f3a9c53c904d17bcd0ec49c71d7eeb : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/489b18aaf5b64d7d8ca4de729afcbfaf : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d77e55358d634075a50ee3355e48c67b : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2bf5fc2b1b3443a4b0795096b15c8f7b : SUCCESS in 3m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a452ffc25f9747edaa66336766d688c8 : SUCCESS in 16m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f5985aaf7d8f482cb425fe35a12725f2 : SUCCESS in 8m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/427f670551b24fabb812c0d0c8ec21b4 : SUCCESS in 13m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c3e4da5d5f44450dae0d46dcaf343612 : SUCCESS in 30m 21s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/14c0338a4c3c4d11aa13aca41cbe76a7 : SUCCESS in 5m 10s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7b2f8aa5ea7e4fcd93c53e7f10b1a6ec : SUCCESS in 5m 37s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/31c989dc6704473aa2a860df33fe32d9 : SUCCESS in 5m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c850d6ed169f420eb355c38554453901 : SUCCESS in 11m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/37123ed30d314fd1894ec631fd307e38 : SUCCESS in 4m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4e3ea725897b4eef80d5481182211459 : SUCCESS in 5m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0eff5c54591d42e38defc4a43f5d9e3b : SUCCESS in 7m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/960d661bf21847c88c7e966222b6112a : SUCCESS in 14m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/07bda45ba3964cb08a4bc3695057a17d : SUCCESS in 6m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5d7ae88a83fe4be88989357c6ef6185d : SUCCESS in 8m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e155a5bd10484fe3ac4d2b037cc7bc07 : SUCCESS in 12m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0ceef53c42814382925bb3325c93985b : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/304744af56f5408aaa0ac3d0d480adc9 : SUCCESS in 11m 47s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7e200439c97d44eca1b27e58b0fccd0a : SUCCESS in 4m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9d6a7238209e45a09fd90ce82618cfe2 : SUCCESS in 4m 48s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3a759bea9a8d4fb7a47963d504088634 : SUCCESS in 10m 26s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9882b7ebe696480face594f61b7c0360 : SUCCESS in 17m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5f242ae23c3b4125a07f8a1712b30f1b : SUCCESS in 7m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0b5ce563cffd4180aa0464207d816e0a : SUCCESS in 14m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6e1fc9c4ac1c4b909608b8d66a16300b : SUCCESS in 6m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e5e651a9b1814547b2b26b3b8c11d7e3 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/387ec2ccd54141d2b46b6d2efb53901a : SUCCESS in 32m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/33ff9eb05e5c4a00a4502e0ea36c2284 : SUCCESS in 9m 37s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/db94fbade80e433889022b42cc931a8b : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad0dd5813bf54d699887fee89fde3f33 : SUCCESS in 6m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/913ab99ca2154512a4a2f73d59566d2a : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1169b3b51c4b4610ad9e4a918bf264b5 : SUCCESS in 1h 15m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06a4712154c541c89b0aba98685134be : SUCCESS in 1h 36m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9630b7e73e533688ff9fa5a592609b6d3e6b10e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 21:52:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61f9fe5a0a2ae769c8677aa02dad14091c7fb403","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 00:05:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d88fd1ee47364e92b6c143b5263c43ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ad5ba6484f45c6b8f1dcae4a6c0e2c : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/490fde4527e9496fb5f8bf24e3eadb57 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7aa718d176e7473d9bd46db0e043a886 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc5ab23e0b534add84437dd829957173 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c0fcaec4ce6414296a4fe48924cd83f : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e39d8325d034e4baa39b16d0d39578b : SUCCESS in 3m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d8480119874041d9b007d75fae0be76f : SUCCESS in 17m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4dd6d2fbcab341e599a426c99180bb95 : SUCCESS in 8m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cd8f32ae419d4a20900a912227c7cd13 : SUCCESS in 15m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ee6b6255e9b14060a1c0db9413301598 : SUCCESS in 28m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d113c5c7bd4b4285a30cac6e9373436a : SUCCESS in 11m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ef7459606495477eb02bc6e2204a7140 : SUCCESS in 6m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2348aa9e902f4f959e5c3871113263de : SUCCESS in 5m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1235ef77c3c44ab2a7831bf03456a977 : SUCCESS in 7m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/73b490e2bb6e4bbda2885df0d1a4d609 : SUCCESS in 15m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91e838dde70a46949b476ad71137eeca : SUCCESS in 6m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/db4a4a468ae94b649525b8b583d1d95c : SUCCESS in 7m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2d0a77a67b6e49e2a22d147a23d52164 : SUCCESS in 12m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/363e4e2837c8476ea0c6f7a719d08eb8 : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/814d26ea6df74d07ae6d0a2e35e28948 : SUCCESS in 11m 34s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1ccc7ae9eb38462bbf286725810fb029 : SUCCESS in 18m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3d2d1db1270c46aabdb7bf998a15ff35 : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e78f81d5e1884a96baa5f0566c7a46ea : SUCCESS in 14m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bb0859efc5074d3bb769426c8ba43222 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/81d2168ee11f40c1b30cd8690eea6d3b : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5981e82e3b204b2696188b5f2ce00752 : SUCCESS in 23m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d9c6c196497c4927b4f5d073e09b6e97 : SUCCESS in 6m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ae2139d5eecb4134bf5ff148d8b70054 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb4883bd372e4acea688d8b6eebd8e9e : SUCCESS in 6m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cc6ff3be7e4a4915ad176e7b56cf3d50 : SUCCESS in 17m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28a146799dfb4176b4dad5744710a6f4 : SUCCESS in 1h 03m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1405908dec0c4f76b5129714d42dddd9 : SUCCESS in 1h 53m 30s","accounts_in_message":[],"_revision_number":1},{"id":"dc1ecbb3adfab373fec3365ab40243a1cef36dfc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 00:05:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d2c7f67d6964845e5ae12db0ad8335dde256e388","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 00:06:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27034fd1018548b3a9f459724d6e0bf1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52cd25eabc3b42ea9918adcaa10a6320 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6773c45a2a03c56df6065d3fb16e51751709784b","revisions":{"6773c45a2a03c56df6065d3fb16e51751709784b":{"kind":"REWORK","_number":1,"created":"2025-02-28 16:48:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/943033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/943033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/943033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/943033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/943033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/943033/1"}}},"commit":{"parents":[{"commit":"ec0b2548dbcf0dc894309ef8f9c8d578cf70c008","subject":"Merge \"update constraint for python-octaviaclient to new release 3.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec0b2548dbcf0dc894309ef8f9c8d578cf70c008"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-28 16:48:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-28 16:48:29.000000000","tz":0},"subject":"update constraint for cliff to new release 4.9.1","message":"update constraint for cliff to new release 4.9.1\n\nmeta: version: 4.9.1\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ic59a401c2c936a672691f6b07f083ac9dcb164d5\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3e472b09282b62045e4fe5cdfb6e680ab862f90f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6773c45a2a03c56df6065d3fb16e51751709784b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6773c45a2a03c56df6065d3fb16e51751709784b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
