)]}'
{"id":"openstack%2Frequirements~943048","triplet_id":"openstack%2Frequirements~master~If86dd98a6790f30b292e02f7a17a95c830d5c585","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If86dd98a6790f30b292e02f7a17a95c830d5c585","subject":"update constraint for python-observabilityclient to new release 0.4.0","status":"MERGED","created":"2025-02-28 19:02:54.000000000","updated":"2025-03-01 06:36:52.000000000","submitted":"2025-03-01 06:36:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943048-new-release","meta_rev_id":"df4631140f29e81bb7508ed478ae187627e47169","_number":943048,"virtual_id_number":943048,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-01 06:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-01 04:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-01 04:33:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-28 21:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-01 04:33:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3c06f067da59fd0ca871bc9c4e42ecc3d9c86bf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-02-28 19:02:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8defa73dc184f7353864a689a4f7aae721973fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 21:02:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/020a50e37ffc4ce8b6de4c898dae1c39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a464bddc4a44848fcb394924847fa9 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20b8290a690443c5ac5eb74f3af17a03 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4316472efd8d44c98454be90286f982d : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b41979b3f15e42a3bc4c58649062c4b0 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8404c58f9c8344248856fc54f890c86c : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/54195fbc34034ce1845c493504eb5520 : SUCCESS in 4m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d4e55b6a311b4ca09759eaf6688d8222 : SUCCESS in 16m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7d66d5e20ef242f481dbc71e40fb2da4 : SUCCESS in 8m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7af9fcca06144c44a7315f802324164b : SUCCESS in 18m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c633af7ce1c447829994555baefdd776 : SUCCESS in 18m 55s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/955adc76b32b4fc3a00c149f6deab48a : SUCCESS in 3m 40s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/46b9bb5e2e8f4735a205d2ef7882a78f : SUCCESS in 7m 46s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6d974d4117254d7d84b5b209207d519d : SUCCESS in 6m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9d77a72bdd2743b99a966eeb98ac0e74 : SUCCESS in 13m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fe3dd16aec9747009c645fa936be75f9 : SUCCESS in 5m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ed71b58364574173a57048b93719d33a : SUCCESS in 5m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/98aaaf8a40354b4a97942518accad9f5 : SUCCESS in 10m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8173b91d783e41abb9bd5edcc9e51229 : SUCCESS in 13m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0381a95949c74272931f10c5a65c9d43 : SUCCESS in 8m 28s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/de9f72cdc5af427ab9c8618ba99d9cab : SUCCESS in 7m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3a37d08a5ebf496ea3cf0bf5cc1d2e2b : SUCCESS in 12m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/408e33b8b0a148c4b05ff4e26b09f314 : SUCCESS in 3m 47s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9fdbc41271a04ce794ccc9f3dd6e8899 : SUCCESS in 11m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/11bb556d6ebf45499045834edef47768 : SUCCESS in 4m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/032edbd2f29c4783aca4d5cd4ce35e44 : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f5c8632f28c44579b97cb351ce8b70ae : SUCCESS in 11m 07s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3396fd0ff41a4c71b8cfaf931f851996 : SUCCESS in 21m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f1e0fa425d0f4b4b88f2f80e4b38c5a9 : SUCCESS in 10m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9e50d5c8b01941648d7634a670690287 : SUCCESS in 12m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1f639101d9084b2997b23817fcfddbea : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e93fc6105e5a47fb886aee2efc7aa4b8 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7df559cd3127413fb5c856dc3310d080 : SUCCESS in 23m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bc68cb9a7b8e4b5a8bb6b4d710be0c86 : SUCCESS in 7m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8a5b391937bc4b3bbb37ab586103b203 : SUCCESS in 7m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/244e0f0c3aed4f18ab2fd03f66378e7f : SUCCESS in 6m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3a35359b0e884573b822ce71ae134909 : SUCCESS in 16m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7a6b36ee13c49e1b1465f5b53212450 : SUCCESS in 53m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/700ec5e495284248ba4740a238c4999a : SUCCESS in 1h 56m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e7a210c876b339da8784b41f2d1f6ca920219710","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-01 04:33:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fcbcfa20c549c3984cbdaaf812061cdb0eb75e93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 04:34:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"218e695265264e50477cef5b200446d42714a19e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 06:36:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9690928173f74921938a00bcea3bbe17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d693289ea2e644c490c87e1406269029 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07e80e2f971f47e8b70ffbfcbdc26bb2 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44299da7cb5041eb9bb4e53a49479e1c : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1dc2f5ee2d324abcafa54ff44dfe3f53 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a956ab84ef4c419e9d89d3519c7dc432 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8468137d7de47f2be96d2b8afbae585 : SUCCESS in 2m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7460ebe8499f4133a85716c9893b311e : SUCCESS in 15m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/39fb9beac623458da82e689e6d2ef5ce : SUCCESS in 8m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dcc68e7431bd441eb70c62e5e2f2868c : SUCCESS in 20m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f8ab88b2f45d4b90ab8e8ed429b1a32e : SUCCESS in 25m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c16c507eb66f4d4db4fb743d1cbfd649 : SUCCESS in 10m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/dedabe7dca0e4513b068f74097279d70 : SUCCESS in 4m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ffe43cf6a3354b04baa2483cbeb0b2a4 : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/90eca38bcda0439080de7d7005b18093 : SUCCESS in 8m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1f24b3faad1344cb9915eb5ea05c58f6 : SUCCESS in 12m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc054f74272044c2a454d69b00dcbef6 : SUCCESS in 6m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/068ea23a0a4c4685976f01656ce87459 : SUCCESS in 8m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d80f3923811845e782988ce83255694b : SUCCESS in 11m 59s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a4a6f698fb0042be9fba02658f15b996 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/31c52f40d73b4fc1955e3577f3780fa3 : SUCCESS in 11m 02s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7fa5ff2db8e54c5eabd47c365fbdf431 : SUCCESS in 19m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2db1aa0772c44f928340d96eb7f519e4 : SUCCESS in 8m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7c296f44c4724a06a848081c69bfdb16 : SUCCESS in 14m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9a0a840e7c94486381346e64edabf1e3 : SUCCESS in 6m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc1cf312b475499fa30a7a260df5b151 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1916152e6cb448fbaa603b13a93845ad : SUCCESS in 28m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/84a195b55a884a1482e0f71177d836b1 : SUCCESS in 4m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4ebb91992aa841c49d8770865565a65b : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/053c885c5b994190aa2627650a39f7e1 : SUCCESS in 7m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/36077931d89048a7800c823858d10b53 : SUCCESS in 17m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2931519e87d74dc19f3cf4f0c829a3eb : SUCCESS in 1h 05m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64f460b15a1843de8de9faecaa59c76e : SUCCESS in 1h 28m 36s","accounts_in_message":[],"_revision_number":1},{"id":"df4631140f29e81bb7508ed478ae187627e47169","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 06:36:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b8a7672166612c420a513b698c1cb434cfdc43d","revisions":{"9b8a7672166612c420a513b698c1cb434cfdc43d":{"kind":"REWORK","_number":1,"created":"2025-02-28 19:02:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/943048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/943048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/943048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/943048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/943048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/943048/1"}}},"commit":{"parents":[{"commit":"ec0b2548dbcf0dc894309ef8f9c8d578cf70c008","subject":"Merge \"update constraint for python-octaviaclient to new release 3.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec0b2548dbcf0dc894309ef8f9c8d578cf70c008"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-28 19:02:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-02-28 19:02:52.000000000","tz":0},"subject":"update constraint for python-observabilityclient to new release 0.4.0","message":"update constraint for python-observabilityclient to new release 0.4.0\n\nmeta: version: 0.4.0\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I9f6b35d856cdd131ab7791943dcbf2588514e635\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: If86dd98a6790f30b292e02f7a17a95c830d5c585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b8a7672166612c420a513b698c1cb434cfdc43d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b8a7672166612c420a513b698c1cb434cfdc43d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
