)]}'
{"id":"openstack%2Frequirements~943089","triplet_id":"openstack%2Frequirements~master~I093b2b7013bc7ed75f4a01701888a6975ec59863","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-04-30 09:37:19.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-05-06 07:19:35.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-05-06 07:19:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I093b2b7013bc7ed75f4a01701888a6975ec59863","subject":"Updated from generate-constraints","status":"MERGED","created":"2025-03-01 08:21:27.000000000","updated":"2025-05-06 07:20:28.000000000","submitted":"2025-05-06 07:19:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943089-openstack/requirements/constraints","meta_rev_id":"528336a7a7b8d1456b647b9a5c509f1459ad4562","_number":943089,"virtual_id_number":943089,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-06 07:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-06 05:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-06 05:34:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-01 10:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-02 12:47:47.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-04-03 22:01:41.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2025-04-04 08:29:25.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"ac22bf7bf8805ab189a9e978329942da8bc20da6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-03-01 08:21:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"753ea2a0807f911020d6ee8d15e6dd0df8266b33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 10:03:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0033b27de29b44b9998506651cf590cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb36f04b5f7440e9abdef73551c8db7 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f28d3080a8743e9b47d25cf8a85b862 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49e1b5cc04c24e8582cf2a25eedb2778 : SUCCESS in 6m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7af796975bc944188e686f80e1da7060 : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a17dc655427f4294bec6656021e3b8ae : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f7561040b944bebad74a0f4d6b68243 : SUCCESS in 6m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/58275fc3445546c89c5e687b4091a903 : FAILURE in 3m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3d0f428911ae469da7332f846579fd89 : SUCCESS in 10m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/698c244365b94093b74889809b143d7f : SUCCESS in 17m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/db51ce256c244ee59d5c5c54438a6eb7 : SUCCESS in 18m 37s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/44e5890e91284753b4d30479ba31ceff : SUCCESS in 5m 36s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/051ed16e99ee491085535bfbd6e38f40 : SUCCESS in 7m 11s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a9e973eca4ac4105b6821e1f8bf37631 : SUCCESS in 4m 12s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/995fc652c289464a88ca6c02ddd6c550 : SUCCESS in 12m 26s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/03a1e71aac8c404fa6fccb2cfe731641 : FAILURE in 3m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/380a35b9f21e4dca967fdfb3d578586c : SUCCESS in 5m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f12fd7e44ea64fcc807ddd572c4711ea : SUCCESS in 9m 13s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/592c6dd8e39f4863bc1d81a9707bd586 : FAILURE in 12m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb3c8fba380f4d39bda6cf8ed3230dfd : SUCCESS in 6m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/38ae79a5118c4665b6a993fcb185a092 : FAILURE in 2m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b51c0060d1514515af8db9a522143248 : SUCCESS in 12m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/581bdedf07224dffa41cdf1e3d4d9efd : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fbff254b19464389b668315b802eb39a : SUCCESS in 10m 20s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ae0a8e425d0e46998e61492c5c16f0cf : SUCCESS in 4m 07s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9400a912ae594f678afe310a2ff56327 : SUCCESS in 4m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/717594b9d84d4fbdaa58354e25258496 : SUCCESS in 13m 10s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7fe4c4dc6df3412ba682e9265c347850 : SUCCESS in 20m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1fd89e389a19456f99ff575eb1b8b023 : SUCCESS in 7m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ab799c2023d54db4a14a1c9835c619e2 : SUCCESS in 15m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d9e1a538c65542e2917d90eec07ed2c5 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/992fcb74b7c746d588da68cd7e4cf452 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8cfe0900d11146b0a73e35b95b1159d2 : SUCCESS in 29m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/22c8718922b94618abe7dd6eac635ce7 : SUCCESS in 8m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f52f912377b449c29f44bce4489a6252 : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a87332030504d99ad456f4bdea3c9f9 : SUCCESS in 8m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ffaaf948d92c4e5792d0dabbce1f28bd : SUCCESS in 7m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bd910985c684ea9b60c37de8a936385 : SUCCESS in 1h 00m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34b96a1e947b4ccb8b0f5269dde4bf19 : SUCCESS in 1h 40m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e3fa7f0e2781b39a21c9ac5a2ccdd3d97ea1f0f7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-02 12:47:47.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0525262b53ab1cbf88aa96d47a8eb7f8d437e27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-03-08 08:09:10.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"14ba894ac97ddaf231178c416d1a11d8914e2d16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 10:08:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fea570565fca4c90bde1bca872166956\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edd8aa73c197455b85c95c1baf07017d : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de7bae737e34f3194c9b401a0be97b5 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fbfc39986754b4a92756793f4b1aa45 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/408e0a70e70749c49fad2587ca16d28d : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30dc8e40167b4f9a98b9898b7c8efd4f : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6022c2b0a7e64d0bbbd5c97043a51624 : SUCCESS in 2m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/57a55bffb0e94bb1b6c50379828cc112 : FAILURE in 1m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2f85e1196afb427798f9ca3959798c21 : SUCCESS in 5m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8b37f5abb0384d039d222603574f2864 : SUCCESS in 6m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e0746fde603c4c238cecc81971d7949c : SUCCESS in 10m 01s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5fd0cd9503764df09a63dc7be2c260b6 : SUCCESS in 3m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5c80b79c9d524105b2cacefb00162ef7 : SUCCESS in 2m 54s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7c314d67c3fa4d4cafccb0e6a5fc7a88 : SUCCESS in 2m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2c94d03bb22c4ee29eda54f62cfe331f : SUCCESS in 6m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/07676ac6a2de44c5b88370e20578c0b5 : FAILURE in 3m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/246cc025afc24a64bb2b1b3b5993127c : SUCCESS in 3m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9b3bf1886faf47c98e7b2a04d2e49690 : SUCCESS in 5m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2a478c306649480e8a0637462fd83711 : SUCCESS in 6m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bd262558004a4a96aca0d80764d91884 : SUCCESS in 6m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d68a4960a76e4fbd89d67fdf282bd122 : FAILURE in 2m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0fe9b26d48684d189f0921deb55573ec : SUCCESS in 11m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2e193b4dba28400eaa3fd87d33ce9614 : SUCCESS in 1m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8aac66196be848559e2d070ada2421c8 : SUCCESS in 11m 37s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6abace81939442e79b578c8ab7b3b092 : SUCCESS in 2m 50s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ef63bab0732f481f846db42db31ca13f : SUCCESS in 4m 36s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/426419cd74f24d3c9b59303e58af5ba1 : SUCCESS in 9m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c388177c4b8e409c9c59cf1a6c251901 : SUCCESS in 22m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/16e83db0c6df41bf91693236c38acd17 : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/72c096590e904765aa27ac8e93b4d344 : SUCCESS in 13m 22s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c802dac6979b4896b50b4efd385443b5 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d8aeac28ec034b529b1869732572bcda : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9654c2f094214d3aab89b6e0a9ba34db : SUCCESS in 25m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0cd4da341af043088b7369e7ddaee471 : SUCCESS in 6m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0fdb2f34ae7e43278de114ddf044aa6b : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a2f1c4c097541b3bf58f7303fc90d3f : SUCCESS in 4m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/057c58635bfb46f5ad18ad2384863f6e : SUCCESS in 17m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5171c11616f344c7b1f88cbb74b769b8 : SUCCESS in 1h 06m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c75a889398214fb0930cadaf9243e499 : SUCCESS in 1h 52m 18s","accounts_in_message":[],"_revision_number":2},{"id":"122b0e27407c93fe14bafcd2758412106c3bbda0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-03-15 08:41:35.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bfc2b7f4019458ba9a025ea57e8f53a4533ea18b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-15 10:23:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/896d7ecd276c47c1add59aabf803ff00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f3638d7f14543559a685527d869413f : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a90da50b9ea4fc39d5643858a870679 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e1ea788bdb943ea84824d633bdfde27 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d99870aa572e4a3aa4d34cf5586d96c9 : SUCCESS in 1m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/806a77994ca144bcabb30f5c7b760792 : SUCCESS in 1m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f9fedfdaf0c7490eae710f7b1afdbc78 : SUCCESS in 1m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6ef79d54455249a8ae373a2047d7dffd : FAILURE in 2m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/86fb33cb08d844fabfad8e14de4af274 : SUCCESS in 5m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cbea479a9a0a49e8ab6bce3f5ea52592 : SUCCESS in 8m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e3748a767236439199f0056156d52d2c : SUCCESS in 20m 22s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5d9bacb14a03448685146308c8ddba7d : SUCCESS in 2m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6864549945664aa19ce4cfb5e8c32520 : SUCCESS in 3m 37s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/29eab0f468fa476eb848da77aaab4c85 : SUCCESS in 3m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/934a253e1376497985182968266d7fa0 : SUCCESS in 7m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0b4428be35b34257b6d5cdecc1f437fe : FAILURE in 4m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ad79d79656dc41a7af99f40b49521f1b : SUCCESS in 3m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/61daee9e076547238c8a6cb47a1bdfe4 : SUCCESS in 9m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/87f2665ec2e84aef9036a09ad2592113 : SUCCESS in 13m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de7e6d7b2ed84fd68d59f435f1963526 : SUCCESS in 6m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8e4fb23b1a1b427388b6756e23b9109a : FAILURE in 4m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/735c45d1ceb14175bf05f2c6a305128a : SUCCESS in 12m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5f3c37b439164572b51ac1e4316a14cc : SUCCESS in 2m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/48b069a72c54474bb2df8d1011455a47 : SUCCESS in 6m 05s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c518d763054940f6af11bf57a415a267 : SUCCESS in 4m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8b49be0614b142f1b42a6121366bc230 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f4a58b75f31843afbbc6541b59bfd17b : SUCCESS in 11m 13s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ae1f7160049c4d0a815cee568683e74a : SUCCESS in 16m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f81810b552a540d0970c489da60a63fb : SUCCESS in 9m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/866f79312673469e8895e0d56d0aa0ca : SUCCESS in 12m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/50acc307c04c412abbfa3a6bcd7c56ba : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e72bd76d3d2d445797fae2cd687cffac : SUCCESS in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbd868f0f4b0431d8d2dbd028d2f896a : SUCCESS in 26m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f2b71525816a4cbc839984bf6dc6174a : SUCCESS in 6m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c4de90100f4d4e83bc71cb65028721d5 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7e26c9b724c4b7691761a07285a2395 : SUCCESS in 6m 08s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f34d67cb61fc48b296545962949d7345 : SUCCESS in 11m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b3a51871fc64902b5b5cd1631297aa1 : SUCCESS in 52m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/883f3a1aafc340e7906a10503f7084ec : SUCCESS in 1h 39m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ef5dd2598dc122e538518e9d57e956364be868fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-03-22 08:42:06.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"167e7b85874a9475aec291967a53dce7a5945346","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-22 10:17:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d5f79a6fbd04707a9f0c2931cfd55c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466e4f0b32074a8a8631ef98ac4edf19 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb22bac20c90467eaba21ac402d642d8 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d80869c6d7ac4e4c8c1781359612e6c5 : SUCCESS in 2m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f22f4ee215f4816be7e95c8d9cdc893 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3972dc7110bd4220ba94fb970dbffcd1 : SUCCESS in 2m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d85a1d1460db4fd6961f7d06bab6c56b : SUCCESS in 3m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/64a9e099e84d4e85b09322e48d00bbf9 : FAILURE in 1m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/472b76f92c1841e08f554c57f066d62f : SUCCESS in 12m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a356d21d01234ab8a196005a6b16e2f3 : SUCCESS in 16m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4438867f7e864dfab6411f51786e7fc4 : SUCCESS in 20m 29s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6daf2058443541728837401cf7af098a : SUCCESS in 4m 33s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/880e1ef1a1454393a52315e3b2b4498d : SUCCESS in 3m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f41011cebdae4c70a9bfb1b2e555c64e : SUCCESS in 3m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d576361c831340b184c3beac20ae3164 : SUCCESS in 9m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7c3be29fc05c43aebaf2cd6626fb8a05 : FAILURE in 4m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9cb00a249a7a4a52acb6deb42bdc543e : SUCCESS in 4m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/32f690e9ec7641d7ba15bf166dc43905 : SUCCESS in 5m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b14915c801a74f21bf8d7d25cebfaf16 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/378d223ee137425588eea1331f9e8816 : SUCCESS in 6m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/830d32d27bc44bb9a15e3d0001a5963e : FAILURE in 3m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ba954cf462ad429d8d666e298ad6dd7a : SUCCESS in 11m 07s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/84671dbf66974e2b9650b60bd9f0214e : SUCCESS in 2m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8c5afbe8abb641b086c2227103090d32 : SUCCESS in 7m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d9ba535d825d4637a5ace9e787015138 : SUCCESS in 4m 31s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6ee955a8bf41415f93ec12b91fad6447 : SUCCESS in 3m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/cd11042848ba401597594557f4a8ccc3 : SUCCESS in 11m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d89af6b39d87459a84031e5c09d74b4c : SUCCESS in 17m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ada18a7e59f84487a58b84b8907119f2 : SUCCESS in 7m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c5c98d8f24a34d188df761dd27e2610d : SUCCESS in 12m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e515c6a52443407e94803d6473ce83ad : SUCCESS in 2m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b6ca17f8a094b7fa69e3e32b2383e69 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e532c20caf5c4c97aea0f9f63e240515 : SUCCESS in 22m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/08777754117e4429a5dcf6b39d2fd9d6 : SUCCESS in 6m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8719864d3dd4416ca14ea8343207c629 : SUCCESS in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1eafe675c9a4a28aa0b9a72377beda0 : SUCCESS in 6m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2b0f0577cc0649518e58d90235a46355 : SUCCESS in 6m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/159fd84f28bf4844b2bc53cbec6ec2a7 : SUCCESS in 58m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cec89587c4da431680d9f6600ad7ff94 : SUCCESS in 1h 30m 16s","accounts_in_message":[],"_revision_number":4},{"id":"cce230d92def34e671157600bbeb908817ef7781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-03-29 08:18:48.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f23bb3eebfb1aece770afa70e85e2f0115679a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-29 10:02:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1ff8ef48d054c368bb74e369f565c6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1136ed4935d64b0399e6dee37e35d3d1 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b821d0a10d8f45ffbe961576e4442e76 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6fec1de4bc243938621590340bbd08b : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39702bee88ed496cb76e44a7ece5b7bc : SUCCESS in 1m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0feee95e2fa1487c9fdb546fb3b93738 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb00093c4f2547c286b10fa2011623dc : SUCCESS in 1m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/497e278fe4cc4676a281450fd37c1ed6 : FAILURE in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/207a4f6bd02448e79dd2226c8348e81f : SUCCESS in 5m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/14d3936d96d24628a79d70ce846848bf : SUCCESS in 6m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9a7246c5baea4c98a354b0b2598aaed7 : SUCCESS in 10m 50s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c7f76416c3744ab9926aa04c4c79b3ed : SUCCESS in 4m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b4cd5f447ebf47c29b7397d117e859c0 : SUCCESS in 5m 17s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/af390b64bf374c16aa26d3a0621486b7 : SUCCESS in 4m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2b110a2e9fa04b57924382571b7d639f : SUCCESS in 12m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b428cf108b99419990746fbc2f6ac11f : FAILURE in 3m 34s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2769cebe776d44389815b0fb363e5dcf : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/926b79ea12294c32bb53127a592f8561 : SUCCESS in 7m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f5fb21c859244c66bdb1fcf756d072d6 : SUCCESS in 13m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1fd9fd803f8c41e4aee089d0536cce60 : SUCCESS in 6m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6ceb7e2a7d454107bc1c3fc25e8aba63 : FAILURE in 2m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a4937a33d0b94ae18d8353eebe4ecb83 : SUCCESS in 12m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b8eb3a533f954de58b32049abf8ca61e : SUCCESS in 2m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/37117e48bf334ae38f9290712d0ea33c : SUCCESS in 9m 37s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7db7e7f05b984a3a8cc139776e56a3ef : SUCCESS in 4m 13s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/63cc7fbf39f5479e9bbd7f0f899ba43e : SUCCESS in 3m 22s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7b19a6c68f71434f842f7f61637a38ea : SUCCESS in 11m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/33627137a17d4558ba245b9c3f942013 : SUCCESS in 20m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1f33557fb5a144d9a73cbf54c660c7b5 : SUCCESS in 6m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9a605e424d2a4ccbae73e3324817f8b9 : SUCCESS in 18m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/59261a7b5c414710bf7987cf77e9be9c : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a5e7603997da43c7a6569c3826bca155 : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25241111f3da45bdb44685ce26463563 : SUCCESS in 23m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b8c71a007d6b44e0b088aaad6668b33a : SUCCESS in 7m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c6fc7fa368a445dfb6d78f0a3ecdc192 : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57f56cc5b9a84a35999b8d6fa0f079ab : SUCCESS in 5m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/157c0321b2c7425fb769967a5f0ce0a2 : SUCCESS in 17m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a69053e23dd24fe4961018f32e097643 : SUCCESS in 1h 14m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87d5abe515e042489235767be2ac7e8f : SUCCESS in 1h 38m 00s","accounts_in_message":[],"_revision_number":5},{"id":"9192cdd839707719a517c9e401c27bbba05bd78b","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-04-03 22:01:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"862648449816c2f9d6ab7be9b591f7350989187d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-04 08:29:25.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"016a6d2dc32df9e299696099bcc7e2ef4bd94a6a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-04 09:08:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3e0fcfbe2e62dd76fff9d0ab232234fa21a95989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-04-05 08:37:51.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"b74932f7b6fa1a5fd7bcc4600c495c157690aba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 10:46:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b13bb0c060414f7f8b32b7574d568677\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae3a856b3a64a3db637aa3eccf6530d : SUCCESS in 2m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e53c0e72cfd454b9f6fc92ded3a0a76 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ac5b24ae06c418a8d8fcee8e2b9efb0 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfb1ff682d56449eb617712c482666fd : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52a29c9ac55348b4b3c716461ab88e34 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/88dbb6d7fd0844808233ceb1b977a5c4 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4888bc3ff3c4d2abbbbea86fdb2b817 : SUCCESS in 3m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/346d3d3ad955415bbb876a0433b8cbaf : FAILURE in 2m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0bbc41048b824ec3ae54c355b6cd2e4e : SUCCESS in 12m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f52dc62de876485bbae3b0d0cc01faaa : SUCCESS in 16m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1ec8a9a4370244089bce1f7b94f903ca : SUCCESS in 9m 05s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6bfa7c060acd48bfade2f78d2f1be738 : SUCCESS in 4m 58s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e836a16a4525479ea595421b80379583 : SUCCESS in 4m 20s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/af323681a24d4bc387bff3be75a4c6db : SUCCESS in 3m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4e02870d4aad4492806e81b8b9a0017e : SUCCESS in 11m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/750ab297a5dc499c8716e7da54ce1f52 : FAILURE in 4m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b75093dff88c4aa58e0ada02f72c0843 : SUCCESS in 5m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5ec2ccf562de43849c3b386d67b2673f : SUCCESS in 7m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/833dc77a639b4afa9cfa6e48326b9400 : SUCCESS in 13m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8a627bb60f74330a76d14d2053a3670 : SUCCESS in 5m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/78fac503d0ca4d2bb6bc3dccd01e9ebc : FAILURE in 3m 36s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dd3678d637bd47deaa98865f21bb67e0 : SUCCESS in 11m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4383d9cc28354327b46a9cce37be883f : SUCCESS in 2m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/80842a20515f4e7c8393d2032ad8920a : SUCCESS in 8m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c7a4bf90bf31459c9a3eb605173f03a6 : SUCCESS in 3m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cdcee081bb934c448c1b5683210b9158 : SUCCESS in 4m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/42c23585526849e8b678cfd30abb02b2 : SUCCESS in 11m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e962626e684f4d8187272940546faeca : FAILURE in 5m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0462a0bfa58c44beb3d1d65b8d910012 : SUCCESS in 7m 18s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/001e2e51b3c1444f97e7f0c9df2e449c : SUCCESS in 12m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2b0d9c276c4b4d7789e13ea370d33a84 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ce42c92a2f3c475c842d3a94aa6f4e4a : FAILURE in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/edfce97443d44a97a5437114d0ebafb6 : SUCCESS in 25m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e7603fde2e344a0a8901d3ff738c1044 : FAILURE in 4m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/13828d3906c94d31915e94199af627e3 : FAILURE in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1299263b4636494ca8ae0625e9bab325 : SUCCESS in 5m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ce28fecc8248490bbc5aa2006a4e0e59 : SUCCESS in 6m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9599483a9064d1f995027450ec6910d : SUCCESS in 1h 02m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51c8184cd5354e059d07887726296619 : SUCCESS in 2h 02m 49s","accounts_in_message":[],"_revision_number":6},{"id":"513e5a43f7427068f1242e1e338d99bdc2e5356d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-04-12 08:19:55.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"3a7f42b1611383f5eca51b224ce203066612d7d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 10:02:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d4ea07ae3d44cefac532907e5f85c3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e38715e25f40bca635523889471ed9 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3d6160733a7452589cf43a257f93b89 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/954b1f9d32c347da96930bcb8c159d93 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10ed82fecb694ce1bc78b25d18dadd94 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48d832856ae9441fa6198ed3c75af0d0 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11e20fa67e6540159951fbbca6680e41 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88c618f5e12a4dac92aba811c9bbebd9 : SUCCESS in 3m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/db0f14d32d2746aba30b947bc441cc7b : FAILURE in 3m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9a65b43093374d0abf7fb3e9b54471ff : SUCCESS in 11m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/039bbd755be4469d9c17f17216792a90 : SUCCESS in 16m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d6157c58326248a6b6e4d49dc8b20994 : SUCCESS in 19m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/19e4b3370ac046ffa5df4119186a3b4c : SUCCESS in 5m 45s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/277cd9f5e8664720bbf88c558516bfed : SUCCESS in 5m 04s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4d14b99316d44af2adfc5e6068634f9e : SUCCESS in 4m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/95d6f62090194116bf18cccd5b5eb0de : SUCCESS in 10m 37s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ebedc03b8ef048ca8c04e847b1a21fb1 : FAILURE in 4m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b226c39c7dbf45a78caaa9dc237e7a06 : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0cafb04d7c5446be8b68c47d398c2697 : SUCCESS in 7m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6135fb5d65864c8d9b6d86d5a2dbac7e : SUCCESS in 13m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a54dfd5d1a34c78bda560ca321682e2 : SUCCESS in 6m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a3c5335cccce4d0dab38b70d53ac7cb3 : FAILURE in 4m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8deec5b957784be69107af13547f2315 : SUCCESS in 11m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f52d9eec28584951b970bd35938ffdf3 : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4debdfdc6361423bb4a8f90af2b745e7 : SUCCESS in 9m 07s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e954e2a1a84149d381401f2bd50eb009 : SUCCESS in 3m 33s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8e464a4ab5544e71ab95de61571711b1 : SUCCESS in 4m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9f8c43fa342f4e25bb3f8ca49396c473 : SUCCESS in 10m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/019cd30373114ab2904764f2e160d3b3 : SUCCESS in 18m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9c1d022db8f148a7ab943a4c922ea58a : SUCCESS in 8m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a331d1909c1e4c10b55b1a6fdf47b4e2 : SUCCESS in 13m 34s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/27cd1cbaf81c42f3bed52731e6206b54 : SUCCESS in 6m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc0612c42480491da86e5fe612f5e714 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b54ab0d584a240b8bd177b9cb684685d : SUCCESS in 27m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3621f8764f554d9da042385ede9144f6 : FAILURE in 4m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/68d8cf5866cc4185b74490315a3b44af : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a046cb1e4a4a4c5f82d49b8086ced16a : SUCCESS in 6m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6eeac045ed984a0ea3d41c122c869c66 : SUCCESS in 16m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98f82b293be943d287e790580da35fd5 : SUCCESS in 58m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8edbc791ac254f21baee26e0e50f3e1f : SUCCESS in 1h 36m 08s","accounts_in_message":[],"_revision_number":7},{"id":"6b89b9952bf07871900eb2b0ac8880aade5ade22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-04-19 08:18:55.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"33698fdebb8e531dd518932f99c941a07c81de03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-19 09:10:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/91e9a1f1b42b43ffaef7a3980d2f4267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312735104f9e4b239b4f54752fcf79a8 : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5211aac84ca6447287ca9529285bd1af : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fc344dd6ff24b0fa80643ab9de9e2c9 : SUCCESS in 1m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d16afb3944b7444c856a35382560f1e5 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc4553124a8b42c0bc84e3132eb70961 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74abdad343fb42719aaa8486a7e04ec6 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/38888f0078134103a9f8343ef02bc486 : SUCCESS in 3m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dc8ed1c16d024a45b9eeec8140860164 : FAILURE in 2m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6245eaf63d834f05bf42042e8150ad58 : SUCCESS in 5m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cff814ada0674dffb38aab86f0d9770b : SUCCESS in 16m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/120ca9fbf414404d9a7064ee13beb07c : SUCCESS in 22m 09s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/651a27be1f6e453bacc23e042b8ce48e : SUCCESS in 3m 08s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b88016d5d403423595623d1c646f2a46 : SUCCESS in 4m 45s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/61d81877bced4f53a611b51b2a3e40b8 : SUCCESS in 4m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7d0aac2ff05e45c2b6b1cf58658869f8 : SUCCESS in 10m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d9572e662e02443f8585bd81dc472db2 : FAILURE in 4m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f5165f6835de4a3ba7ca5ab2ea5bb857 : SUCCESS in 4m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b2577581403c4e6e94bb35eb71223697 : SUCCESS in 7m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5ef1a061353149868c817897e07e4b7d : SUCCESS in 6m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/093f203e93064148b00075594f704723 : SUCCESS in 6m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4611bd11e663445c984c509316634df7 : FAILURE in 2m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/14d0a61a0126489f9e0057914a9300e0 : FAILURE in 11m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e197198efdce4bf29634c8b2cc5216e7 : SUCCESS in 1m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/87eb052e8c0b4490a5365bb20c5c9cfe : FAILURE in 10m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/652bad614fcb4b9d9cdecd82caf13322 : FAILURE in 4m 08s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6b8f54fa5bd54c6ba86ca2fe1d169345 : SUCCESS in 5m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/50d690f374374553be284071e73c0afe : SUCCESS in 12m 20s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a88e0b78e84f4485ad48a166bca67c90 : SUCCESS in 20m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/30af3e0ffdf54006b302a28e35b57eeb : SUCCESS in 7m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/aca4f45d0715446c9a361c54c45c5241 : SUCCESS in 14m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3352805a58f44cf4a53078f51b88ff9b : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a33cead54efd4b66a84dc3cf29e63cfe : SUCCESS in 2m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2843f3a341d54391977051199597e7e2 : SUCCESS in 21m 04s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0747366f1ce74f7896b80ef3200f3a84 : FAILURE in 4m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b0a585a903dc43c9874bec3902e706fc : SUCCESS in 2m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b9c777fb6db4817801af2222fe22e05 : SUCCESS in 6m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e5f2634cb0764cd9a82ab2a91523de55 : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b185ac4da6d43b1a86bf4fd0170eeac : SUCCESS in 47m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cb0af8761134be8a068c568499f0a64 : FAILURE in 45m 41s","accounts_in_message":[],"_revision_number":8},{"id":"cd943006e7af6fc0ebd7553ec8080227351186dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-04-26 08:06:49.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"2f05613d59059ad86b17db9b8a418de579806ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 10:04:28.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4560ff06ecc447968fc46a44e72b40f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1bae6a0529408d9fa5536c8a0d9996 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23075a1251414a8f9d4c9cd68b11f935 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bd70e853c7a49018f87fcd6b5373fb2 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1dce8ab3b2e849569c149dac256c85fa : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/82838c6fa05f41d1887c6c471fa88a3d : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f35633ea995e4c59af8e017f60ea9b62 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28dae2d0621c4a3880a8f0da8f43a306 : SUCCESS in 2m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/60f18f69d3a943cfa19c6c7a21de76e4 : FAILURE in 2m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c42da9dc63f64b7d80a0780caf8d22b6 : SUCCESS in 9m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bbfd05b6975a4389a5c79a87a8934cf7 : SUCCESS in 16m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2a19f68799074a579de30bdc32ee9e7b : SUCCESS in 19m 31s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a984b87a15264665afc3b2519400ed75 : SUCCESS in 5m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d2f9d2439b80429e92f622777496d15d : SUCCESS in 5m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3e7952cb427046fb80106a6136b1338f : SUCCESS in 4m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8bd993a89d30467582e7b1498bd9a739 : SUCCESS in 12m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/35016b9d97bc4cdebe8bfc0d8c18625a : FAILURE in 4m 10s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5873c65ec75743ccb35b93bbe0d972a9 : SUCCESS in 5m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f107cd24db814b1f8a226bc83025eaa7 : SUCCESS in 8m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b24843a992be4450bb883a52ba72c9d0 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da15ed0cb74c4daa8097076a0a001b1b : SUCCESS in 6m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a6e2b49cd98c4b32bf631424df013cd3 : FAILURE in 4m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4a89a8967d7344a48f47a59608ede7d0 : FAILURE in 11m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6a29785623ec49ac8f98dd01c990eb91 : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ac5ac7c451e74b06a96f5fa967411ebc : FAILURE in 8m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0d8e61eda2104444832c3564d8ccab0d : FAILURE in 4m 25s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/68f52aa3a57d45fb83c52a699ca39ebd : SUCCESS in 4m 05s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/dcf37791f2474bf5a87f9eccbe796e66 : SUCCESS in 10m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b36843f04b9640fe9b62f3ceb4d8db01 : SUCCESS in 14m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5277fb3fc7ca4defa8a9c948664f5dc2 : SUCCESS in 7m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f90d3ff104bc4678a45829e16e6d034b : SUCCESS in 13m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f5afc416c7d147a0b91968f454f7c0c1 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fd987c4dcc2845a3afa9290a0d2d792a : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f74e02cf26cc4cf899a71ddf2d4efd20 : SUCCESS in 24m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/884c0f6b72014b95a2ea0b6e51abf229 : FAILURE in 4m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3b8bd76d38ed4b6288ef038fb93ec749 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf942cd9ab4e43ffbe8d0c4de68338d0 : SUCCESS in 5m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c458fb8e577d48058395570204123283 : SUCCESS in 18m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/747f3f9a595f4b14aa8dc8d7b1f337e8 : SUCCESS in 57m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c37697ec4504e94b51f16a63a12310c : SUCCESS in 1h 35m 30s","accounts_in_message":[],"_revision_number":9},{"id":"7f6e56e6c71f116933b1e2302896e2c6fb572872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-26 13:04:33.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"7292778a57a1846a51ee9320ba59f5feea39e538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-26 14:06:51.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":11},{"id":"eccd6e8cb6364f552aef5855db7cf3a7ddd6f695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 16:06:09.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a2a5bbce9a34f0893a6bf9bd191c4d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2ac25936cc46faae4fa928f8a810ac : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c82322ba7b4238bd47772aad0e2102 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7df9f1051534d8886263d70c22ed6c3 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/469c978e6cc140509b6f74b4de9ba40a : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4846f025b9f4eebb33bce103499a5f8 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dafbcd2e805340bf97e1247a33adfb31 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7894429f9b924e2eb59978449a06e4f5 : SUCCESS in 3m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8a901a723c6846c1a4404b1b28eb2c83 : SUCCESS in 5m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fbe26ef4540347d7a3b145282547476d : SUCCESS in 7m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/202f3435c97f4451a88586237ccca071 : SUCCESS in 21m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8922b2d0d20d4fae8a11559a6a46952f : SUCCESS in 23m 25s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3cbc1931c2b9472aaf585a5f7809d55b : SUCCESS in 4m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/96f7b59194bb4a159fb679670553525e : SUCCESS in 4m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/eff4c7d0e230476697644a10e87ee800 : SUCCESS in 4m 54s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fe9b8841c4194b7697ec192de70edcea : SUCCESS in 8m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c0765cdeae8e4f2984667674f310bd33 : FAILURE in 4m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ec622a7b431d4265b8a85cdb916e4723 : SUCCESS in 5m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9896493ae2b945368647861130f00bcc : SUCCESS in 7m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0564b3a5d07d4c4486c7ccd81045cd02 : SUCCESS in 5m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0b33a675683b49f0aa3b816dccd72e5b : SUCCESS in 6m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4259ef46587e4c20aeb522c44523467e : FAILURE in 4m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f61f05f20f1d48228e34304d32fd08c1 : FAILURE in 8m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6b5114daeab049208afbfd4f41a02613 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0a56259f11a94e06a5bf6ec7ad38fbe8 : FAILURE in 10m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c817fd0d22f6422cb731231a331e85bd : FAILURE in 4m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/510739882d4c49489bbb9c714df07d97 : SUCCESS in 4m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/97e735b0dc3748e0982097a204b0b8b4 : SUCCESS in 10m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d791ee4662bf44e28b6d303e73220444 : SUCCESS in 20m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/67b5a9451ce948f5aac8a72df807b65d : SUCCESS in 6m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f8096f610bd743e18442b2b377f29cf0 : SUCCESS in 12m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/53178827653e489fad17cd56980ac8cf : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/15f5d01ba804459499726908584381b1 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/539cce12981a4445a463fb84dc4e5cdb : SUCCESS in 28m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/90b54dc4062e48409ba0cfc28cfa3050 : FAILURE in 6m 05s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/df322e0c664641879881d382cddf09e0 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50c94a67f85f4e4592428f7e9be7d2d0 : SUCCESS in 5m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9ab1f3c405b04807895184c5869f928f : SUCCESS in 20m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e1fdb0a46114e6e86b29c93d71b6203 : SUCCESS in 58m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5f193eedbfa40708f4206d1ba9f994a : SUCCESS in 1h 52m 14s","accounts_in_message":[],"_revision_number":11},{"id":"4056c6a8d08b7ce678b38abfaec2c82b19d21ecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-27 10:11:42.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"dfeb859724051c771162a5df8992ba2d2cb44163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 11:57:01.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41c2764332ee4c2da35f7d21a9f120ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840c5814d61a4eb4a971f786448bdbc3 : SUCCESS in 1m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03aa72cf3361489fbebf6dbd6d1c3458 : SUCCESS in 1m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/beb5a43f62074c319423ce8288ee12a0 : SUCCESS in 1m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9467c9d2cb343229471891802dc5610 : SUCCESS in 2m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5625e871fe134fb3ab9be5483dc51538 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54e0654186ea4355a71a1ddd460ad3eb : SUCCESS in 2m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a20f93ac94848e6b58fd2bf979e3671 : SUCCESS in 2m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e06efd6e6bb44108a2cf5fd4f83c3d7d : SUCCESS in 9m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e3bdde9d0f9a4d79af7ec9ab4b5666fa : SUCCESS in 10m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d63b56b5ee454a738e6abd58894fcf35 : SUCCESS in 13m 59s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6ee2ce96a24c491c9ed9f12cebfaaf7f : SUCCESS in 16m 31s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3ae5a54b2aea4e3e9af4ab25d8212899 : SUCCESS in 4m 09s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/99bb1e31bd5c4f96a989868cc18d33bf : SUCCESS in 4m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/eb0c37bc879143dcae25b28415159e49 : SUCCESS in 3m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6972032a0f7d4d838fd0efe43f4d4fc8 : SUCCESS in 8m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d6a616cdc89f42edbd42da2844c73623 : SUCCESS in 3m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/417bf4b445e34e29a77dfe94fec1fbe4 : SUCCESS in 4m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/61c6ac4aead346a2b69f74d93f6facd3 : SUCCESS in 5m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/168e8e2e691b4f76be035a9eb2a95e94 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26d6476838fd47cab42b56e555d3bc4a : SUCCESS in 5m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/16faad3be2594cf097f20c37e29cbb7b : SUCCESS in 4m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2e0eae5233754d57b77fd2929542e064 : SUCCESS in 14m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/68c8700ab26f4c999241e707298cb6de : SUCCESS in 1m 35s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/80b5e94b6e674c83bb74bbe77daa5bea : SUCCESS in 11m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4f8a8d748fa044eb90575913e806787f : SUCCESS in 2m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9c488265473b4c65a4bce9df2cbb8351 : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/077209cae2c94c6f9e905890b62364b0 : SUCCESS in 9m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/583abf58c91b49b58dde14f6bce50015 : SUCCESS in 15m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ef1c04ac4ef04d7fac2ffed9c6ca15e4 : SUCCESS in 6m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/91f3ff80cf2c4d2f860db2ecac8c75b3 : SUCCESS in 7m 43s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/18d3d5e9cdc34bca8e09c057a46cd59f : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/726aea5a5aec4a0bbd95ffeae556d8e3 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/635dacfb784545d69150a01e6a662077 : SUCCESS in 23m 52s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5c9ea131318246d78a0ed99cccf24d9e : SUCCESS in 6m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/181a996b2bb14828b8276fe62a203621 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2a8be3eaa6b492dadad3408ab29eba9 : SUCCESS in 2m 52s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ddebf486a4de456cbd861669309b6166 : SUCCESS in 18m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98c7f9b0ae374f5b8b09335dc74b130d : SUCCESS in 59m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75e1665f440b4c86baecd97f48fe9d94 : SUCCESS in 1h 40m 09s","accounts_in_message":[],"_revision_number":12},{"id":"307c75a76478954400ec0ccb2d765eeaf485d99b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-28 11:35:34.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"1640d6f72540bc20c76ba6934b6ee03530b52e0d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-04-28 17:00:57.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"1e83a49833e2c11cd0aa07161bef81d526ebe294","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-28 17:42:09.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"67ca7592b458841a4af5fbf141474038d4c98226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-30 09:28:59.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"31558d0bb13430043a7c06bbac58a005f4280aa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 11:33:10.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e7c64541f9a4630b1ba0f109933f76e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4e62be650bb4137b4daa3e61ea392fc : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87bffa276ee944eca1f7c95ac1632355 : SUCCESS in 1m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fff7c19060f49fa885134e200700c68 : SUCCESS in 1m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e91e117f3bc47d2ba9083d6c9d34727 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8450ab03bee41008d8b50497501d865 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86ffb2fc155c4faebb66235e4b8abc3d : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2873272e8ac9480a9bac851efe337de5 : SUCCESS in 3m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2834e7119e17419e963630d4e54a5674 : SUCCESS in 11m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a8b97a744f334f05a3211c80657371a8 : SUCCESS in 13m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/98b59d05eb47439fbcdd2c1e9a48bc0a : SUCCESS in 15m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7660bdf86433478eacc4e9a0e587d31f : SUCCESS in 20m 52s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e12d989a705740338a6b959c092acc0f : SUCCESS in 4m 25s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a87e1ae979a049a9a73b2c708a2388c6 : SUCCESS in 4m 28s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/bc31ce756c7b4344afbe47e23ab6d0bc : SUCCESS in 3m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/261e3dfdbd254fd49cb592b7ffc99106 : SUCCESS in 6m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ed2fc48f995b4941b41d1900dac08ee2 : SUCCESS in 4m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9095bc0eef5146a0bbe87041543bc260 : SUCCESS in 3m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/20a0ba60c82941fcae8e25e812b7291e : SUCCESS in 5m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b1b0682cb359479db83822404a052644 : SUCCESS in 5m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1137f1c2e3f8436fbb960df7f4202b97 : SUCCESS in 5m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/eb19c3bb83e440c6b23dafa72b6f8d49 : SUCCESS in 7m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/65b7a2a61d794f3c9cefcb55100b86ce : SUCCESS in 12m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1d14ce2cdfc14d6dbd6bbc61d0392fd2 : SUCCESS in 1m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4e0adec6d657414697525b665828358d : SUCCESS in 10m 58s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3fdd7a1e01794d9fb2950e641537806a : SUCCESS in 3m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/96cbfbdc50114d1eb28f15b5646b3c5f : SUCCESS in 4m 38s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2a80e4e9d7b14b0a8aa5e8775176cd3b : SUCCESS in 10m 39s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9efecf00da8a4a8082f6930039b69bcf : SUCCESS in 19m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7bc5f49f36b847e7aad9eda2a798eaf1 : SUCCESS in 7m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5a94b54badfb4a0fa10945e8641e5d37 : SUCCESS in 17m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/56028b341d704556b3d64eed914dc486 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d196e7182b6a45168d5b60460eb1d60f : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b760063fc6d54e9fb07e3449f8a2cf40 : SUCCESS in 29m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/af82f97242ff4028a68fbabc469d57c5 : SUCCESS in 6m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/485f097d22ad427796e6551d84661021 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f9971692b7b24a2c984a11b297a509b6 : SUCCESS in 7m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6df9751944a24098bc12a9639a1984f7 : SUCCESS in 17m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ad60bd6daa04e91b9cbe333cf54420d : SUCCESS in 59m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49cf378e04394a02ab0b85502730eb48 : SUCCESS in 1h 58m 26s","accounts_in_message":[],"_revision_number":13},{"id":"6eb143fd4f970c9562362e0df5969a5a86c51777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-03 08:18:13.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"d345eee8acc42ec5af30cc7b0e6a96a4cf07d1d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-03 10:10:25.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/caf5d6b6ee364644a1aaecda32731fc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3cb2729b5b421bba51d04b897b1c4a : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af07a75cf4b847ad8292e7f15c0dc47e : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7fdd45cb3dc4425b5851a95830f004c : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8402bdc8d93a4652b8e5243ebfdf829f : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c02398106a64061aa551779f2c1fcac : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ff981923d3341f6a7604fb2191a1b9c : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17118cf0f96d46bd8c1f4c91de73aab8 : SUCCESS in 3m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/97aebbd081be427aa9cabe0dece030c2 : SUCCESS in 14m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/704348748e0546f3aa813ae64884a75b : SUCCESS in 12m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fba0feabb81e4f44822531371f6fab0e : SUCCESS in 21m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a04a81f8298d4f6dab6895bd83beb40a : SUCCESS in 26m 01s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/21b1b2bd2f924b2ab18d351808e35dbd : SUCCESS in 4m 46s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/84a7e7c74f9341ce91b0b20b06caa219 : SUCCESS in 5m 15s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/14af8ed389c343e3b28667b7a4b962e6 : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b76112dedf3a48aa97ac13d97ddf9137 : SUCCESS in 11m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/af681ae7141b497cb7adca46a1fedd90 : FAILURE in 2m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ee13b77b66cc41c5a7eaf13bc12acd71 : SUCCESS in 5m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c3b164af76f24c8980a4e2b91e675d0e : SUCCESS in 7m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e9bb4f4b1475431d9edafe1775544399 : SUCCESS in 13m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a2264a9afc8429d84d1316fb05964e6 : SUCCESS in 6m 18s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/459b49c47b4345498263dfea54b2491c : FAILURE in 4m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/24a5b33401784472a0c926ca95d10c51 : FAILURE in 12m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7ffddfae133847f5a6e9654e26d5a0ca : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8564a7922e114741a53dae3535695e5f : FAILURE in 11m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a8f346dcc23e462790498f5badf64586 : FAILURE in 5m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/da179b951d57473587e5cc2327977fa4 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bb01a04b911f4af0a4b25167f164ab18 : SUCCESS in 12m 10s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/80afa5737b5e49a8953cde6a723d8a16 : SUCCESS in 20m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ec009125e0e341b1a8184faed5875ec7 : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/05bbfa10618e464086f69687762ea034 : SUCCESS in 17m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ddd62511d6ab42c88261869cd13d8522 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1663a380e67e460ca5f65aa7a6693ec2 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1aa2c2ccef34f01887660234d8d3338 : SUCCESS in 28m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5aabc970ec13404db57eaa772a88461e : FAILURE in 5m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/45b3a44bf6c84345836696f7a1d4b52f : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05f33524bb764814bba9c9c90fb06e5d : SUCCESS in 2m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6f9b5aa428714c4f87c10f9f78e7cb15 : SUCCESS in 18m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8429dcf507f44397a92dca9dd4d6c79d : SUCCESS in 54m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2bdacc4266f4411c84e08ffbd5c6d19d : SUCCESS in 1h 46m 03s","accounts_in_message":[],"_revision_number":14},{"id":"c6416e3c478b7b843149373188bd08d7d7679526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-05 19:39:12.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"4a1163b01af97dfc614464205dbb920ce586e218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 21:37:24.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92b766d5735a4455a640276c4dbf8238\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/747dc85b16c5412fa18b2f20d6b03fde : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b248195ffcbb47c5a94390b7b97108ab : SUCCESS in 1m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be702b43d9d64d5a854eb58e42dba614 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0bfbf2be03fb4abcab9597382926af1c : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13f372ffd8374332ab78546dbed14eef : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f72cb271ddf84fc6a8aa89d5e1cbe152 : SUCCESS in 2m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/558fff39eab24296bbc9cfd95557f1b6 : SUCCESS in 3m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f4355a657e6949829e9fd1f92e5efaf8 : SUCCESS in 14m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f7747529c64e4fd5b2b0ddfb5cd7d014 : SUCCESS in 18m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/009e784878b3426cab75baf4e404aa4f : SUCCESS in 24m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5ee009ad734e4f2c8c228e6e7aaf0023 : SUCCESS in 25m 23s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/218d2e3f061044d6b97b7256d54dac7f : SUCCESS in 4m 43s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/082d7804bd564a1887fb3371b7090eae : SUCCESS in 5m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b1691d23aca34d5ba3a056cf6b4c9a75 : SUCCESS in 4m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8434b1ac80ab45faa32c351d4182a830 : SUCCESS in 12m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1f45d8a3325b4f438afe20d8220047b5 : SUCCESS in 4m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9a841d6af0294af480e532ca1a8cff4e : SUCCESS in 4m 41s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/db5353d2a0de4394a91cdef0706802cc : SUCCESS in 5m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0d1c4f69b2264caa85b866dd33b6dbd5 : SUCCESS in 5m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b65df86eb51949f39256ee5fb636339c : SUCCESS in 5m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a35889d8e0e64127872b35160a17e634 : SUCCESS in 7m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c1be0c7770604d29a41b74c6dbc621f3 : SUCCESS in 11m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/43401e9c412f4c5391c2a3022768df71 : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8ef98f983a0140219c9774f7fe0a110c : SUCCESS in 8m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c0d84b3da0b84556bf14422f1b31b835 : SUCCESS in 4m 08s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4e6247b092774f35a8b236908725e710 : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e717d61696ef4a80bb7ea36012ead37a : SUCCESS in 11m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/40e4248e1bc346a99f50956f98501613 : SUCCESS in 20m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d0129cca568742cbb68e0dad8da3f386 : SUCCESS in 6m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/41c4b00dd60e4a04b3b13cfb09946be9 : SUCCESS in 13m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/cfbf53eef0f24b28ac375b221d3155d1 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d48935bf6935451b8ef10c530e542631 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/571c82f7a9044491b0272fde7eda4915 : SUCCESS in 26m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/09b81cd2ea9b45eebd9a9efdf276a7bb : SUCCESS in 6m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0a6b4d2be0f84733b9fb75a0ca596831 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/318a34b63e7c4620bb222842b1932a34 : SUCCESS in 6m 03s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9e2997adfc054029bf89985e7b91c5c4 : SUCCESS in 18m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68fe6dbac4d84863ba62f702b19873e1 : SUCCESS in 1h 01m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e600b8f83171492bbca29ce1194b7658 : SUCCESS in 1h 52m 53s","accounts_in_message":[],"_revision_number":15},{"id":"1c9871c13f35180530dae70d269f719c09a609b8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-06 05:34:35.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2913a12b15e33d7d7f6e7dde494999896dd74bf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 05:35:07.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"376628eeffb9f7e3f94dfe657384cf7308bf23cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:19:35.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ed663a97f114497ad886caf3f26d15d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca781e74f464dfe9a867d2b42d6bcfb : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1212019ce1d64e1b80d46b460c5a6363 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e8ac4db37d648e98ef22bfa07c2218f : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea5c192c94584cb58a30821262a55c7a : SUCCESS in 1m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/232d12ae00d54ff4954e1a22dd946ac4 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f861b298287749db98ac31a4704bb941 : SUCCESS in 3m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9d8d89186ef845e2911faca2c24784c8 : SUCCESS in 16m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6629c38b95cd4a4198a086798bf4152c : SUCCESS in 17m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d103c6d32ada436f8a820a8b21fe615f : SUCCESS in 19m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9f10264ff92c47c2af36b040cf153e72 : SUCCESS in 11m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/56c9ec5a2f7249ca9470e183fa6299fe : SUCCESS in 4m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/301c3aa7d5204ee0a48217e960c272aa : SUCCESS in 5m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6a578c8835f8424199b38efb2a676046 : SUCCESS in 5m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c6e989eff2f14a4bb0eb5c17fe89aa11 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91d2c0d0af6b45908a62db44bf9cfaba : SUCCESS in 6m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/29cbc657dcfd4bcfb0160bd342c42dec : SUCCESS in 7m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9ef6d6d83fdf4e73807424b23a8223f5 : SUCCESS in 12m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/23a316fd217e4448ae7d59659520e035 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f27f4843fabc4235a2840fee432be8fc : SUCCESS in 9m 26s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/56e84ce2d0c2474cbb99dba8c0ded2d5 : SUCCESS in 21m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7102653abdbb4621942bbfe15678873f : SUCCESS in 7m 18s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4fdff3b514314063b7772dc435f026ce : SUCCESS in 13m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a79269a08bb64f449fa5369019764c2e : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/11ea65178b464f1fbf61c27f1c5a52f4 : SUCCESS in 1m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c0d80b0eae64d5586c58d58b70645ad : SUCCESS in 22m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/58a6a084878d44c5958672c9464aaab1 : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/89bee9c60281404c870c92978abba517 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f4c5692164d4019b2ff34d83b5397a1 : SUCCESS in 6m 03s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/35d257bbd5784a9c8f3683d11399cf1e : SUCCESS in 6m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0634a68747e44b05ae6a5a6e818b48b9 : SUCCESS in 59m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3fd8386e2f74ac794e0f81fe1d9c4dd : SUCCESS in 1h 39m 59s","accounts_in_message":[],"_revision_number":15},{"id":"2adb1c51f444c74173dbc9b17e8f305300aa3b84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:19:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"528336a7a7b8d1456b647b9a5c509f1459ad4562","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:20:28.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/630cc0ce3a7c46068d829bc541807d12\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6fb4bc8faf421990c3d79463c1e328 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"04bc4b165469f4ad1e6a66d250ee7c537486126d","revisions":{"35ec6cba8ce020f6c0b6bcc7bd11e327c2e4f840":{"kind":"REWORK","_number":1,"created":"2025-03-01 08:21:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/943089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/1"}}},"commit":{"parents":[{"commit":"c4ead3c16c003cb4631850e4f201c7b3ee0507ef","subject":"Merge \"update constraint for python-glanceclient to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4ead3c16c003cb4631850e4f201c7b3ee0507ef"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-01 08:21:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-01 08:21:26.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35ec6cba8ce020f6c0b6bcc7bd11e327c2e4f840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35ec6cba8ce020f6c0b6bcc7bd11e327c2e4f840"}]},"branch":"refs/heads/master"},"b9c0b51e955da11b85eacf8bfd1dbaea93f7000e":{"kind":"REWORK","_number":2,"created":"2025-03-08 08:09:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/943089/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/2"}}},"commit":{"parents":[{"commit":"39f3fb28302cb384aac7113af4f2ebaadf385e9b","subject":"Merge \"update constraint for openstacksdk to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39f3fb28302cb384aac7113af4f2ebaadf385e9b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-08 08:09:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-08 08:09:08.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9c0b51e955da11b85eacf8bfd1dbaea93f7000e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9c0b51e955da11b85eacf8bfd1dbaea93f7000e"}]},"branch":"refs/heads/master"},"c760f7c6b424878b15d3c7328f3afae8d56ef41e":{"kind":"REWORK","_number":3,"created":"2025-03-15 08:41:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/943089/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/3"}}},"commit":{"parents":[{"commit":"39f3fb28302cb384aac7113af4f2ebaadf385e9b","subject":"Merge \"update constraint for openstacksdk to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39f3fb28302cb384aac7113af4f2ebaadf385e9b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-15 08:41:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-15 08:41:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c760f7c6b424878b15d3c7328f3afae8d56ef41e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c760f7c6b424878b15d3c7328f3afae8d56ef41e"}]},"branch":"refs/heads/master"},"59d917962a99dd65a68cab38b75994b828e8d49f":{"kind":"REWORK","_number":4,"created":"2025-03-22 08:42:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/943089/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/4"}}},"commit":{"parents":[{"commit":"7a304c74f76991597bfe7eb0666b2b2a2bf2ffa3","subject":"update constraint for taskflow to new release 5.12.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a304c74f76991597bfe7eb0666b2b2a2bf2ffa3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-22 08:42:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-22 08:42:05.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59d917962a99dd65a68cab38b75994b828e8d49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59d917962a99dd65a68cab38b75994b828e8d49f"}]},"branch":"refs/heads/master"},"9a0c7163843466b3375cb333c952e91d0b89eaf2":{"kind":"REWORK","_number":5,"created":"2025-03-29 08:18:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/943089/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/5"}}},"commit":{"parents":[{"commit":"be97f2507812c9047c5a39ee4a35b88ea7f677d7","subject":"update constraint for virtualbmc to new release 3.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be97f2507812c9047c5a39ee4a35b88ea7f677d7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-29 08:18:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-29 08:18:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a0c7163843466b3375cb333c952e91d0b89eaf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a0c7163843466b3375cb333c952e91d0b89eaf2"}]},"branch":"refs/heads/master"},"fbd9487f0807c501c8ada741eeac688511846fd8":{"kind":"REWORK","_number":6,"created":"2025-04-05 08:37:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/943089/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/6"}}},"commit":{"parents":[{"commit":"aea4bdb03846d4b08c0b3decf0ef6dec618a14ad","subject":"update constraint for os-traits to new release 3.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aea4bdb03846d4b08c0b3decf0ef6dec618a14ad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-05 08:37:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-05 08:37:50.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbd9487f0807c501c8ada741eeac688511846fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbd9487f0807c501c8ada741eeac688511846fd8"}]},"branch":"refs/heads/master"},"90bce089d1c1ded6551ace1b9a2e2546cd996091":{"kind":"REWORK","_number":7,"created":"2025-04-12 08:19:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/943089/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/7"}}},"commit":{"parents":[{"commit":"a795befd0341722114868ae51bba9fc3eeea8a88","subject":"Merge \"update constraint for openstack-release-test to new release 8.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a795befd0341722114868ae51bba9fc3eeea8a88"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-12 08:19:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-12 08:19:53.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90bce089d1c1ded6551ace1b9a2e2546cd996091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90bce089d1c1ded6551ace1b9a2e2546cd996091"}]},"branch":"refs/heads/master"},"f063fd44f24ca9aa2c7068cccdc6c7b034216ac0":{"kind":"REWORK","_number":8,"created":"2025-04-19 08:18:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/943089/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/8"}}},"commit":{"parents":[{"commit":"1cecf886b2df680590d5eca285fae3ffe899ce38","subject":"Merge \"generate-constraints: Make use of ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cecf886b2df680590d5eca285fae3ffe899ce38"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-19 08:18:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-19 08:18:54.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f063fd44f24ca9aa2c7068cccdc6c7b034216ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f063fd44f24ca9aa2c7068cccdc6c7b034216ac0"}]},"branch":"refs/heads/master"},"ac8add854c9ac7941001254b0ffd5772c9e22a5c":{"kind":"REWORK","_number":9,"created":"2025-04-26 08:06:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/943089/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/9"}}},"commit":{"parents":[{"commit":"c0c86e0a09a6395d75d03d874ce8f73d3886399d","subject":"Merge \"Bump pyroute2 to 0.9.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0c86e0a09a6395d75d03d874ce8f73d3886399d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-26 08:06:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-26 08:06:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac8add854c9ac7941001254b0ffd5772c9e22a5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac8add854c9ac7941001254b0ffd5772c9e22a5c"}]},"branch":"refs/heads/master"},"6607741cbc3f793d1cdda57fe753a0feeb5ce333":{"kind":"REWORK","_number":10,"created":"2025-04-26 13:04:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/89/943089/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/10"}}},"commit":{"parents":[{"commit":"83fd8c6af7bdf1d99cb79da942c2f723ff8922a2","subject":"Drop python3.9 from upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83fd8c6af7bdf1d99cb79da942c2f723ff8922a2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-26 08:06:47.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-26 12:56:05.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6607741cbc3f793d1cdda57fe753a0feeb5ce333"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6607741cbc3f793d1cdda57fe753a0feeb5ce333"}]},"branch":"refs/heads/master"},"0689e0f594d0d4b14f53ac34f5871fdcbf209f83":{"kind":"REWORK","_number":11,"created":"2025-04-26 14:06:51.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/89/943089/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/11"}}},"commit":{"parents":[{"commit":"dfc4189062d8a638cc04b6cdb2e2c35de21680e6","subject":"Drop python3.9 from upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfc4189062d8a638cc04b6cdb2e2c35de21680e6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-26 08:06:47.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-26 14:06:14.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0689e0f594d0d4b14f53ac34f5871fdcbf209f83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0689e0f594d0d4b14f53ac34f5871fdcbf209f83"}]},"branch":"refs/heads/master"},"f3a964cb95626ae234456b7a3992af43d416a751":{"kind":"REWORK","_number":12,"created":"2025-04-27 10:11:42.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/89/943089/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/12"}}},"commit":{"parents":[{"commit":"dfc4189062d8a638cc04b6cdb2e2c35de21680e6","subject":"Drop python3.9 from upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfc4189062d8a638cc04b6cdb2e2c35de21680e6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-26 08:06:47.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-27 10:08:10.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied:\n\nKeep version specific pins for:\n\n- async-timeout\n- exceptiongroup\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- oslo.middleware\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3a964cb95626ae234456b7a3992af43d416a751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3a964cb95626ae234456b7a3992af43d416a751"}]},"branch":"refs/heads/master"},"f07fcb81514c246231ba89f1bb91e6739c497d4a":{"kind":"REWORK","_number":13,"created":"2025-04-30 09:28:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/89/943089/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/13"}}},"commit":{"parents":[{"commit":"1bebf7a7f8c0009e7afc5711d8d4a91d62bc04e2","subject":"Merge \"update constraint for osc-lib to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bebf7a7f8c0009e7afc5711d8d4a91d62bc04e2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-26 08:06:47.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-30 09:28:32.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied:\n\nKeep version specific pins for:\n\n- async-timeout\n- exceptiongroup\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- oslo.middleware\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f07fcb81514c246231ba89f1bb91e6739c497d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f07fcb81514c246231ba89f1bb91e6739c497d4a"}]},"branch":"refs/heads/master"},"e33a5a8bbea6ff01735972923f88878f44aeef84":{"kind":"REWORK","_number":14,"created":"2025-05-03 08:18:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/943089/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/14"}}},"commit":{"parents":[{"commit":"5ad45c0c1d4150b0bada72ab8bd1f0fdd2c6fc01","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 5.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ad45c0c1d4150b0bada72ab8bd1f0fdd2c6fc01"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-03 08:18:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-03 08:18:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e33a5a8bbea6ff01735972923f88878f44aeef84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e33a5a8bbea6ff01735972923f88878f44aeef84"}]},"branch":"refs/heads/master"},"04bc4b165469f4ad1e6a66d250ee7c537486126d":{"kind":"REWORK","_number":15,"created":"2025-05-05 19:39:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/89/943089/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/943089/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/943089/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/943089/15"}}},"commit":{"parents":[{"commit":"5ad45c0c1d4150b0bada72ab8bd1f0fdd2c6fc01","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 5.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ad45c0c1d4150b0bada72ab8bd1f0fdd2c6fc01"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-03 08:18:12.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-05 19:35:09.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- async-timeout\n- exceptiongroup\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- oslo.middleware\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I093b2b7013bc7ed75f4a01701888a6975ec59863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04bc4b165469f4ad1e6a66d250ee7c537486126d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04bc4b165469f4ad1e6a66d250ee7c537486126d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
