)]}'
{"id":"openstack%2Frequirements~943344","triplet_id":"openstack%2Frequirements~master~I78c0d53fbfecffa9d904e9b717c01b7b79c940f0","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2025-03-05 05:07:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I78c0d53fbfecffa9d904e9b717c01b7b79c940f0","subject":"Bump openstacksdk to 4.4.0","status":"ABANDONED","created":"2025-03-04 19:14:04.000000000","updated":"2025-03-05 05:07:33.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0135f1fce7e04fc3a9f8a33487540526c3572369","_number":943344,"virtual_id_number":943344,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-04 20:50:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-04 19:50:26.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2025-03-04 20:50:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b1180830b0e0a6b805b70080b3924c7dae3d702","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-04 19:14:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a68a1607c24bb615c24494c10f83bf8b22956a6e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-04 19:14:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a185705cb1bc33a5730416d1a4baba5bcc045e44","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-04 19:14:39.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"63dbfcaf015ce53e15e98718651cb5614faecd21","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-03-04 19:50:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65b623dfdcc073f60fbfc497b92c7d6fbbd05d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 20:50:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e745b3eda7440d0911c9eccada78bab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c2f99e190f4b9596ab68ca9548be32 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7877925435e4e0682ab37fca2ce04d9 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba1eaf95f51645ce89936289cbce37f0 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43a3d5011a264f40bb8c00e468afdd03 : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e91923f91b174d19b9d572949ac40bb4 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4c98ccc8a5614d3398622d06dcdcae58 : SUCCESS in 3m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8d61bbbb4b8941f8aaffc450fb9f57e0 : SUCCESS in 14m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2905ce78bdfe4154a4cf56048095ca90 : SUCCESS in 9m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3c33c7689fd4420d9af960a044ff2d20 : SUCCESS in 16m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3fc1cd8a95474cf1827affc7ef49cd92 : SUCCESS in 22m 49s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b5ca7a3a7bae482b8eb01eb8a92b44c0 : SUCCESS in 5m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bc3bf3f77017421689fce48ba113a8ef : SUCCESS in 4m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1352bf80d5364a5ca3ab25f624a44852 : SUCCESS in 3m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c28ca59250314634bb7260ba8f99e032 : SUCCESS in 11m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d204af92dbac4c10bb785da8978f31aa : SUCCESS in 5m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f3a4ea495efc42d1b57edd3330e882f2 : SUCCESS in 6m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8b869b4878ae4728954de81dc0892301 : SUCCESS in 8m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4b05b40adf154ea2ac91b0e90d147cb2 : FAILURE in 7m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a97b4cb44d2462da6ce98c2b8883c53 : SUCCESS in 6m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f039be3bdbe34e69a64562f7c79a1f9b : SUCCESS in 9m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/56da434c1c6b4308b36c44048e9b4573 : SUCCESS in 12m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8de111efd3c54e2d93cf4ab095a85f11 : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e6d823cde30f4d44b86e456592196399 : SUCCESS in 11m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f187ab7c616d420d8883e4a8cbcc6505 : SUCCESS in 4m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4f8ae01f74564e95970cd54ddde6057b : SUCCESS in 7m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/848e7bcfcca14265af2ce0b7b2dc348a : SUCCESS in 10m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/137a365fd5e34b829dc1e15780a16360 : SUCCESS in 18m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d42151a10bd147b49c2a65baafdb6111 : SUCCESS in 10m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f6d8a6081c89488c8c93aa773485cc27 : SUCCESS in 13m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3fb0b6e90daf442d98cd11ef55bcbf19 : SUCCESS in 7m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1380171cb30a40e189cbc66ddd5451ef : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2088f1e788544a7a2c1991bf3fa15ec : SUCCESS in 25m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c6cfa6fbe2c34d65a82c889e57416048 : SUCCESS in 6m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/02763d2b9cc243dd8fd5275016c8f668 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbffb4fd03f84ef3b3190fa26aacdf98 : SUCCESS in 5m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e084fc7dc5894d4f9eb1ca58d2bfcee2 : SUCCESS in 13m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09b29a3cd5214b18a7a91af432deb9e6 : SUCCESS in 1h 05m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70cd53bb8bc94f1b8f2b93bd6189193a : SUCCESS in 1h 34m 06s","accounts_in_message":[],"_revision_number":2},{"id":"0135f1fce7e04fc3a9f8a33487540526c3572369","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-05 05:07:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97a2f7d926dfd021ae44f8b459cc635b669c6b8f","revisions":{"c0ff998806eaa554890ac220ef06e47b925ab007":{"kind":"REWORK","_number":1,"created":"2025-03-04 19:14:04.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/44/943344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/943344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/943344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/943344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/943344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/943344/1"}}},"commit":{"parents":[{"commit":"c4ead3c16c003cb4631850e4f201c7b3ee0507ef","subject":"Merge \"update constraint for python-glanceclient to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4ead3c16c003cb4631850e4f201c7b3ee0507ef"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-03-04 19:14:04.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-03-04 19:14:04.000000000","tz":0},"subject":"Bump openstacksdk to 4.4.0","message":"Bump openstacksdk to 4.4.0\n\nBump openstacksdk to include fix for baremetal client.\n\nChange-Id: I78c0d53fbfecffa9d904e9b717c01b7b79c940f0\nRelated-Bug: #2099872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0ff998806eaa554890ac220ef06e47b925ab007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0ff998806eaa554890ac220ef06e47b925ab007"}]},"branch":"refs/heads/master"},"97a2f7d926dfd021ae44f8b459cc635b669c6b8f":{"kind":"REWORK","_number":2,"created":"2025-03-04 19:14:37.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/44/943344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/943344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/943344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/943344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/943344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/943344/2"}}},"commit":{"parents":[{"commit":"c4ead3c16c003cb4631850e4f201c7b3ee0507ef","subject":"Merge \"update constraint for python-glanceclient to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4ead3c16c003cb4631850e4f201c7b3ee0507ef"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-03-04 19:14:04.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-03-04 19:14:37.000000000","tz":0},"subject":"Bump openstacksdk to 4.4.0","message":"Bump openstacksdk to 4.4.0\n\nBump openstacksdk to include fix for baremetal client.\n\nChange-Id: I78c0d53fbfecffa9d904e9b717c01b7b79c940f0\nRelated-Bug: #2099872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97a2f7d926dfd021ae44f8b459cc635b669c6b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97a2f7d926dfd021ae44f8b459cc635b669c6b8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
