)]}'
{"id":"openstack%2Frequirements~943582","triplet_id":"openstack%2Frequirements~master~If51dc75a38c8335f9e14286574d3569845855047","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If51dc75a38c8335f9e14286574d3569845855047","subject":"update constraint for python-magnumclient to new release 4.8.1","status":"MERGED","created":"2025-03-06 09:55:36.000000000","updated":"2025-03-06 14:54:45.000000000","submitted":"2025-03-06 14:54:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943582-new-release","meta_rev_id":"d308e552d727892a5b63cead252ddf021fd0ac36","_number":943582,"virtual_id_number":943582,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 14:54:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-06 12:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-06 12:38:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-06 12:03:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-06 12:38:16.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"021a48b9ea9515cf06980ae14f5b63ba4a128698","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-03-06 09:55:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5faea9005a36ff7c6b9c525a932b2f95811fb7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 12:03:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1af77ef848b4d85a331a207f435b79c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebdf842bfcca4c548e3248302fac64e1 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a1d912b13064093b9469144afe010c9 : SUCCESS in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c9bc82a981d4643820e3e1506fac763 : SUCCESS in 10m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08510c758b68401da9d5ff07c2b3de8f : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9c28b217ec4460ca8f2b50d665d0258 : SUCCESS in 8m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08b4f4f2aa9e4b099db3dc7fc5780cbc : SUCCESS in 4m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/014637e7fccb4898a49635048ccce55c : SUCCESS in 15m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0271a6b8b34a46b9bd78d7cb4eda7557 : SUCCESS in 9m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cd3e388a53a74f2097aa050bc095a006 : SUCCESS in 16m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0c39a1c9d86749f7aef935c82cc9e1dc : SUCCESS in 25m 02s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f30c9c807c3342828eb9a861c34c3c59 : SUCCESS in 5m 11s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/85739a39f8c14df6b4918e6191d9e83e : SUCCESS in 5m 25s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d8307a28b9bc4d1180f77f6f2adc3efd : SUCCESS in 4m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/35f1b65bac394ceebd50985aa84a8a57 : SUCCESS in 11m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c1c7a5c720454fe4a4ad1982cffd3da6 : SUCCESS in 4m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4ed5362c57374b6da1250f675ef2442d : SUCCESS in 5m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a3c63bef8a3848eb8f46a483fb979159 : SUCCESS in 7m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ab66e3cfd06a41169d4d56406dc23359 : SUCCESS in 18m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6656578bda0c415096ac6a86161dffa0 : SUCCESS in 6m 11s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ddf6a124a2ab4586ba05873cc4195697 : SUCCESS in 7m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6693408d0f5840ceb411230920806530 : SUCCESS in 13m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7487e529db484f8c94c9fda3eac47bf4 : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2349eb4bbe784a229c1e2e6321f8cab5 : SUCCESS in 13m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5499bfa188e746a69ae8c8dc868fd085 : SUCCESS in 4m 18s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/553c9351ebf24b24ae8f59df18a22f9f : SUCCESS in 5m 14s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9bd26169452c4f919c1ef421d1148717 : SUCCESS in 18m 05s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2b3c05a1cbb14eb4933cf6c3bfe7acee : SUCCESS in 25m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f741daecfc49487694b4fea1a2febce7 : SUCCESS in 9m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a05fe31f39c84e02bf84df156de91f8a : SUCCESS in 24m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8cb5731157ee4499bed701740e47c0ca : SUCCESS in 6m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3345c455d9f462098efd900b2767cef : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/da04dd2ed8ed4503a008102fe8abe1c3 : SUCCESS in 23m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b503304a50ac4e8eb6de109c089d96c7 : SUCCESS in 14m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2c3eacaedbf349358d12ddcf87e855a6 : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e1fd650ece64636b4bab11bac68088e : SUCCESS in 13m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/14403ec6622f4ff39f1161cfc51f9321 : SUCCESS in 15m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e654576da4774ec194a68decfa77a7de : SUCCESS in 1h 03m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2039398c042244fe9608480cc17f72b7 : SUCCESS in 2h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1051d51a921a19cf41a40906560d36aaf899c0eb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-06 12:38:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27f384479f2e0a687e97f6bfd8ad58aca6341ce6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 12:38:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"043d25d08176deb9d738f1cb9f82b3ef078255ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 14:54:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f0c51b75bbb43da908e0457ecc16649\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cfc0aa9cfa44b890ba6c9c06987247 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/296aa841f3214330b260446e0d125998 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9303f838c7ff4ec59e1464d034d74aa3 : SUCCESS in 8m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5000c1962f9b43cebba11ef4990301d7 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/72dce79877b34efe909f6c8eeb0c5fe3 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/645f16066f2c4990a7ae6b8007096fe0 : SUCCESS in 4m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d24666cfbf5e4def947f9d7051e439ef : SUCCESS in 16m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f1e8fe4dbc34acc9fcc3edf17c3f4bc : SUCCESS in 8m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b4ff4a13bacf48d19fa9e4f06b81f951 : SUCCESS in 17m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a1ef867c7c7c4ff79d45a9c4d91b6be0 : SUCCESS in 22m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/92ac331c3bc04377ba099767de114bdb : SUCCESS in 11m 37s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/447db0ba022e4f749c126d6d3bd596e2 : SUCCESS in 5m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/390145e86cc045c0990b0d29c29a55ae : SUCCESS in 7m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fca59406aa494fed9b8f0ed41c637e5e : SUCCESS in 7m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4f51e271483e41eebddb0a7c1d410729 : SUCCESS in 15m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7f2e8193e2b42cab4272bd113c6fb72 : SUCCESS in 8m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f8d7d3d965a54f168a8eb00702b579aa : SUCCESS in 8m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a7e292c040c84ffb906cbbaebef0d00d : SUCCESS in 11m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6b016319d1c04a88afa2bb2630ef882c : SUCCESS in 2m 42s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d3d25ed83d5347b7810952311953366c : SUCCESS in 9m 21s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2404dea964cd4fd4a62a13b5c6be6c36 : SUCCESS in 20m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/229e5b874ca04af99944c5378a399147 : SUCCESS in 10m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d18a5edc6109476491bf67fc1edaaa43 : SUCCESS in 12m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/52b2c16a98ac492cbde9ab14db9bd233 : SUCCESS in 10m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a2727e798b80461a97734ee72cd0be68 : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b2434185e29a4744b4813fa6df5de2f8 : SUCCESS in 25m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3e69bc344fb845c59652df3ba515ca49 : SUCCESS in 8m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5c047191c8784705a51ca13b01055b17 : SUCCESS in 6m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40d9938fdebe4d11a07b1a3bba430a08 : SUCCESS in 5m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8264b19a28154121800fc2d235210f1a : SUCCESS in 16m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9592d4606fd64d6b84ee3a72e961817d : SUCCESS in 1h 15m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4e803210a66433fa57f7aa7d75a224f : SUCCESS in 1h 56m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d308e552d727892a5b63cead252ddf021fd0ac36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 14:54:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d143240dfac96a655ca14cea6a21144eaba4ec6e","revisions":{"d143240dfac96a655ca14cea6a21144eaba4ec6e":{"kind":"REWORK","_number":1,"created":"2025-03-06 09:55:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/943582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/943582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/943582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/943582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/943582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/943582/1"}}},"commit":{"parents":[{"commit":"c4ead3c16c003cb4631850e4f201c7b3ee0507ef","subject":"Merge \"update constraint for python-glanceclient to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4ead3c16c003cb4631850e4f201c7b3ee0507ef"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-06 09:55:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-06 09:55:35.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 4.8.1","message":"update constraint for python-magnumclient to new release 4.8.1\n\nmeta: version: 4.8.1\nmeta: team: magnum\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I8051be03f46ea76791fed5f016bda805b5bc912e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Dale Smith \u003cdale@catalystcloud.nz\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: If51dc75a38c8335f9e14286574d3569845855047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d143240dfac96a655ca14cea6a21144eaba4ec6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d143240dfac96a655ca14cea6a21144eaba4ec6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
