)]}'
{"id":"openstack%2Frequirements~943583","triplet_id":"openstack%2Frequirements~stable%2F2024.2~Ia67fb94213e4a3e251923b67696fc1ec96246351","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"Ia67fb94213e4a3e251923b67696fc1ec96246351","subject":"update constraint for glance_store to new release 4.8.2","status":"MERGED","created":"2025-03-06 09:58:39.000000000","updated":"2025-03-28 18:08:08.000000000","submitted":"2025-03-28 18:08:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943583-new-release","meta_rev_id":"a9baeaa5ecb1a43bd9f1e694e921a64c4b450a45","_number":943583,"virtual_id_number":943583,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 18:08:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-28 14:49:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-03-11 15:31:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-28 14:49:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-06 11:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-11 15:31:33.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-03-28 14:49:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"837eed8a141ad3bcdd2f92c36ca2b598ad0af8ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-03-06 09:58:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"402c11daf139e23f6293ff22d243ac8811774d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 11:37:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccb9f5742fdb42cb9eaef66d257e3d91\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/919065217d6342d18d96e270a3b26e23 : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e9497458d474208b7bdfc772bb2c477 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df739d5814e74496bf276aea2c4d5c2a : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a34ad3f0bf746de91a137ca4c131e2e : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ddd07316428541839d2b36cb42f1d88d : SUCCESS in 5m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45049205bbbf4e7ba8546d3f1de33763 : SUCCESS in 9m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/283987d366c04175ba8c042dc795d514 : SUCCESS in 29m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d2e1366f3e0f4b3fb6bc2324897c5d57 : SUCCESS in 15m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f758c1268244f4f82f285f9f00d7c46 : SUCCESS in 13m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f653bb3e3a7644769b89e1090b036058 : SUCCESS in 14m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/360b2cb3915340e9929f811b90ddc583 : SUCCESS in 19m 19s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/3292ae54833c4e749a55ab645ed3c503 : SUCCESS in 4m 16s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/fc6382faba01432ba6797a8cc0028c5b : SUCCESS in 5m 00s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/1e9f4b3000ac423ca9bd52d34011b115 : SUCCESS in 4m 15s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/49658144af224f7ba9e6df9a71870b3a : SUCCESS in 11m 35s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e955338fee044ea8a60b35743ebb9fdc : SUCCESS in 7m 22s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b454df94b1e44f6785bbb365c8afc71a : SUCCESS in 8m 01s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/905bdf6aa0ee4ee29d804dbb0e02eee1 : SUCCESS in 6m 26s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/80c4348b6deb42b389c4bcde851bd966 : SUCCESS in 13m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92d3367ed861438da33b62222733ea93 : SUCCESS in 7m 36s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/8dec0802ea8b4a19b39fec43a6d03e1b : SUCCESS in 8m 08s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c0ddbbd74f2a438fba1a224485d11f59 : SUCCESS in 21m 37s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/34e0e5a7191d457e83637ed5ec815c65 : SUCCESS in 2m 14s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/7c7f96655cca431281bd3f4872eeefd0 : SUCCESS in 10m 49s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/99fd4b4f909c473481f32afa4387eb90 : SUCCESS in 4m 06s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/148a9734a51f40a3bc43365481100e7b : SUCCESS in 7m 00s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/35d85d5b376a43c78fbf68cf93194762 : SUCCESS in 10m 41s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/b6e00df705444586a48f2cf212475eb6 : SUCCESS in 20m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/23fad78c50a24f8bae8cda7077f92b87 : SUCCESS in 6m 12s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/3f703db53fbc46ea9699e256e8df3b9f : SUCCESS in 10m 48s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7c13f3cc65fc46a69ddf79430f45e86e : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/139c6235e10e41a59bbfb47fed3f26ee : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57b36e8b2c9c4b82aa2fd197fde964d9 : SUCCESS in 22m 02s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3a452d4066a2499b99377f222ed9abd4 : SUCCESS in 4m 26s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/d62075f5c24e4d028e505476c8c853dd : SUCCESS in 8m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/78b73e07a3e64eb3b7794ee2a5c23a7c : SUCCESS in 5m 17s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7ca39bc43c8847c180ae648cd3375aa5 : SUCCESS in 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8261908120574e1eb0977b65eda15dff : SUCCESS in 59m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85a5aca3d5d34547a7060fdbb8baf382 : SUCCESS in 1h 32m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8f85b64ec51deff24d4e4d0aa29b1aa5657776b3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-11 15:31:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5aa6fabb70b3883b14ca6af8c70b1230b86285f7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-28 14:49:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b9307cf2e8b768a4091dd3cd90443fa7b4957a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 14:50:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec78bc29b7876a2a53eeadd7bcaeb4f246b9755c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 18:08:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3c3f5f097584e109352733d1f348c19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dc1d8ee49384b769dfe781db7a05a21 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5acea2b466954e2caa735b07b3a31082 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82ddc07d42a44258b73cbc1cfbe7b2f5 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55b0d3251e444ac296c1a3ea80071583 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3d56e67fd6740d2a55714cd7b30e5c4 : SUCCESS in 2m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8f459f5a05f447dabef9f232f597c4cf : SUCCESS in 8m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/131c6c6826ad4bcbb27eac41ef347986 : SUCCESS in 13m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dc27d5c9ee5c4a20ac23192bb9d38bb8 : SUCCESS in 9m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/96ce019514144a94a21610986b34fdcc : SUCCESS in 15m 07s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c20f44c6184444fabb697c02f619077e : SUCCESS in 13m 21s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/1a4fd5478ac443799ece1a18598e3967 : SUCCESS in 5m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6f6ecce6923341ffad5af0db326190e3 : SUCCESS in 4m 54s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/7cf905a20947482aaccaddb4f32694ab : SUCCESS in 6m 56s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/67ce6883c78a448195062b0f49165197 : SUCCESS in 13m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c32e502db56d4160968e4c932bd7427d : SUCCESS in 5m 17s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/24ee01bd2ded470fb1d8df4468f5a0ba : SUCCESS in 9m 03s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/cd7af00d40904b9f97a89b68db0d9dcc : SUCCESS in 11m 06s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/057c08783f474f6cac17ea1daefd5751 : SUCCESS in 2m 31s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4eda3d8eec18456b859061e676a97dec : SUCCESS in 10m 40s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/246c45ec23f242c2afb1efe0f3effe89 : SUCCESS in 18m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/65ac19094c9e45a3b71369a60352e08c : SUCCESS in 5m 52s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/109b459a72674085945dc2d5161327e4 : SUCCESS in 9m 46s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2cf6d9071a51474ebda339f1233e7b70 : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3a9334f72f3b4c4180938bfba6f3777f : SUCCESS in 2m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fcc6120bb68b4651a82c2cbef18f62cd : SUCCESS in 26m 16s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/566900c6b07e41fdaf91ef1bf27b21ba : SUCCESS in 4m 02s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6ae710e509d74b26bab1ab526a6ae2c0 : SUCCESS in 3m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb2ae7593354470a87cdbe824aece0f8 : SUCCESS in 5m 41s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/755165ed40034d02996b12142194c7eb : SUCCESS in 20m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55d197d783f04cc7aefbadcb2ce39318 : SUCCESS in 48m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21e0a495459a4415b4f40d5fa07dd818 : SUCCESS in 1h 32m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a9baeaa5ecb1a43bd9f1e694e921a64c4b450a45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 18:08:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f8a83ad7fa37e9656d0a7fa94b4f71653c44656","revisions":{"6f8a83ad7fa37e9656d0a7fa94b4f71653c44656":{"kind":"REWORK","_number":1,"created":"2025-03-06 09:58:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/943583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/943583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/943583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/943583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/943583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/943583/1"}}},"commit":{"parents":[{"commit":"d620ff557c22c4b8c1398b6dc84772c341fb4d5a","subject":"update constraint for python-openstackclient to new release 7.1.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d620ff557c22c4b8c1398b6dc84772c341fb4d5a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-06 09:58:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-06 09:58:36.000000000","tz":0},"subject":"update constraint for glance_store to new release 4.8.2","message":"update constraint for glance_store to new release 4.8.2\n\nmeta: version: 4.8.2\nmeta: team: glance\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Cyril Roelandt \u003ccyril@redhat.com\u003e\nmeta: release:Commit: Cyril Roelandt \u003ccyril@redhat.com\u003e\nmeta: release:Change-Id: Ifcccd372dc517e0ce13d47f597160d01de436659\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia67fb94213e4a3e251923b67696fc1ec96246351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f8a83ad7fa37e9656d0a7fa94b4f71653c44656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f8a83ad7fa37e9656d0a7fa94b4f71653c44656"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
