)]}'
{"id":"openstack%2Frequirements~944392","triplet_id":"openstack%2Frequirements~master~I265fb5d983d3eefadedfab8cff1421d6b8591479","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-05-01 10:40:06.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-04-09 07:57:46.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-01 10:40:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I265fb5d983d3eefadedfab8cff1421d6b8591479","subject":"Re-add tap-as-a-service to project list","status":"MERGED","created":"2025-03-14 14:49:39.000000000","updated":"2025-05-01 10:40:50.000000000","submitted":"2025-05-01 10:40:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944392","meta_rev_id":"451a5bfc8bf71688b15dc789e95f6fb8fb1e72a2","_number":944392,"virtual_id_number":944392,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-01 10:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"all":[{"value":1,"date":"2025-04-18 01:01:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"value":2,"date":"2025-04-09 07:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"value":1,"date":"2025-05-01 07:52:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 16:53:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-15 12:39:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-04-18 01:01:32.000000000","updated_by":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"reviewer":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"state":"REVIEWER"},{"updated":"2025-04-29 16:21:20.000000000","updated_by":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"c636fdac37bde7e4a915c1591606544b0dfcabd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-14 14:49:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c89b2816d85ed26522f55c0d0a681a3fdd3f7686","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 16:53:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b605f670dd4840fba3868bfd610ff333\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd0d1630ffa4f5fad6d0b958ea15baa : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82fa5eda8ee244ea808df36c157a11ad : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0843800accb8485496111be674f71603 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f7a492b0d244b13895ce04fa9095b2c : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70e862ae315e4990a33d02d5b9f87f16 : SUCCESS in 3m 29s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/2399780cb1174b8faaf68719c677fae0 : FAILURE in 3m 32s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5acf47fcfe444581b61e43f2764aaad5 : SUCCESS in 1h 12m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86a3e012a55f417485a5cae025a00260 : SUCCESS in 1h 57m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ee4f636e40d262e858377a41f853a5dd151e0578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-15 01:46:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"61f42c1cd51d870e0c2c6e5f93e7f5c2bb5c9ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-15 03:40:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eb23a9146154779bef797b5b01333f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8749eb9bd50542d2979e4c2a3d54369a : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c1216edc5da47c9bbabe61b921b07ca : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/217054f0b11947d197ca425871733ef0 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25ea6d22e8a54764908920bc83dc556d : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4fff913b74b4bcc82339125881e9f0d : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bdabb8c52bb04605ab46646f3a6418a6 : SUCCESS in 4m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/adc70b96faa34bc695e1f8f7a819f54d : SUCCESS in 16m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1878a80e74d044128dbf041b261c008c : SUCCESS in 9m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/af0624747b6f4424a35d3d1ef57f4c2c : SUCCESS in 13m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/740c0a52173b48f4ac77f79fe96f21e1 : SUCCESS in 8m 41s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/7e1d5970f52445019b569b34d26cfd2c : FAILURE in 3m 18s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d7981c5b065347a296be26008d9161be : SUCCESS in 4m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4248d4d3d8dc49d19f7938b59e8a6763 : SUCCESS in 3m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/989be8828729470c8bc36cd19760513e : SUCCESS in 4m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2e84f05e24af43f1966774eb52094951 : SUCCESS in 9m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c6d392649fab4279954ad7c289d501f6 : SUCCESS in 4m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/408741458164449cbd738aca98beede9 : SUCCESS in 4m 51s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/673fc5c987294be08ec3a2c16d2b7c34 : SUCCESS in 6m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9bbcbc26ef7842ddaa3f3245ea61cca1 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/482f87d4875a4ee385a45e4bd892abae : SUCCESS in 6m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e198a2f7f77240bd8398f1bdf2f97892 : SUCCESS in 6m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/03eef6dded69499c916dcf5dd4ed5a4e : SUCCESS in 8m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3e0ca1658a204816b1f11aabed9f0b98 : SUCCESS in 1m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7a96af18a55e4c73a6d943d4a2b81e78 : SUCCESS in 11m 20s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/058e037f5ff041c6ba66b171b758ca51 : SUCCESS in 2m 36s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a6fccfe891ee4a90b94bf94e8f6ba723 : SUCCESS in 4m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7e46538d324341e2a8db7344f5bedc3f : SUCCESS in 8m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c1b5aa40b65a41338133ee8f040f1a98 : SUCCESS in 16m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d5bb77763414044bdb2c5a77ce0e780 : SUCCESS in 8m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c8f0406422c34d0786a78f1712f3bf79 : SUCCESS in 9m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d3ed8a367ae5452a8e3dc46428606dba : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a77a5f55756a466fbcf768b4c20e6b79 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b0829e46fa14a67904838a3080d8239 : SUCCESS in 24m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/267014aafa364f83b011c663aa3cacca : SUCCESS in 7m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/faabd2da1bb645e0bcfcf191355c7ea8 : SUCCESS in 7m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bfea3e801b6246cdaba821da74a409ad : SUCCESS in 6m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1c692144e78e4f9780791f7cdd78e7f0 : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/160fc7e35482479a94610f0dfedd8da9 : SUCCESS in 1h 08m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/752ebc4a508248a2a5fee11e0a68e37f : SUCCESS in 1h 48m 15s","accounts_in_message":[],"_revision_number":2},{"id":"0e4cfa7f5abd47a8c3d881eb944ddc7e999c85b0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-15 12:39:26.000000000","message":"Patch Set 2: Code-Review-2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"330d2bba140235909d766fd61a52586c7509ab6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-16 14:30:40.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3e0b8e5e48f234bfb16c7cb78be5c4afc0a0efb1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-16 14:30:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1fab2e804a1922b8a18ba7b288862ae84e3d348d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 15:48:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8ed9abfbd4e4eebb48ef541832d3b41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7186497978864ed3bb71412a2f61e2bb : SUCCESS in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9702630b7fc48deb5ff2de4161d8b0a : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed4c7b6c84234acdbbb6e8ea0c6686b8 : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69aa025c71a04920b692f54c80744939 : SUCCESS in 2m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0fac50dc93cb4704b76f14bd0ec89329 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5e9900574dc429c9bf02e3b13d902a2 : SUCCESS in 2m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/923846fe95ba499b8a617b3d05964246 : SUCCESS in 8m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/38da3ac8a3ee4057bbb51603ea8d5c6f : SUCCESS in 8m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/97407ffac176472d9cfdb9c9e7be442f : SUCCESS in 18m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/84c0e227362e461392e1a9437a77b15c : SUCCESS in 9m 12s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/d2bff7ddd28e42fe95a1d15d9da9139b : FAILURE in 2m 51s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f026f70404fe4b36939ba6d3ac8eceee : SUCCESS in 4m 29s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/30580a987b704294ace6b6854c710667 : SUCCESS in 4m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c6ab7188794b435faac29ed30809bd74 : SUCCESS in 3m 36s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e3aba41b71bc4d2b994c66b00eaeb864 : SUCCESS in 8m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8600f929193147e1b6d402137f9250a4 : SUCCESS in 2m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/615fba42d80045b3b2072bd1c1df8001 : SUCCESS in 4m 40s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1fab93dd13a54be9a159a1f1b532ac8e : SUCCESS in 6m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ee4c7854ce1f44488f32e8f00af37821 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea7bc0d9faa44e6daa2f92f6735123fb : SUCCESS in 5m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6f869b72c6a5441ab86782c5ebf91796 : SUCCESS in 4m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c6061f20ff16424db51aac7c802a2fe7 : SUCCESS in 9m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3ccb05f4933e4ba789c853017e786a04 : SUCCESS in 1m 58s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f3d9731830ee486799c9ffd93cf9d846 : SUCCESS in 8m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/049395f5a9c440a394f82f8540819efc : SUCCESS in 3m 49s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6eed8301142d4e81ba74393235bc2726 : SUCCESS in 3m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/fd0585096a004f6ca9aa1f3aeb96ced9 : SUCCESS in 8m 40s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ced54568621f4e0fa8b1543469ec84d6 : SUCCESS in 15m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5d6edbe82d2c4ffcb17685915c1af093 : SUCCESS in 5m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bbc887e614c84373991bf9521ff623b0 : SUCCESS in 10m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1066767af2fc4d8f9249a0ca29bcca56 : SUCCESS in 3m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cf4e8e7619544bbd9b1e3755ce506855 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9eb72649bca045999367ab1712d4ecbc : SUCCESS in 25m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/305e0168ad5145359f01768b0fb85945 : SUCCESS in 5m 58s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a28c82e42d6045409c50632d8c6b0299 : SUCCESS in 2m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62b261f667ee4ceebbf4c7f8331e7283 : SUCCESS in 5m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bec2a20cd1a446afb2f5aabb8801cc2b : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/925538779fe14a67a9dfcb417df6803f : SUCCESS in 1h 10m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4967b0b152c74efea00f1822cc4f603e : FAILURE in 1h 06m 56s","accounts_in_message":[],"_revision_number":3},{"id":"cd25e90a879b4b327c5df1bc9e28c2e13ba7ffb6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-28 16:16:28.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d402417a5fd302ff39efb61b03facf3f360248da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-29 02:05:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1ffb4209d2af794fd5ab746d67c7ad8d30084ed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-29 03:52:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b9c087c93814f7b9a8ecb54687eeecb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae689341b20e4ed38cb4c521ba0180c1 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4d6545015b446959f9348d511dfec8c : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb414ada561b44c1bb41953fd2de9e2d : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/477ac6e09a7343feb2ac6f6cb24c5ece : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4ba0d8a82d541fdb4977c43b013a047 : SUCCESS in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e0d4f1d10b564c339629447c8fb64bd0 : SUCCESS in 3m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a2da977fd5194ee1bde89e3d6399d158 : SUCCESS in 17m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/495128bdd80749fcb96292748a247229 : SUCCESS in 7m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fbde213fa2ef41b1b41d622c5c5c7796 : SUCCESS in 10m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/67526333af224bb1987ac4896bf6ab1f : SUCCESS in 18m 50s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/1a4e01d245d04bed9d5d7937e7c6ecfe : FAILURE in 9m 24s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b2bb14005ec74265a6662ea10a16952c : SUCCESS in 5m 11s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d70778e23cac4c9aa241df688a914975 : SUCCESS in 5m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/716ca5b6c7db4effa439293761a2cb0b : SUCCESS in 3m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/544a164f3dea4b82af4e24c0b3db7e00 : SUCCESS in 10m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1624696716994c6c97a47f9212f61479 : SUCCESS in 4m 34s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/49374268ad614c67831f8b379fca8d1f : SUCCESS in 5m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/af6e3432d18f4011887b453f757823a8 : SUCCESS in 9m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/06e7d076f474433b8f3de251577ad0cf : SUCCESS in 14m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f25cf1a1900d4d9e8bc2e8be78028bcf : SUCCESS in 6m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2445d815d37f4d6884605b5cd2e755e7 : SUCCESS in 8m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ceaa5039fcb1476dabdb8db99d0f5d99 : SUCCESS in 13m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/46518b70627e40eea8f7a651687ac9d9 : SUCCESS in 2m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ac4675bf6b174cdf8be5082598ba365f : SUCCESS in 12m 09s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/41ced4e5b65741b6bbf1e456ff94e49f : SUCCESS in 4m 09s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6493fa40cfac4c2d8897be125e2c08b3 : SUCCESS in 4m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8cb3e75ccf144f6e922c42e2f47514cf : SUCCESS in 12m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d0e3265b9fdc445e8f9ecbad28b4fa13 : SUCCESS in 20m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bbfcde66f5204331a811b5f4b12bc649 : SUCCESS in 7m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7076311ecb3c42b490f0fea6616e3fd5 : SUCCESS in 13m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1f1d4663d81d49199de201fac5257130 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2ee4137bca7d4f8a995f5fb03b8a98ab : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/031b4cd1cf2941b5bfe6165b1f0862ea : SUCCESS in 24m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9f7f4d3357ed41cf8b8004cf62bf9f73 : SUCCESS in 8m 28s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/034d9df28d514cc39ae43e2e411c764e : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2db1520e26704fcfa6b6d18ef7ca132a : SUCCESS in 4m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/67355a9317da4cadae5c2418e93f86f6 : SUCCESS in 19m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e534f837ae014df6be3699b7519348a8 : SUCCESS in 52m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c312f429e5f24d4bb99cfa8d77eeff98 : SUCCESS in 1h 22m 20s","accounts_in_message":[],"_revision_number":4},{"id":"a95ae1b830cf52f709967e28bb21a193a79ba855","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-31 07:58:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"258d07a9102f17b52835edd5bf731cdc1f7357e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-03 07:56:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"88d4226fb1a74a4714972a6422d0cd7643e8dac8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 10:07:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/622caa548f834f459bb854766d970964\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4654ada0907240bda9c126413d1fc60b : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c452a3c1424319b8a2f97367a9b586 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5f323a4414940cfa8366c0d8be778d4 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe4608d86ea5437fa12680215e50822d : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d400ae4e524e496dbefbe96383e8bf1e : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5544edfb4946427fbfe84fa3f23dfbed : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c1c853ec09314e95b5e3d73ab6594300 : SUCCESS in 3m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f03ca1166d57461aa401678682294ea7 : SUCCESS in 20m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b28118c08de444b39e04994749a00d96 : SUCCESS in 8m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cbe932f2db65419281461eaca527ef53 : SUCCESS in 15m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1deebeda678e4055bbcae1001a3d79e7 : SUCCESS in 9m 27s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/c2f4682e9056409d83eb7b17a26614db : FAILURE in 3m 06s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/641f6e4349934e69afefa227838f9ce2 : SUCCESS in 7m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3095096405934ee0a8ef7a239af401b4 : SUCCESS in 4m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0a73df21f9444af9b85eac52bf780b1d : SUCCESS in 2m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8fe1e683d64e4278974310bfeea33552 : SUCCESS in 13m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/51e63ed274c242fe9826ea1e3a1f9b92 : SUCCESS in 5m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d149e5e8c94d496d8c96114473bbeed4 : SUCCESS in 7m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/218ec91f2b62480086d58371ad2ff791 : SUCCESS in 7m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d6c48937cdfc4d22a58ac96a4072956c : SUCCESS in 14m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/defd99eca6af4fa992e1783c5ed4a732 : SUCCESS in 7m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6f17d96b3c46402584a4122d137ac8af : SUCCESS in 9m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/593a11ac4f26488f81d4d14203f23fef : SUCCESS in 7m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/25f677b8e028497fb658821a62dd0836 : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/af7baba499e54c88ac8fae1514e08be9 : SUCCESS in 10m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1a25532939b142ba86c5533302c35828 : SUCCESS in 4m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0a9f54054b10434fbb8f346150455aad : SUCCESS in 2m 36s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/124134c2b65a4de186b68f1a856e80d4 : SUCCESS in 11m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/34115031b63c40a79e226f1b31d92daf : SUCCESS in 20m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/72c2d25908af49c3b67b63d2f127926f : SUCCESS in 8m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/12e11b6504494b5f8fb56b1d184f2a8b : SUCCESS in 12m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/62f0b3da3988440883964d5a924ae972 : SUCCESS in 6m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a0901a5ebad4c91a478ad0a59e4449b : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa0bb3372d78487090d9994573b36f3e : SUCCESS in 25m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/01f02bdcbe8a4ba5a03c408ac6f40d73 : SUCCESS in 6m 36s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3af058b915d648e393ef57bbf22328d9 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b6d1d1631844ba7b3a76930d0f572e7 : SUCCESS in 5m 42s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/262272c209734698b6d62c529620cd4e : SUCCESS in 14m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d589eb2e60a428a8843b8292c2f99db : SUCCESS in 52m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e474e5659f744a518f0b68457ed12c51 : SUCCESS in 2h 03m 29s","accounts_in_message":[],"_revision_number":5},{"id":"fa4d13c3d10cde75b06b97a719dc28820f6a565e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-09 07:57:46.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e455cfc37fddd2e5c6545dcf181a0ee9188743b9","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2025-04-18 01:01:32.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"330c98ef94725b404f63f729149537012722fd2b","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2025-04-29 14:46:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"45d96b117a4a59d546c9fa8a9e50e9eba4937297","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-01 07:52:19.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"647185ef5d9f8c90071748f7a78283f1f6b49a12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 07:52:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"761333194062d2ad0de517bf96176a63f8e8c3cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 10:40:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76665e57ffaa40508158503eaed6e0cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc6c587a0b0464ba5b07ecfe0379c51 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1fe8f91e1bb40c4854074a31db05c62 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3beee46948d64cccbd39cb8406395b25 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf6b27900a3b44579e07380af3f6d465 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f34917e21c7427888bce38bfcf0acfe : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82a6d4d1acd742ffb19b768968723229 : SUCCESS in 2m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f60444f8622141aa85620e8f78ffc6a8 : SUCCESS in 10m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b9e7214ebd464acab788d4a2ca72ae46 : SUCCESS in 16m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4524aefec4b644e0b7216a482de032d6 : SUCCESS in 24m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/549795a500574cd8aa08a28e6329dbd0 : SUCCESS in 11m 34s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5fde984a356141349327cbee9a41cd6c : SUCCESS in 6m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/90a887d7002a4994b8757b1b037d8cb0 : SUCCESS in 5m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/42a7b17cd71f4c8c81d35ec688b1eeab : SUCCESS in 7m 21s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/667c9082c85048349b04a6e09262275c : SUCCESS in 15m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f030afc51cd4c528c3e261338833ffd : SUCCESS in 6m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4f6eb0f1ebcc42da9c4b9f5a82b28a55 : SUCCESS in 8m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1c8ee9f480fc427289f3bd8072cde593 : SUCCESS in 12m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/44c1055165f1406c9cec05138d3a327e : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7d11070a89b34629b5e492106ea1c62a : SUCCESS in 8m 20s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/53abbea669ee4df1b1a9a115cfe860e8 : SUCCESS in 19m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1c228b3f800b4635b555a95c2a9e20e1 : SUCCESS in 7m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/898ce042ce84466aaeee77be170c5bf0 : SUCCESS in 10m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/eb833b9560cc417a939cbbdab53e0975 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0c87b65f3e7a4f8a9ee72c586365ac47 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7beb02ebb344f4f899fa0c2969731b3 : SUCCESS in 28m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/18e13d77bbb843be827d4c203d51e36a : SUCCESS in 5m 30s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2fee6ebab97a4935b94d464f13cad316 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c479198c955e476d9d43cbc1a183e7bb : SUCCESS in 5m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/66cd36320b02431189b894fde4d1d8b6 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d94d560dfd0f4202902920a09a3cf463 : SUCCESS in 47m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a653a097b8146fd8ebef38a0838fcb5 : SUCCESS in 1h 32m 01s","accounts_in_message":[],"_revision_number":5},{"id":"112732d826849f45aa93a5e3699375044a8e58fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 10:40:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"451a5bfc8bf71688b15dc789e95f6fb8fb1e72a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 10:40:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edfe0882d051432087a56283639b9917\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3ae0ff86854f93afac8b35d3ef98e6 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1ad6d7b8beef12198e7da5174e3a5fb2b90a72cc","revisions":{"07bc87683ab04a71702b9eefc563d359c0984c75":{"kind":"REWORK","_number":1,"created":"2025-03-14 14:49:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/944392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/944392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/944392/1"}}},"commit":{"parents":[{"commit":"dfbacbf502a016f97874176ec13c0a2516afab94","subject":"Remove tap-as-a-service from constaints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfbacbf502a016f97874176ec13c0a2516afab94"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 14:48:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 14:49:29.000000000","tz":540},"subject":"Re-add tap-as-a-service to project list","message":"Re-add tap-as-a-service to project list\n\nIt was added back to the neutron\u0027s governance.\n\nChange-Id: I265fb5d983d3eefadedfab8cff1421d6b8591479\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07bc87683ab04a71702b9eefc563d359c0984c75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07bc87683ab04a71702b9eefc563d359c0984c75"}]},"branch":"refs/heads/master"},"d0c908a7a0944e11bb252fd665abad57d2c0341f":{"kind":"REWORK","_number":2,"created":"2025-03-15 01:46:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/944392/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/944392/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/944392/2"}}},"commit":{"parents":[{"commit":"39f3fb28302cb384aac7113af4f2ebaadf385e9b","subject":"Merge \"update constraint for openstacksdk to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39f3fb28302cb384aac7113af4f2ebaadf385e9b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 14:43:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-15 01:46:13.000000000","tz":540},"subject":"Re-add tap-as-a-service to project list","message":"Re-add tap-as-a-service to project list\n\nIt was added back to the neutron\u0027s governance.\n\nAlso, it was added to upper-constraints because of networking-midonet.\nHowever networking-midonet was retired some years ago, so it can be\nremoved now.\n\nChange-Id: I265fb5d983d3eefadedfab8cff1421d6b8591479\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0c908a7a0944e11bb252fd665abad57d2c0341f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0c908a7a0944e11bb252fd665abad57d2c0341f"}]},"branch":"refs/heads/master"},"2f07b832390bb716d4ab765d7d4a73dc61a03247":{"kind":"REWORK","_number":3,"created":"2025-03-16 14:30:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/944392/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/944392/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/944392/3"}}},"commit":{"parents":[{"commit":"39f3fb28302cb384aac7113af4f2ebaadf385e9b","subject":"Merge \"update constraint for openstacksdk to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39f3fb28302cb384aac7113af4f2ebaadf385e9b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 14:43:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-16 14:30:40.000000000","tz":0},"subject":"Re-add tap-as-a-service to project list","message":"Re-add tap-as-a-service to project list\n\nIt was added back to the neutron\u0027s governance.\n\nAlso, it was added to upper-constraints because of networking-midonet.\nHowever networking-midonet was retired some years ago, so it can be\nremoved now.\n\nChange-Id: I265fb5d983d3eefadedfab8cff1421d6b8591479\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f07b832390bb716d4ab765d7d4a73dc61a03247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f07b832390bb716d4ab765d7d4a73dc61a03247"}]},"branch":"refs/heads/master"},"15e142d7fa0c28fc9d31b64168b3397b22e5bffb":{"kind":"REWORK","_number":4,"created":"2025-03-29 02:05:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/944392/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/944392/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/944392/4"}}},"commit":{"parents":[{"commit":"4a006431f60be70eea411a5c5f2e9119b8cdb567","subject":"Merge \"update constraint for horizon to new release 25.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a006431f60be70eea411a5c5f2e9119b8cdb567"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 14:43:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-29 02:03:14.000000000","tz":540},"subject":"Re-add tap-as-a-service to project list","message":"Re-add tap-as-a-service to project list\n\nIt was added back to the neutron\u0027s governance.\n\nAlso, it was added to upper-constraints because of networking-midonet.\nHowever networking-midonet was retired some years ago, so it can be\nremoved now.\n\nChange-Id: I265fb5d983d3eefadedfab8cff1421d6b8591479\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15e142d7fa0c28fc9d31b64168b3397b22e5bffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15e142d7fa0c28fc9d31b64168b3397b22e5bffb"}]},"branch":"refs/heads/master"},"1ad6d7b8beef12198e7da5174e3a5fb2b90a72cc":{"kind":"REWORK","_number":5,"created":"2025-04-03 07:56:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/944392/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/944392/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/944392/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/944392/5"}}},"commit":{"parents":[{"commit":"aea4bdb03846d4b08c0b3decf0ef6dec618a14ad","subject":"update constraint for os-traits to new release 3.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aea4bdb03846d4b08c0b3decf0ef6dec618a14ad"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 14:43:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-03 07:56:11.000000000","tz":540},"subject":"Re-add tap-as-a-service to project list","message":"Re-add tap-as-a-service to project list\n\nIt was added back to the neutron\u0027s governance.\n\nAlso, it was added to upper-constraints because of networking-midonet.\nHowever networking-midonet was retired some years ago, so it can be\nremoved now.\n\nChange-Id: I265fb5d983d3eefadedfab8cff1421d6b8591479\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ad6d7b8beef12198e7da5174e3a5fb2b90a72cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ad6d7b8beef12198e7da5174e3a5fb2b90a72cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
