)]}'
{"id":"openstack%2Frequirements~944752","triplet_id":"openstack%2Frequirements~master~I8f95eae13a67263940cb2c59085c531c76d887c0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8f95eae13a67263940cb2c59085c531c76d887c0","subject":"update constraint for ceilometer to new release 24.0.0.0rc1","status":"MERGED","created":"2025-03-17 11:08:09.000000000","updated":"2025-03-22 19:29:58.000000000","submitted":"2025-03-22 19:28:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944752-new-release","meta_rev_id":"6866def2d7a80fd529fb5979e97b5e36c324f01f","_number":944752,"virtual_id_number":944752,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-22 19:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2025-03-22 17:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2025-03-21 08:11:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2025-03-22 17:49:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-17 12:59:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-21 08:11:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-03-22 17:49:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a791683dd4109e7e55a6aa668a56a20754d2a221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-03-17 11:08:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2447e47a2e5bbd8e3844cc660283144a90c124ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 12:59:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c432b6256f274f89939fa852b16bfd17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7b119e41a6a47dbbb5b9ea1c0c34c38 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d9e8fcfda94f9d8c7ba55a326cc5b2 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e517ddf311b2415487905ade2a67bef5 : SUCCESS in 6m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7ea5f57cad0e4640a3fb0275064e5826 : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d724cf7cc4ff451daabc79f899ee5e6d : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ced3919091b4765bd5dba245688959b : SUCCESS in 3m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9cbadfd79e74476c8a1b9d31c7539056 : SUCCESS in 18m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8ca7c287b584467a9732b7346bc5386b : SUCCESS in 9m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e48e52cc1e484f9db3eec62faa262273 : SUCCESS in 19m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8458790cb58545d5b3a9b1112c014d2a : SUCCESS in 28m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e059a4f274ca474ea8aaf3126f2ad813 : SUCCESS in 5m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/83c95a874db747f09a22754b564a6e9d : SUCCESS in 5m 15s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3fd3e49b5bad4e17af049c2238149824 : SUCCESS in 5m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/17306a6e3eab49c7aabcb92c9ea2dcc8 : SUCCESS in 11m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/288af848c937460d87ec44440c9e7b87 : SUCCESS in 5m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/47c06a33538e4879b2ce886a99e8c8bc : SUCCESS in 5m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1659c7da4d5f4c8f957213cd28997916 : SUCCESS in 8m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c85a04ef50764085a6829729bc05fa5f : SUCCESS in 13m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ee5247b2b074f6eab94f7740cceba01 : SUCCESS in 5m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e0e9d197803b481799b5049028980833 : SUCCESS in 4m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/16f25de3de17491c8ee0fc493942d80b : SUCCESS in 13m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/108d71523ccf42dcb9b509ec9e7bc964 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/aa17f2b9447a4f61889f2bd913ae6323 : SUCCESS in 6m 40s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/38a5f4a623674459b5dbb86ebfa160f3 : SUCCESS in 4m 57s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/faf806e7cd6f4ab7bdd1b8828ef122a1 : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/dcc50283db074f2e85b7be21a07956a5 : SUCCESS in 10m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/230fc1ca6c8d45bc884463a299bf01f4 : SUCCESS in 17m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/74cd9996d35c449ba514eb58fa0ac4a2 : SUCCESS in 10m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/807cb66eed0d4c4cba506335a8ae49bc : SUCCESS in 8m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4f795e33a12442e083092e6414fac8ef : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5aa07f9ca62747a19387ffaeed0958ee : SUCCESS in 6m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/292d281d638f436b9b3dccdbf57b46fa : SUCCESS in 31m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9a31a0f0dc684e0faa5df5149ae8827e : SUCCESS in 6m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b9b5b2e9ba404bf9a5babcdfe0cdaa43 : SUCCESS in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83a6f5ce30f249e8b74ffe8020d7f9fe : SUCCESS in 3m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2f68990166ed45d7911fc857b61ddf1a : SUCCESS in 16m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c05962b954be4056a431b3d6d230a89b : SUCCESS in 1h 09m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89f47e4a2c0640c0bbe055bedbf5041c : SUCCESS in 1h 45m 16s","accounts_in_message":[],"_revision_number":1},{"id":"40c0f92526b45aed9c9e016dfca50a4c4c42629b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-21 08:11:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3aead83db2363a13b42cee9809c501e54b54db9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2025-03-22 17:49:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56474de6f6b7a4b1f08c60bd4b0afd5de5f93557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-22 17:49:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02a74fbaa8939ebe6cf2192f5ca26404e44da0d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-22 19:28:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b519fc504b9147d28c0633b675b5dcd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d9c978afddb434998dd8bc9f8d79f37 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcacc0a166e542c9affb2c9d071ae02c : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00f06f2ca2c14083b49e6ac665ba643e : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19382764f9c640c5a3e2acb4bd2c5459 : SUCCESS in 2m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32abb1040c6549e8860105af81ac85c8 : SUCCESS in 2m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7cbd7caba7214d06baddcaeabe65aa29 : SUCCESS in 2m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a60ebd392e29452fb360f0fc0cbc2495 : SUCCESS in 6m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0d38a40775984cb7b31660402cdca49c : SUCCESS in 8m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7f575deded9d460fa8f55c5bb5ec7e78 : SUCCESS in 15m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/24064da2a35f4bb988f6c62558f9d8a9 : SUCCESS in 16m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1a99668a6852472383360416a98141a0 : SUCCESS in 11m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9ab121ed18984a1c9e62b48d1f3a73a8 : SUCCESS in 6m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ea636ed30bcc4762b3b0d23409c8be46 : SUCCESS in 5m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f44693db4e0f4a71abf3dfae346586a8 : SUCCESS in 7m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a31b3367841a4911956abb1873df0765 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/254802146b5c43bdabcb5bc6319605f9 : SUCCESS in 6m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c8e2510afe2e446b840e7a04925cd7f0 : SUCCESS in 6m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bd06dde9b7c148dda2068f26382e7995 : SUCCESS in 13m 28s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4f6c567764154eb0b2c6acd3a314a146 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/90503527b3ca470ca15628a77ef0cf55 : SUCCESS in 11m 30s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c5c918d385634ff986a19623c1b8ef45 : SUCCESS in 17m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/48dfbe5383024807b3dec642ff786c08 : SUCCESS in 7m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/82937f3f525947edbb23308006f3941e : SUCCESS in 10m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/561d75ad81134b318d695e9a27b182eb : SUCCESS in 6m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d945fb46f3c64d3482a387a5486a929a : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb678fc094be4ffe9b7356b20bd58982 : SUCCESS in 23m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ae5a378255d84abc974b0ec7127388a0 : SUCCESS in 6m 52s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/622f4ce8194b45e1bdfba2114f737999 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46648aab3baf40b8adcf21010c8f1e1a : SUCCESS in 6m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/18179d9e92114316a2e76b0d07575cef : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f2dbaed41d746448be600fadca0efb9 : SUCCESS in 50m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23edf32a31c74842a7f56b846c65628e : SUCCESS in 1h 28m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8deaeb7e8ea8e7eb65a9ab7d4b9597b9ea5b3aeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-22 19:28:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6866def2d7a80fd529fb5979e97b5e36c324f01f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-22 19:29:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9ea7d215c9f476c9f64a2f5d64d86b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a36d50f9ae94a06915848bc7cf6df8f : SUCCESS in 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"926dfd07b589ea5f9efc433f1d0b8bace8dd1579","revisions":{"926dfd07b589ea5f9efc433f1d0b8bace8dd1579":{"kind":"REWORK","_number":1,"created":"2025-03-17 11:08:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/944752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/944752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/944752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/944752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/944752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/944752/1"}}},"commit":{"parents":[{"commit":"7a304c74f76991597bfe7eb0666b2b2a2bf2ffa3","subject":"update constraint for taskflow to new release 5.12.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a304c74f76991597bfe7eb0666b2b2a2bf2ffa3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-17 11:08:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-17 11:08:06.000000000","tz":0},"subject":"update constraint for ceilometer to new release 24.0.0.0rc1","message":"update constraint for ceilometer to new release 24.0.0.0rc1\n\nmeta: version: 24.0.0.0rc1\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release candidate\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I0d6352808cbc271d0ee2807b4f8cb549f16fa2a7\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8f95eae13a67263940cb2c59085c531c76d887c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/926dfd07b589ea5f9efc433f1d0b8bace8dd1579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/926dfd07b589ea5f9efc433f1d0b8bace8dd1579"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
