)]}'
{"id":"openstack%2Frequirements~945272","triplet_id":"openstack%2Frequirements~stable%2F2023.2~I8214a2b0b27f24f8cc802fec068b1c34949d59c3","project":"openstack/requirements","branch":"stable/2023.2","topic":"new-release","hashtags":[],"change_id":"I8214a2b0b27f24f8cc802fec068b1c34949d59c3","subject":"update constraint for python-vitrageclient to new release 4.8.1","status":"MERGED","created":"2025-03-22 16:30:07.000000000","updated":"2025-04-29 15:52:25.000000000","submitted":"2025-04-29 15:52:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945272-new-release","meta_rev_id":"3c4e9e49cbefee08312a0dd21240b4d92e006dfd","_number":945272,"virtual_id_number":945272,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-29 15:52:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-04-29 12:42:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-04-29 12:42:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-22 17:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-28 16:38:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-04-29 12:42:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"441d157c06911187e3107b08722e44bec56f62f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-03-22 16:30:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43a74fea0168ffd34b9901320ee33f74915553a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-22 17:23:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e4caa4b8bfc49e4ab960bee47d2c9c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c21122e84645eb8da3e808a0b1c1ac : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/062c4a078e244b6da8083673715e7128 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/270641a74e60469eb1148f493325c4da : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5446f4e9e91406d999494b7f3f9914b : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a2173c956d0444099d0810bf485e702 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c01086d6d3b5464884f5e66df32ac922 : SUCCESS in 1m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a5447fc21a34591806dc235f1e01b45 : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/27332daf0930492b99453098c7b5a503 : SUCCESS in 8m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2b33b85f5ac74c048720512572e5d93d : SUCCESS in 9m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/652543c915b045d0a48c576f556f68b2 : SUCCESS in 8m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/15b9d883238541e59a1bdcd238e5063f : SUCCESS in 8m 47s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/31de011c6d1a4ee2afa7ac4cdf1a9d44 : SUCCESS in 6m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/29b0616b854249569ce5655acca487ea : SUCCESS in 3m 29s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c17d197d1f8c4065bb421dc52784d9b5 : SUCCESS in 4m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7bb6a03158a94eac8324b8b4661b825e : SUCCESS in 8m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8524e061864441cc8a4b582a44d86750 : SUCCESS in 5m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/aaa774df6baf49bea479e9aeee8e574a : SUCCESS in 6m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9a7c1b800d5f4e99b526b679d84929ac : SUCCESS in 7m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/39e66e8be1234142a97471b988b99147 : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4b185bf77494ceeba921565e2219e21 : SUCCESS in 6m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/64a07bf5b7e74f45aa230630576be192 : SUCCESS in 8m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c5795a8c5a224a549495898d8d582cfd : SUCCESS in 12m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b2d11652e936413d874806d6834dc90f : SUCCESS in 4m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ac328ef5283b4b858ae1eb5c02522516 : SUCCESS in 8m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/41963b85b595413db43a0dbfa9e4eb52 : SUCCESS in 5m 29s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/07793cc42bf54c7591c453f78acaac51 : SUCCESS in 9m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/449c1f89661f460f9e868b6787c44e76 : SUCCESS in 29m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/49d43a0133d3415b837a3670fcec208a : SUCCESS in 11m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/59fbef9dee2c4a65b5531c1cc179e531 : SUCCESS in 12m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/463b9b2c4e994640b8e0abcb664588cd : SUCCESS in 3m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a46d51e9059041fabbafed9c8f4c4cb3 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ec35f025a164a20b303ab9cd641a08b : SUCCESS in 20m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/673d20477486488081041cf0663c564a : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/76ae6d91e782489b922df00a7db51564 : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82810e0cef784f62b9add7e6182ad6bb : SUCCESS in 5m 19s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ef1f1bd8412445db8b4b472ccdae0b18 : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84f5c10ce6f04731a4965887c541feaf : SUCCESS in 38m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/144502d6593c45bd9b570e3cadc7b89b : FAILURE in 52m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7ce5f3919633a92de4a6a02bc6e9c35bdd13d70e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-28 16:38:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d588a14c71dc7c638858af9e9e657f2a9a360c1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 17:49:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/acb1f899f8c346c6918a5f45799e4a15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f10b66c56a49b5a3dd18630133cbdf : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e2c4da6be894216860c7eeef1fd2370 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7526aa90c3ac47bfbfccf51ba00440a8 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c80076e5b57f4945b9c9be9727251d19 : SUCCESS in 1m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/580e4f36edbf49d1bb121cc6dc2a00b6 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ef2de5ae4d44ab4a507415f36d33148 : SUCCESS in 2m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b800e4c5c4f548768ae2077801677484 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f0023e29ed3f468793a2c14474abf43f : SUCCESS in 10m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b1b9b05b74634746a26ca6b3f5805f87 : SUCCESS in 12m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f36de3db090a4b748ca6bc7bb51105b4 : SUCCESS in 10m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/31a005a5c0c7433cb1244ab86aabe6c6 : SUCCESS in 10m 40s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8373f170ae424372b0298a360137f1f0 : SUCCESS in 8m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/12579692512143fdaa679acfb5b834bf : SUCCESS in 4m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2f113bb806d1414e9d4ee6e94d1a631b : SUCCESS in 4m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/54d0b07490c14184a002ca8067b9b1c0 : SUCCESS in 14m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3c33c832f2ec442c88d0474d53358d4e : SUCCESS in 4m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/641c29dfa81f41abad404f0c6e8e9f51 : SUCCESS in 5m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4af6db1719b748b981a0b8ffca0431a0 : SUCCESS in 7m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/afb79a73be87436aa99177d8a71dc6b8 : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/017c9a3d930246e7b0848bb2636f9598 : SUCCESS in 7m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/718b729dd1f940348e1c2c90e8b4bc2a : SUCCESS in 8m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/98df064b1f314a509b4a1a118c95205e : SUCCESS in 13m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6bfb969894624cc8bd8a7b414519a02a : SUCCESS in 4m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a07d1ca94c64493681f3ae76aeca65b1 : SUCCESS in 9m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4b04e9be1b1047a3a7ae8c5bf2443856 : SUCCESS in 5m 10s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3641c9306c1c4e5ca4748e1d00060e56 : SUCCESS in 12m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/59a79ab90bb94bfc8abd7620ea85d728 : SUCCESS in 25m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ab70984e6e6b4c9b96721ce2235373f6 : SUCCESS in 8m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2d319c00ec8c48a4939bfb94cffa7e45 : SUCCESS in 13m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a9fb1a9d4378413fbe335def5a2d84f1 : SUCCESS in 3m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6838ded6f91c4fa5bf41bfc32bca4ce6 : SUCCESS in 2m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/64353ce2f562465894b40ae7ad986afb : SUCCESS in 24m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/42e6235d431f47d184e96bf168920938 : SUCCESS in 5m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0d832d6ff25c4c27bae2e7703db12718 : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d4b728e234ca4aef817e7ab5c6525d48 : SUCCESS in 5m 33s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b4b8f1dc1a0d4589bdde8b4bcbc956d4 : SUCCESS in 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/844ba6fd60024ca887935a921a50b023 : SUCCESS in 56m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b0650a43b5f48249740cda3bb892612 : FAILURE in 1h 01m 46s","accounts_in_message":[],"_revision_number":1},{"id":"51b93bc67179bbb9be6d46bf83a8f40a1793c193","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-29 12:42:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba09cc331ba86c2a79affb82f1fa1c1e9841df09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:21:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c08fafbab094ef7933c5d38c9d6d906\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df7be8106094d618d1aa71490287116 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e016855196d4d2c9d6668b228ff047b : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d33203dad3e144b4b46fee6d66f4da7b : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f9b8bb58db44398ad3150abaccd8604 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0d999badea8461094e3bf42170002b2 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61621ccdebed4ef980942dcaf333fd7f : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/11c2ef687f644bdaa4f26c33dded275d : SUCCESS in 2m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3f1fb45b526c48c0888d3dec90575846 : SUCCESS in 3m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/aefa196fe24d47b38377e54996b799e8 : SUCCESS in 11m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d04f7d43fa7543ed8fb3e648d0bc80e6 : SUCCESS in 3m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/032b73fcdf984c1a9a9d1bc97618f30a : SUCCESS in 10m 15s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/281b7dce299849fca1f5f46a97a9ae0f : SUCCESS in 6m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/30f31b95c09541db9763d2ace781921f : SUCCESS in 3m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/238324ebb0a746ceb23fd3d543d97c04 : SUCCESS in 4m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8105d84952b24c699ae33df0f0595739 : SUCCESS in 10m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/14098b0ece574a26929466c99bd068d7 : SUCCESS in 4m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bc6a68e0342b480ba80d88e492999ed4 : SUCCESS in 4m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4bd90e4e2c91414998924552ff944005 : SUCCESS in 7m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/989afd52fa484382a9ed34e284f02948 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56c222a22d5649a1920e391ef04fa72c : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4a850440c3204f54a530f9b1d2f7b7f0 : SUCCESS in 7m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b2b30363d5b74b4490eb336e95d7bb48 : SUCCESS in 10m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2aaebec4f2904b79b9ecfa03ae5b4675 : SUCCESS in 3m 14s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6cc10a846c3e40b086490993e5bb835f : SUCCESS in 6m 53s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3235dda864764f4ea85e26195e4aebbf : SUCCESS in 3m 52s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/edeffa01353d454aa7d780d0cca1b267 : SUCCESS in 11m 47s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/64c25ea1942d4aea8b573a1046323ae6 : SUCCESS in 25m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1de5ef21466a49848881f136a28537d0 : SUCCESS in 10m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9c2e634b2180422db8e3af91bc18a026 : SUCCESS in 13m 14s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c97d2a40723c42248e17c6901bf8c0cf : SUCCESS in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/56e8ec6a502b4efd8929c63289d545bd : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1ab8cecc7454ff48afb00dc967d00be : SUCCESS in 23m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a71c5a885ae243b6ab349039ac45cde9 : SUCCESS in 5m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/77efc7e059514c2493ceabe38934a559 : SUCCESS in 4m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a680975d77349e19ad59e80a8ad7fc5 : SUCCESS in 6m 29s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/04ca7a21ea394f96bb52bdf6414a5667 : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ef54aa4bbfd4d21925627799be6d487 : SUCCESS in 45m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61d202366fd945aa99708c6cbeb390d6 : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":1},{"id":"de9698dda046ec6cfa26cc49fc12eeeecc9fb326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:22:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cbc46cbe72d84e06f35c598d75736352538846f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 15:52:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfdf0a602ce54a9d873f59ade48bf8c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a16a89d2532e4b1084beb5fd929e1303 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/076825d7deb844ca9711ec1879c7c32a : SUCCESS in 2m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/252e32cf98234b8d9251bc9cdbb50808 : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/429af1f5cc3341e1ba8e9b685e81a580 : SUCCESS in 1m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57d00cddba994971a2135176bc53c8d0 : SUCCESS in 2m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/544390b3d89b4c47b31db8eef27c0023 : SUCCESS in 2m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c031266e43984487ad0e8e70652d0216 : SUCCESS in 3m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/144200eaf6c2414b9d41e11edd96749e : SUCCESS in 4m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c71b25ad1fb54c9f906287b1b4a48b7c : SUCCESS in 6m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4c38a5b510694f4b82469ddbaa2a0f14 : SUCCESS in 10m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/88126f627fd0435c8c37035272b6adaa : SUCCESS in 4m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2d83bc5c27bb429f91a52f6bfb6eaebf : SUCCESS in 4m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0691c3b78cb3419a8e8e0fa5c43ccd97 : SUCCESS in 7m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/21d18805361a4889b0d0104e918998c8 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09f550c416654cc2b32d51741f8497b4 : SUCCESS in 6m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/20e18b0be9034b6081ff2caf957ba719 : SUCCESS in 9m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/04c9a1ea133a4d5c8ef96d390a260dee : SUCCESS in 11m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e58a8b7cf57d425fae0c2cb16433c8fa : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b93952b8298f42d2a14043a462d3f6a1 : SUCCESS in 7m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/39dd930ece0447d28ffb15805595d376 : SUCCESS in 27m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a528f50d37894a5288d0a3228a4e9bfa : SUCCESS in 10m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c92c2a4a168f406e942388717b0d6b12 : SUCCESS in 15m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2d22888fb3444d45a11494e89cec7df8 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/15a8d9cdc7b3405cab5087e560be6a06 : SUCCESS in 2m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f5c5d1237a043bc919d10da4dc9c42a : SUCCESS in 26m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8c0b22e2c1594663a457bc724673c0ac : SUCCESS in 5m 32s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/53e1ef9f3e404ff888b738d2b7fe9967 : SUCCESS in 3m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f00e03652601410abd7be8711f86495e : SUCCESS in 4m 40s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/13354fdf73d744f8a8c3a14f304230d2 : SUCCESS in 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fdef5903c8846c69c735c6835626f2d : SUCCESS in 54m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79b1042e362a4e8790d90a568980b9db : SUCCESS in 1h 24m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3c4e9e49cbefee08312a0dd21240b4d92e006dfd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 15:52:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84f41f5ed9c339a9f1353a273cfd8bf441a5f8d4","revisions":{"84f41f5ed9c339a9f1353a273cfd8bf441a5f8d4":{"kind":"REWORK","_number":1,"created":"2025-03-22 16:30:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/945272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/945272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/945272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/945272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/945272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/945272/1"}}},"commit":{"parents":[{"commit":"9042ccffadf3de68b5e2e0ee76e3d085842fd161","subject":"Merge \"update constraint for oslo.db to new release 14.1.1\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9042ccffadf3de68b5e2e0ee76e3d085842fd161"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-22 16:30:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-22 16:30:05.000000000","tz":0},"subject":"update constraint for python-vitrageclient to new release 4.8.1","message":"update constraint for python-vitrageclient to new release 4.8.1\n\nmeta: version: 4.8.1\nmeta: team: vitrage\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: stable/2023.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I980557a77b4dc34fc7d1d64d61667cd442db375a\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\nChange-Id: I8214a2b0b27f24f8cc802fec068b1c34949d59c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84f41f5ed9c339a9f1353a273cfd8bf441a5f8d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84f41f5ed9c339a9f1353a273cfd8bf441a5f8d4"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
