)]}'
{"id":"openstack%2Frequirements~945488","triplet_id":"openstack%2Frequirements~stable%2F2025.1~I32bdbbed486b3f362f04e8678dfd725532dba0b5","project":"openstack/requirements","branch":"stable/2025.1","topic":"create-2025.1","hashtags":[],"change_id":"I32bdbbed486b3f362f04e8678dfd725532dba0b5","subject":"Update .gitreview for stable/2025.1","status":"MERGED","created":"2025-03-25 16:46:48.000000000","updated":"2025-03-28 19:25:35.000000000","submitted":"2025-03-28 19:25:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945488-create-2025.1","meta_rev_id":"27f419e14913af0e29c4afa6240da9468744361f","_number":945488,"virtual_id_number":945488,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 19:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-28 16:39:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-28 16:39:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-25 18:28:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-28 16:39:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3ef702cfd92b5026140c2decf78f07de94bf3267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-03-25 16:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96014e586ecba642d0d2509d64dc1cddcdafed73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 18:28:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7b9446ce3a7423aa668e197f1c4567c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/578d175c881e4bdcb24335e41925b203 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0731de4bbabc4b64bbc5e03ae3208825 : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5a27dc696fc4b4d86e5ef1e4a61b6b1 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07470ef9b69d40e5b43b3079f85340a8 : SUCCESS in 3m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/095ad2e5725949aa91d86e1fa3c98b36 : SUCCESS in 54m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d75ca9e2d1c346db83966197cbc19dda : SUCCESS in 1h 30m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e18607b0e713523e281a1837030f7ff1b8c5f32f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-28 16:39:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd1fc4c5acf034f72df7e0b3419fc019f45cd52d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 16:40:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82669ce5f8d98ab431c0d7a87f693ffdd61387af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 19:25:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f12d8bbb40c94594b830bc2cba55211a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56b8852473d946daa455c8bd6b33b2a4 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8296b18272540738c4900e17ff3741e : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad91b423ce284a23a64a17668ef07591 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef090cdef62e47a3bb1dd9ccbc28c472 : SUCCESS in 3m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5def8c17a1f4c1f84bf9a34837ce843 : SUCCESS in 54m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/770514545d8a4e3d992b9b5de01b567d : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"27f419e14913af0e29c4afa6240da9468744361f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 19:25:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb2e6f8107b43b7ef96bdcc1e5a0a2ab1004f7d5","revisions":{"eb2e6f8107b43b7ef96bdcc1e5a0a2ab1004f7d5":{"kind":"REWORK","_number":1,"created":"2025-03-25 16:46:48.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/88/945488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/945488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/945488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/945488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/945488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/945488/1"}}},"commit":{"parents":[{"commit":"4a006431f60be70eea411a5c5f2e9119b8cdb567","subject":"Merge \"update constraint for horizon to new release 25.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a006431f60be70eea411a5c5f2e9119b8cdb567"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-25 16:46:47.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-25 16:46:47.000000000","tz":0},"subject":"Update .gitreview for stable/2025.1","message":"Update .gitreview for stable/2025.1\n\nChange-Id: I32bdbbed486b3f362f04e8678dfd725532dba0b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb2e6f8107b43b7ef96bdcc1e5a0a2ab1004f7d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb2e6f8107b43b7ef96bdcc1e5a0a2ab1004f7d5"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
