)]}'
{"id":"openstack%2Frequirements~945574","triplet_id":"openstack%2Frequirements~stable%2F2024.2~Idce121b7795947dac41266abf0ee83c5178c7dbf","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"Idce121b7795947dac41266abf0ee83c5178c7dbf","subject":"update constraint for keystoneauth1 to new release 5.8.1","status":"MERGED","created":"2025-03-26 12:15:34.000000000","updated":"2025-05-18 08:55:43.000000000","submitted":"2025-05-18 08:55:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945574-new-release","meta_rev_id":"af96baf46020778131c64266075a3943dc2da80d","_number":945574,"virtual_id_number":945574,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-18 08:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-18 07:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-18 07:29:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 13:41:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-18 07:29:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"faa720a174767d927684b5e7dfe785c1c287a94e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-03-26 12:15:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"725ed8cb2ba957249f985aa32d6f4283859bcf0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 13:41:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5d4e3a4ce5645a1896c92eded544805\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d0acc2e4d864449b87247222c44fd50 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22e364f6c0114ee59c96edd75ef23ab4 : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/517979d664f042e5b5b583b77ea79de3 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e83ebe61b014ef6a2109119ab5772a5 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3d44d98506244ae8a0974e07df882cf : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5144973001e4111b7eff8bb20acb39b : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40fafb38776949bd80af9be166c32894 : SUCCESS in 9m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d460e121b0fa404f9204a6b40491e3dd : SUCCESS in 19m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3d0841f2ed784e73b1e9b9362f80b4d3 : SUCCESS in 8m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f3937aaa3e2d400abb4b2c6791f06714 : SUCCESS in 17m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1d6a766bb919499db53a665a2b4d5db1 : SUCCESS in 30m 58s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/eaed03c2044c48e695e80711d2f031bb : SUCCESS in 4m 49s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/9a22ff1d5c6846d98ff51643ef2e7bd4 : SUCCESS in 5m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/a6f6e721a8fe4727b46cd4dfe0f14e2a : SUCCESS in 4m 23s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2b86d0b6b7ac4effb712ef765ed5d440 : SUCCESS in 11m 12s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2b65b1b4c8104adf8fd593e52255c350 : SUCCESS in 4m 16s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e33104826ddf4950ad8b7cbf9ac3d899 : SUCCESS in 3m 08s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/0c537dfc908f4f37864a0433bb62bac0 : SUCCESS in 7m 44s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b5d61d6a4ea34e4c9b57183ed43f2f29 : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dfd2251fc78430e902d0f5ef797dba7 : SUCCESS in 7m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/30e669a9782c403ba001ea4ca382bef0 : SUCCESS in 5m 24s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/43eadae0279e4c238331d1fd72d661e8 : SUCCESS in 8m 48s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/eefd4898bc8b4629904c2c4d62a12227 : SUCCESS in 2m 16s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/42bf080db153454da1f84558713f9438 : SUCCESS in 7m 31s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/c9718ec0555c4d5ea0ae34999440c8ce : SUCCESS in 3m 03s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/9571fa0d75c7402c8acb4c999c9fd749 : SUCCESS in 3m 56s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/e4239487089e44c2a089c1a47efb4b68 : SUCCESS in 10m 11s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2abee454fa5042859bea60cf5f07e7e4 : SUCCESS in 14m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/41e91a8db6f94ba1bf88e50e2c9d1a54 : SUCCESS in 4m 15s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/61c67218178d47e2be08bd665625fdde : SUCCESS in 12m 15s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/cd0195e06f7046c6b8af2f85e6a7daee : SUCCESS in 3m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa5ee0abd1ba46cc9d44bb0e499bf525 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ded9b510cfb84285befcccc7a0a951a0 : SUCCESS in 25m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5f38270e9cf6492badd95c07c4fb67f4 : SUCCESS in 4m 20s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ae6a21b34a754341b54cae7c84bf64ce : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e20b5cdae7bc482ebf7e1f796829b592 : SUCCESS in 5m 49s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ae3c183bdd2c49f1bd94555175c48fc6 : SUCCESS in 17m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e85a80d008404cebb3b9527d14f5ec21 : SUCCESS in 59m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eac39021a3d64a98b589ce5e9c286d14 : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b18e7a5765da5ba17f698d5c8f7e77019d338e66","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-18 07:29:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95ee00b4918355bd2eb311ff7984519b316cb32f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 07:30:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4496712ecc7ed453a7cf6ff4cbbb5415b186d6e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 08:55:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bb7c437884446e9a993fab665459015\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b360c51e7e5f4092b2efe90696ea1d6f : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a293b1f7604945c0991f369af4b74e0a : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd1f30c732454a55ac77832c2cf82e4a : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/639a9518f5dd46c3be7b62963d1ed5c4 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1afbf9ce345474a855facf4f01f679a : SUCCESS in 2m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/555f5e4a9e3347298ae63ce492abee9e : SUCCESS in 7m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c8bd99debaae416e86b3a806d33ff4ac : SUCCESS in 15m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e6d2dd5f21704f36bc463774b94a951a : SUCCESS in 6m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/621f9a05cd5b4cc2861b854775245894 : SUCCESS in 17m 52s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/99ff1a34274b484cbab6c9ee8b2f73a3 : SUCCESS in 11m 00s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b8965c00d289428090b000622a1a4f4b : SUCCESS in 4m 06s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/800c22a3030748cfb9ae12f19b67f9a1 : SUCCESS in 3m 52s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1a5d79d4b65c4f748febd47d1c15bd48 : SUCCESS in 3m 57s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/f61f5928acf240deb317c945c14fa9e5 : SUCCESS in 13m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/06f5c90bb8d64fbba7483f164dee7d1f : SUCCESS in 7m 55s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/548b2c5b76df4baeb4d0c76936099ae6 : SUCCESS in 6m 01s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0f7bd829b31b4044b80083a52e09de6f : SUCCESS in 8m 30s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2c9c5c37ad5f4ab781cc1da7deb63cb8 : SUCCESS in 2m 17s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/76b1455e817f4dc095f699d2f37314e2 : SUCCESS in 5m 47s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/91fa7192ecca43c195dcba5f50f946e1 : SUCCESS in 20m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/02a5d61668e04c278f331620a04a1537 : SUCCESS in 5m 54s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d31e9a2d3b464c6698e75837f3c8ae60 : SUCCESS in 9m 06s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c848c9ea6c5e487b8d24fdd3a18d8754 : SUCCESS in 2m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/937d994379d84b2191f338fcd1c4e3ea : SUCCESS in 2m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/28cab41f128f465abdf329cb8e334e23 : SUCCESS in 23m 44s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a21b3e63a28c4280b1f7476668c4a6ca : SUCCESS in 4m 23s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/053080e196b54703bf319b2be2324277 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/806d54f7a55c4e03b9ce56e3c2b7fd34 : SUCCESS in 5m 33s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2c9f0ed6cb874498a4e6d52d2ddd9a37 : SUCCESS in 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18dfb411a84c4e6d8f704e4dad5c9b5f : SUCCESS in 1h 03m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f402d29a6df745fb9f414116d6c94de8 : SUCCESS in 1h 20m 30s","accounts_in_message":[],"_revision_number":1},{"id":"af96baf46020778131c64266075a3943dc2da80d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 08:55:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5809fa7c60edae7e9b9c99c75122ff6e6dd18bba","revisions":{"5809fa7c60edae7e9b9c99c75122ff6e6dd18bba":{"kind":"REWORK","_number":1,"created":"2025-03-26 12:15:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/945574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/945574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/945574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/945574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/945574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/945574/1"}}},"commit":{"parents":[{"commit":"d620ff557c22c4b8c1398b6dc84772c341fb4d5a","subject":"update constraint for python-openstackclient to new release 7.1.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d620ff557c22c4b8c1398b6dc84772c341fb4d5a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-26 12:15:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-03-26 12:15:31.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 5.8.1","message":"update constraint for keystoneauth1 to new release 5.8.1\n\nmeta: version: 5.8.1\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Felipe Reyes \u003cfelipe.reyes@canonical.com\u003e\nmeta: release:Commit: Felipe Reyes \u003cfelipe.reyes@canonical.com\u003e\nmeta: release:Change-Id: I7df462db54b00a6bb550d3bdffba6d4894a5b815\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: David Wilde \u003cdwilde@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Idce121b7795947dac41266abf0ee83c5178c7dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5809fa7c60edae7e9b9c99c75122ff6e6dd18bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5809fa7c60edae7e9b9c99c75122ff6e6dd18bba"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
