)]}'
{"id":"openstack%2Frequirements~946499","triplet_id":"openstack%2Frequirements~master~I702394b5f925d6a3e1833109a82f9827cdc3088c","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2025-10-14 01:21:47.000000000","reason":"Change was submitted"},"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-09-11 13:22:26.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-10-13 23:17:15.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-14 01:21:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I702394b5f925d6a3e1833109a82f9827cdc3088c","subject":"Remove ldap3","status":"MERGED","created":"2025-04-05 16:04:34.000000000","updated":"2025-10-14 01:22:33.000000000","submitted":"2025-10-14 01:21:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946499","meta_rev_id":"37affc1e10a5b00518059592f86413dca24d2066","_number":946499,"virtual_id_number":946499,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 01:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-09-11 13:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-13 23:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2025-08-08 11:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-10-13 23:17:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-05 16:36:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-04-05 17:38:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-31 14:35:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2025-08-08 11:14:02.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-08-29 08:28:13.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"0d3e0c1612250a5498785dc1de135a87e326efc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-05 16:04:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60335fb913a19d68fb82139a41831e1fb6f9baa6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-04-05 16:36:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b6b355416c8db860bf05f8cd329a1e3dec5f80e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 17:38:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3218a27230746f3afd6344afd989327\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c7b242e7ed474fb02865c72491cdbc : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7674d8cfdd3b45e998699815bcd83802 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d422d4e66b04423a8bb05e4e77c8051f : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6045e0bbb1045d39671ca5044d93588 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6d074e11fed4792849a63647e456ee9 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4fd49cc2cfea4d1db01293de71936789 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/358ddf3433284b18a71a09c93a7171f4 : SUCCESS in 4m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/25c1f329eba344959744a5d69b2b7383 : SUCCESS in 12m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7a70f1625aec429d89630686324a0f3a : SUCCESS in 7m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b4f5733869c748308d121f27138cdc2c : SUCCESS in 17m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fb9113af668e4207aed4ffa0e969aada : SUCCESS in 15m 51s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3e8d3d7f875d4cc5a79fa45d71b51708 : FAILURE in 6m 36s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ff95bd05f6e34c6db937f3386bf92dcc : SUCCESS in 7m 17s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6dbf92548a7046c897917d8138a4be46 : SUCCESS in 3m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fadd9ebc028a46c0a412ca49013a7cb2 : SUCCESS in 9m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3b98b0613aa9402eb75b5719cf585d23 : SUCCESS in 3m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5635197999ff4f259271669d45ca60ac : SUCCESS in 3m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/300b41a3f9484c39b0ef06f9647aee24 : SUCCESS in 5m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f4ea457ade6e4ffa92ece8e3ee6d6d05 : SUCCESS in 7m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/551d2a3c2c724f11a712439028ebafa3 : SUCCESS in 5m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/eaac596bd9094e01bc82f5cf553e1482 : SUCCESS in 6m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1d017cc52584462a9e9fd703c9ed4c7d : SUCCESS in 7m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c2fb00c2c6884bdfaaf738d83edacedd : SUCCESS in 1m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/77c1a9cbd8234caf92775823e240ab45 : SUCCESS in 5m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/95d351c2dd3d400f851ec7f07333fe26 : SUCCESS in 2m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/556d760df38f45f1af52384b25a4799a : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/429982128be3453091d0487b8cd2c134 : SUCCESS in 10m 11s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5b969d307b1e4116abeda4365ab8f256 : SUCCESS in 25m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d26b7c5ce5b8427eb317c602bfa7eee0 : SUCCESS in 3m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3e5a8cf4fc7c4c1aa43ec2f5858a745a : SUCCESS in 11m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8b8e8151d4264b669fe8a50cb717f6c8 : SUCCESS in 6m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e260db4417624cf7a9e21c7b1058e2e9 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dcefe19737b14e3eb73fa12ced484610 : SUCCESS in 25m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/01c6e43c232449b389502d1af57bd3fb : SUCCESS in 5m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/63af38cca9a241bfa39670d0e3141786 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f01afe81d0d04a6cb4a78f03c053f36b : SUCCESS in 4m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d26dbd37f4d74fd2aa4d83026bb12d04 : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70401441c6054e04b0aaf784fd6d7b86 : SUCCESS in 46m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2c6f8aba8e64e7b850f840dbb715a5c : SUCCESS in 1h 30m 47s","accounts_in_message":[],"_revision_number":1},{"id":"72ee3555e16d1952fef30447a3c7ba6002523051","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-06 15:25:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f386a8d078fbaf330ff6828d942346313a39cd59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-06 16:25:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f29f7d2b0092485194832828b4856786\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ef730513334292a911b5d8a125af3c : SUCCESS in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e31310816384170b630e905ecdff349 : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f66eb6f8f22e430c914db957332289eb : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26aadedc22134ef6ad5767fc3bb21dad : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/443c91f5adea47948488c2d01fd1f0e8 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19e456ceb2294ef4a9e6f81e72778ede : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e9d0c9fec9b4305be96f624e7790620 : SUCCESS in 3m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1608e043472b4deba258ee493d079ef5 : SUCCESS in 6m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ce4f494a1ef74c50978f3f113ae398f3 : SUCCESS in 6m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/70fa54a3a92a47cfbfee8c84c3542c50 : SUCCESS in 15m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7f112e4c70e34dd2810ba9ac9217a78b : SUCCESS in 26m 21s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7175eb4b6aec4ea6a8bbab5e9e3db7c4 : SUCCESS in 3m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7741fbe0d69c480fbd7ff3749accf4fe : SUCCESS in 5m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a9a9f869136344a498a48b8f7274a13d : SUCCESS in 4m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a124fdd58bdd41f4b12626270a20b6aa : SUCCESS in 9m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9b62bc5fafc84d66941c5304a5805f57 : SUCCESS in 5m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d1eb832131684e4692b8167a1ab3050e : SUCCESS in 6m 22s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2c9c1285c04b4348be3747bc4259f251 : SUCCESS in 5m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/68c9f616cf9a4f189c3e1844a190d73b : SUCCESS in 14m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b24fb9dc56b4961a97bd9814950fed3 : SUCCESS in 6m 23s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/29b068de59154cfc9d0365e3aafc1cb9 : SUCCESS in 8m 28s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/11635d0396ba46cdb2eb86afcd1ed66b : SUCCESS in 6m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/775fbdbd7ddb47d2bec4143d8e20c45e : SUCCESS in 1m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e2983bbaf8f34dfdaa1894d8c2056a4a : SUCCESS in 5m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a990639f60e94228b63f85f1797de1ea : SUCCESS in 4m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4d3e7ea380064899ba61d2f300d19ae6 : SUCCESS in 2m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9c1f95639fb540a1a05294c30e0753d8 : SUCCESS in 11m 05s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8bd82d67695f464e895af7807ebce0f1 : SUCCESS in 13m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d72879281f6841c6bf21ad6654cf5547 : SUCCESS in 7m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ef0870053bde4220b80501fb855a18db : SUCCESS in 7m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/42eaee390a4948e99c0a9c9ac756b4c3 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a91efc155eaf416290ce42d1c042b45d : SUCCESS in 2m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d65e01de4ace44418377fee9dd18298e : SUCCESS in 26m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d7b8824436034e91b7ccf33e52d9fe8c : SUCCESS in 3m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cd6936bc8b784b55b719df63ae7f9408 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/541d6ee2cb274527ae1f378598e1cae1 : SUCCESS in 3m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/abfd31eed39145febd563e0d3d78032c : SUCCESS in 18m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/610a29404beb4e498a02c21557e06705 : SUCCESS in 59m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ff364e06ace4e3d9e79bd319ff2bcac : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":1},{"id":"4340d25a9c9bd4b3bd84c3f99ab458aed8c91c9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-07 13:28:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"addf6ae851c3bfec09ff143239a0afc0756f164a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:34:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ee3bf9c29754ccfbe169ace8eb1dc06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c912369fa4dd43768e74ec455c3e957b : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dafe0fa63e48484e9dec10586a32ee0a : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a774bdb884934c5db22c64cb747ba723 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3dfa40b46e234a9e98d4ebf0953b4733 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c0b49929cd74cb994d0741d76b1ae54 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfbcd42ed4c24e2ca3b5202f278d667a : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f38d8448f0f84b0e8f02e7215a442930 : SUCCESS in 4m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/73dabb56911b49bf81a3984f29b5e592 : SUCCESS in 7m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/02fa8ae375604fe79c762d07f2947c01 : SUCCESS in 8m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f983fc73a3cf4060ae3278a3e928a5e5 : SUCCESS in 16m 22s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0acd50ea3f074c7aaec4d2a370bdfea2 : SUCCESS in 27m 34s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7783074ca6ba45528310e84ae7d874e0 : SUCCESS in 4m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/96cf600bb2874ead8775e2d616d66b42 : SUCCESS in 3m 30s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e6b9fdce87384fe683ac7bd14591a510 : SUCCESS in 6m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8e78c6627f4d4f9f941a22de643e0337 : SUCCESS in 9m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9343c07189654e33b856b45239af23b8 : SUCCESS in 6m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6426b29a86b1443394ef291482e33482 : SUCCESS in 4m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/18d92c6ef859481b95ce5a52b5621928 : SUCCESS in 7m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dcc6705bea5042adba37774c76986208 : SUCCESS in 5m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/06cf2d474b7b4c239a2c653d92983092 : SUCCESS in 9m 23s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c7c8a2211f42436cb051df00a0310da2 : SUCCESS in 7m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c2c3cf47ff2b4e33b15d8ae294f8c019 : SUCCESS in 7m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2fffdd0f8a0a40389a287825dac8793d : SUCCESS in 1m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a5e9650b21574c85bbe8306fb461db59 : SUCCESS in 10m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c2594301a48246b0b6cf75754726c924 : SUCCESS in 3m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e9d1712424ac45e49da7e60de3384598 : SUCCESS in 4m 23s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4eb9f8ee68cc49679bdb9bc2bdce4a54 : SUCCESS in 9m 31s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/80a851459cdf4f2d9ea47a6a40d88448 : SUCCESS in 14m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/463d98e63d494025b127a4d17cc87876 : SUCCESS in 7m 06s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a963d6d9f42742c981ba970da4361ecc : SUCCESS in 12m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/14350390c4974852b4af15de4054a0cb : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a2fce47353d4b619015951a9dec0c7c : SUCCESS in 2m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b5c671de96b4c2c9866939dbbf6b314 : SUCCESS in 28m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9d85b23b07724258a48a2dec851bab39 : SUCCESS in 5m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cb049c2b6ab84f8e87fdbdd1074cc0f1 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/11a4b8e76d904a32b03926222f443a76 : SUCCESS in 5m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c85345f42f9b4643afb89434dad6c787 : SUCCESS in 21m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/632221a71efe4327aab281f266991b76 : SUCCESS in 1h 00m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67e8b8718cfd4d58882b605053be471c : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":2},{"id":"6379e1f4c43a7be010df069cad585ef2de7903e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-16 11:10:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9b9d7f1886797a5d02d223120a086f8aed5b3c42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 12:37:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18d1301c8b614b929b787ac6fb7ba962\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0dc65c19a94cb084efa64c2c27c340 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5f2bfe3283946a4b4822fbf9cf1d6ef : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e00e537ed0744b17ad7b8aa7bc9f260e : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6570f81f3bcd43b88bd4484a1fc576d1 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1b908de01804173ba39fc4125346b28 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ade5be18abf445d8784169ea2b35cf3 : SUCCESS in 1m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e91d01169014fde811bd1c42034b88b : SUCCESS in 2m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/263397dc87fa46eea63a25ef8538dc99 : SUCCESS in 15m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/656f1f4df8cc45599eace81fa1e4dd11 : SUCCESS in 8m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/95c08b9826a241fd88c542fbc9d45010 : SUCCESS in 18m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5d7eb1c78e524b849f39308ff5757ee7 : SUCCESS in 8m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/108aa765050f4a95bd85db71a1e2230e : SUCCESS in 11m 58s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/85ca237b2c4d4a078944d4a57279baf9 : SUCCESS in 5m 11s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/aaa9423425cd47389740e9c12070b626 : SUCCESS in 2m 38s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4991a98443924546905f67bd6fd7f336 : SUCCESS in 3m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/16212c7939d245d78d8f3b54aebb72e7 : SUCCESS in 12m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4517f083b3a046c882fbdc87aa496756 : SUCCESS in 6m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/aa3ee0b5d9e44dbba41a8af9fdc00382 : SUCCESS in 7m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7b2b33209f8943b182cf1daf03a0d63c : SUCCESS in 9m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3aa1b6fa8f35476897eed0f80dcff180 : SUCCESS in 14m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e902f7afcbcf4ac0b80cd9ba72f50453 : SUCCESS in 16m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bab29b20377d453ea11e069eec10507f : SUCCESS in 9m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e81ac55808854fa58a08a253e68b558f : SUCCESS in 11m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8de83787ad184567bf01efdc6e490e0e : SUCCESS in 3m 36s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/09511c7994fa48efa7d76d21642a2ea3 : SUCCESS in 12m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8142069b109b444cb1dd9eb85c08f497 : SUCCESS in 6m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/eb1e7187627b48569d8d4091f2f887a0 : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ae575a47c35d488c9250cf7a83e47c4b : SUCCESS in 9m 38s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b84fe2cfbb104ab9b9070894c9e06a86 : SUCCESS in 16m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3fbf6fae5bea433babc8d57802910d94 : SUCCESS in 4m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/df54c7e53a0047e380193406eb364f5e : SUCCESS in 13m 22s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/81628d3b99ea4964a46f9b7e13ded7cf : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/323846907b834f14ae285edd2c3e7a15 : SUCCESS in 2m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96f422d5daf34bba96bccbf776e341b8 : SUCCESS in 29m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b32bcf34acdf4ec1b2a771198cf4f4a1 : SUCCESS in 3m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/80176f14abcd43d9b3bd04ae702158ed : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc14a8807bfb4bcf80ad410f0d1aab3d : SUCCESS in 4m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9dccf354e19e40d4b4b483727a8cde22 : SUCCESS in 17m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47f8b57b79cc4bea8a016b78d4a518f5 : SUCCESS in 1h 02m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34bd1dffbaf3498fb4161c927e102fdf : SUCCESS in 1h 25m 11s","accounts_in_message":[],"_revision_number":3},{"id":"6a4dbec6f19d7335eba22dbd788e97e36acfe5f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-26 02:54:14.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d4262d43bd6b10d930351030f92cf5ba7b70d46e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 04:39:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca11721769c642fea8219103bd53ec7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/934a258446bd4a8b907067810b595c23 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b81175db32b8426c96c49358f4d7401f : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57aaa18877484bbe921c87360ac576f3 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9857672592ef4bbaac7ea11b6c0d4ca1 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18be88af08c8481fb26e3e241e8c07f0 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40e2df71f8b44e4892d8c989adbcbde1 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a10c8f841c4941a1b5d2ebd7ea20279c : SUCCESS in 3m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/86b0258d7aac4049aaab0bd17be29a3b : SUCCESS in 15m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bcd2f09de30c47048c4e117120b80027 : SUCCESS in 10m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e4ddab2b3dfc442f8b285390ec1e81c4 : SUCCESS in 20m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bc24dcaa9fe8495ebdbd573cfc29e1fc : SUCCESS in 15m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3fb3c2164a0748b4b09fadcb10676d9d : SUCCESS in 22m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9e8cdc66fef6482992a1aabfe6a2e148 : SUCCESS in 4m 32s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/38c39f760e334e00af1d3bcf4034fa2a : SUCCESS in 4m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8ff71481b81640d996a216340ee2ab5f : SUCCESS in 5m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/848cd55dc4ad4795a2ea9ab24a3ecfa8 : SUCCESS in 11m 05s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b5fa01179a6c4136af39202e48b5d873 : SUCCESS in 6m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/05e5c086c5304d56856dcc71fcbf843a : SUCCESS in 4m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/349ea42c621846eb93df215634b3620b : SUCCESS in 8m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a81e1d0c281c491698b58999507a26a9 : SUCCESS in 13m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b289f86ed2741aab4b42dc48bd868fb : SUCCESS in 6m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f725f1a3acdf4aca94898488a935abcc : SUCCESS in 10m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c5c72a65f46c45a8988371c39935b2dd : SUCCESS in 14m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b06c62296fdb4e57951d1c7da1e02986 : SUCCESS in 2m 39s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3a252d42aa6c4f45b063a25396a620da : SUCCESS in 10m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3c9331d7465640b898ec4c2510abe88e : SUCCESS in 4m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/542e861dddac4fd3a8b5da4eada401c1 : SUCCESS in 5m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a29a8b345a4144b9bacccf29f62516c0 : SUCCESS in 9m 30s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1d00b1b963d647fcb175bf8d4b2d4dad : SUCCESS in 21m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e0112aa24d664412b78d09a4823b4929 : SUCCESS in 7m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/29885cfb962045728a2346ab6c1f49a7 : SUCCESS in 12m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6471f654426243f48e7a874335f27353 : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f2680c202e214b788eec6b9d1e507edc : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24ddb5bbba624a64822cfd56df968993 : SUCCESS in 23m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8b2bb16ee296494b8239809864057eaf : SUCCESS in 7m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f7c48be9711f491b9777b20041032d22 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f131ee6ddd964b55b12de66ccf3c0630 : SUCCESS in 5m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3e2f68a36d26433eb1cfd15114ce2fd4 : SUCCESS in 14m 06s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/169b69cf8a5849979de9fb3e0733a81e : SUCCESS in 7m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f239325dd154dad9f8f38c67fdbab2c : SUCCESS in 46m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9545a1ec5d54c9dacdd1f6dd2d42e86 : SUCCESS in 1h 42m 44s","accounts_in_message":[],"_revision_number":4},{"id":"a24db1fbdffd17e80e51623578c423de00385e55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-14 10:16:29.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b9708201651a30c2ba9e8bc948ae248ad0a343a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 11:22:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19330804fdd4472cbedfb3589b78e34b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f2555231874eb38e226280c0555534 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9a6c618ea94c9bb21eb7c22f35394a : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cac21484436496488218cc353748566 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5645070407c4149829c1786c9de9a2c : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68c2bcd80e4c4b19a2e6080ecab744a3 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8632411903f84efa85e652b3a3a9a90b : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f2b064b037944fc6a63a5b93c5a957f1 : SUCCESS in 2m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ca213025a08f40a3a9b0554399d96143 : SUCCESS in 16m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/496f785b17c842828b98b176c5868cd7 : SUCCESS in 9m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/14d764140185400798bf8da088566d86 : SUCCESS in 16m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c2cc6c679a3b47eea1588d7979374d68 : SUCCESS in 18m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2976e17b2d3d47788787f1122e377580 : SUCCESS in 24m 46s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e5eb8b64f4094ca28047febc553351e9 : SUCCESS in 4m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9bc99c97ba4945feb9ca8d90ecd682f4 : SUCCESS in 4m 37s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c0800ef4eae34265bdc9a0a46b9a74ad : SUCCESS in 4m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5cdddca3e5a64fe293094a815213f5b0 : SUCCESS in 11m 14s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/856837219c8b4b229ab79b3a61f2217c : SUCCESS in 4m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e4c1b6779bda48a1aee63b80aee078bf : SUCCESS in 4m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/65efcecddb9d48ceacb90db07294e277 : SUCCESS in 7m 13s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2b94f882b4e8483b82f357a64521ce2f : SUCCESS in 15m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dcae6a51605e4152a0b7bfa8cb1dae23 : SUCCESS in 6m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/938219ef8bf0406ea43b1f67f82aaccb : SUCCESS in 9m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1abce181f29b4e99ad27f61d0300dbf9 : SUCCESS in 10m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f8aabecbc49141899cbefdb616166c7d : SUCCESS in 1m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cf42c65fb86943309fb300f908ff2282 : SUCCESS in 11m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b0ce109d384c4672b0fdc240cdf208d7 : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/34a0bfaaf0eb417aaceef6b02d1eeee4 : SUCCESS in 3m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/85efe2ef99aa4cf191cc643073e57532 : SUCCESS in 10m 55s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5ede6809a70d436fa657b547ecfcb1ce : SUCCESS in 17m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/46f5e55562b64503a64b7f6e50030eab : SUCCESS in 6m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5615094ef23f415a9cafd91340271b04 : SUCCESS in 12m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6edf5b4115be4bd69aa5a804ae7ed336 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e66953508d614a7bb30fff105441ffd4 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/992026370bd34764a7f375ae9186811f : SUCCESS in 23m 28s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2f5b913e0dcb4db69e9d2600adcaab9a : SUCCESS in 7m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b694673ecd8e4452b6f5cf3c882f3e64 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c2a35a5e88e49eba8be510c97d5a54e : SUCCESS in 3m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b8a6762d4c3b41f1be27db8ef1f0b6f4 : SUCCESS in 13m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a89cb56a96e24ebc890740844d53681c : SUCCESS in 5m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b614c6f5224447f8cbf1bbbecb76dbd : SUCCESS in 35m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d71d443dd3245f290c46555277b1a69 : SUCCESS in 1h 04m 16s","accounts_in_message":[],"_revision_number":5},{"id":"c8b96f5c2d56b15430b9ff9f4f8d26927ff06f34","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-07-14 14:28:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"87d378b3efcd5cbd48bf98d3e32ec6e6b7d8d9fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-20 10:21:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"97892e18f75322616dbd11d37bc59c99a32cae0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 12:03:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/448144dfbffe4be9b937b2bad699344a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d3a6b93d4b4a1a9a2b3f2bd31ffb1f : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2de384f3af445c18181494fc27deea3 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfbf976f57224b8a87aedd6b5c321790 : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7ac1112835549819bc2e030c60772fd : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d1d76c83e2b4fb1ae38e98dc2402453 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0aea609a5c74eb290e8c919bdf6a931 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/38c77cc8d7984461924c2861e4e42e8a : SUCCESS in 1m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1255d2ee997d419da64d0a472d963f03 : SUCCESS in 6m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c4cbf973831d445f9b3d5cb2cd3db01f : SUCCESS in 8m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/67f78607b6b64548b6b341c8a8705b57 : SUCCESS in 6m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/374acf6e2c73444fb01927831292673a : SUCCESS in 18m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/847f8d2835a94d00ab6276982ddf4ba0 : SUCCESS in 22m 48s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c9f6874eb4104d95b8fdea2c2003d3b1 : SUCCESS in 4m 11s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ebb805aff9104ee1ac32118091ea2300 : SUCCESS in 4m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2eb463e0579f4964b157ee0e2e99dea8 : SUCCESS in 2m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f1e127a0ff2d40d4b3de21ab2850fd34 : SUCCESS in 6m 29s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8c10b2599bb343dd8f3420c2bd3fcf61 : SUCCESS in 5m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/89c20cd9c8584192a9123b3a85ad43d8 : SUCCESS in 5m 35s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8c264c16ee264d2c9036df7163420040 : SUCCESS in 3m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6714db234abb45a988f24e9c829dd5e1 : SUCCESS in 15m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d1e183c8040143eaae9a100cdb793876 : SUCCESS in 7m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/94bb901435314c2fbc2ebf753df58762 : SUCCESS in 7m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bb7d40db246e4ea8a02d1e60c619ff07 : SUCCESS in 9m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f99be0827c0f4a09ab4f3134c235b0a6 : SUCCESS in 2m 43s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7016fc24e392497a8f0a6932556ffd2a : SUCCESS in 9m 13s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/00cea6e19c4745668da067af22eaf20c : SUCCESS in 4m 30s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f083d2cd6aff44fd852a1af7650d9ddf : SUCCESS in 4m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4e1194a3081a4b9caf83675de3cd7154 : SUCCESS in 11m 25s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2c8b05346d1a4e5ca95228bbe00edabb : SUCCESS in 14m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d0909cb1e4b34b1ebb0a21f5cade57a8 : SUCCESS in 4m 06s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/44b8635e794b49d7a94235d1920f80f2 : SUCCESS in 8m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1743115dc1b94e19af1722de7ffaca8b : SUCCESS in 8m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d19482b06d3a43659aa7916d06744b94 : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ed55c52e5fa49e6948146593f7ae783 : SUCCESS in 23m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e32576d1bc5b4a1d94ec5a3f89c7e904 : SUCCESS in 5m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aa4c3fb48ea642b5b1f168f53f2736bd : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/587704ac2f2d4d7c88e3cf52db7052ee : SUCCESS in 6m 34s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c39245f1d0314027934b985aca3629de : SUCCESS in 11m 36s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4b05eb6ea92147d8a8f2d840cdef08e3 : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf33414daefe4a73818e8ef24419d2dd : SUCCESS in 1h 00m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f5f1866289f4634853205de11785e25 : SUCCESS in 1h 37m 00s","accounts_in_message":[],"_revision_number":6},{"id":"640444da372f86a1d6d76bfd680bb72946c3dbf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-07 15:19:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3fac2ef38d434feba9eccb42c071f5329d6d5e12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 17:26:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cb8b52366b74ab285b2cbd85225959e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4167c3299ddb4bd6aaa99c3c9f5d0193 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518ef5bc4a484007a017e4b848a47938 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cc40bf8b65a4e68a840f7e1a3fe7edd : SUCCESS in 5m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86b6a95ebf1a45fe8dd0e07c143e88fe : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df008c149eed43eab482911a7ae3f048 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a96b766586e4baebdb252220d74799e : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c8d3a9bc3954412991566e461ebb7de : SUCCESS in 2m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0b9406fc0e614b24bf5c6466625cbd36 : SUCCESS in 18m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2263a34e5207400f95f815889dd5e63f : SUCCESS in 17m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b175ca99a74147d3937a7efe3b0fea91 : SUCCESS in 23m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dc2357708ef440de926da82f17d667bb : SUCCESS in 22m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1d3a6fb079cd4967a231f747d6f6c7b8 : SUCCESS in 26m 22s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/181f2ef1ac0e460aaf03984ee0066ec6 : SUCCESS in 6m 32s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c9451d4770384874a5a539e822a4c087 : SUCCESS in 4m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3e48e26f551240e6b6b5168f1d331cc5 : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2af5f016fab04b7db05ef65974c648b6 : SUCCESS in 11m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7bafaf6c95854069bd253ba27a95da5c : SUCCESS in 5m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a671c469524e46bf8daf95a5bc0a273a : SUCCESS in 6m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/930c67784d8a472a947e5232579b7386 : SUCCESS in 9m 20s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ca98a07e3d014666a4ee2c6c76ef08e6 : SUCCESS in 16m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e7d632395c04e77bde3c87a29177b00 : SUCCESS in 5m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/85bfd31d324443f5884c8893ea707d11 : SUCCESS in 9m 48s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c0a60f8ed5894be687101779494d3b99 : SUCCESS in 12m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c235383fb9f04ca99f37e901a201d057 : SUCCESS in 1m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0f9166dc3de24bcbbf086bb2e693c684 : SUCCESS in 10m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/320abeb41fe940e188d020e2bc7a9376 : SUCCESS in 3m 42s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bc1284acfae149b79fe82c6346a5b230 : SUCCESS in 4m 41s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f0134a9e08dc4aefa40bb48f229e656f : SUCCESS in 11m 13s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f65d692d42c543a0b1b4c1683950ea2d : SUCCESS in 21m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e0a700b084944277825e5be7810c3094 : SUCCESS in 9m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/51a3020371064c15af16b44ff43cf2c5 : SUCCESS in 13m 12s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d12765e26817464ab769a6cecff1d82f : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a83f91b002e048d1a1c526ef6cf2d655 : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc76b66659a64676b100a320033355c4 : SUCCESS in 29m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cf128d7cca094210aed0a8ac74e18dbf : SUCCESS in 7m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7aa7ce3d8c304a0ebb5ea679492eaca2 : SUCCESS in 4m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f16f82d334a4e648761cc430c2b0595 : SUCCESS in 6m 44s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/11fd4a779c2246cbb43dc0c8f4de5842 : SUCCESS in 14m 42s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/edfb154489d6439e948f02439a67aa64 : SUCCESS in 6m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f20dac6622944f328632ccc8fea57df9 : SUCCESS in 56m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b113bd064594387af0a8b2555d44077 : SUCCESS in 1h 58m 00s","accounts_in_message":[],"_revision_number":7},{"id":"33b945bd66cd92968db36e54bd136ce675ff322b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-08 11:14:02.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6ec3625795268d5f609b368024a41ba702b664a1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-09-11 13:22:26.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0709acc04f75c0202ee52d72b7b2d87a44219e60","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-13 23:17:15.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"9bffc9ef18fe6c92b9f104a572c60e6e8cd2cb7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 23:18:01.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c99c72f2c1016d5b4b0c47d90a5dbafdfc85ac5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 01:21:47.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc8c339d696c48f5a8d1518eb1ca88e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a67dd13a5dc4d06b39dea0a1dc2243e : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf2debcd146f45ae8c7dc9e989d748b4 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6be0dc332ad641c2af7b3e2a4ea2eba1 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb37e7e0d2c549f1bf000b201b2e27a0 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbedb90f05764c2b96fbb3c59ff14c27 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5007a5f1d9d14e248795e4aba9453bcf : SUCCESS in 3m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/39f6a418e0644b25a178e78306115f95 : SUCCESS in 10m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6e75cbbebf784b639a52e651f21d4365 : SUCCESS in 16m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/087ac0d623a5424e85b48ccc43ab6bee : SUCCESS in 19m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a3bc05e1d6324e4a8d098e4ba823d4ed : SUCCESS in 15m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ee2270d1370345d1bb42cb7bf1aecb28 : SUCCESS in 6m 28s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d7882eee6ae1444bbb6e481e9a934caf : SUCCESS in 6m 59s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/81e545a2fc154d31bb00d12fb3ff1736 : SUCCESS in 3m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/34123907eca84198b8a5aec23d0c2026 : SUCCESS in 17m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/faad33141cf64d899b0817f73d9da93f : SUCCESS in 10m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7b8fb7236a1441b1936ad259e9a808e1 : SUCCESS in 7m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/66cb7e72c6584548b557324f2f00a740 : SUCCESS in 9m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/651e41adca554f3db72728e551b1998c : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/741460c3dbbf47a78b6ba395d705e9d9 : SUCCESS in 9m 00s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4973509b0afe4653a5298459347640d3 : SUCCESS in 18m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ed000d86799b48089592c9384f82c17c : SUCCESS in 6m 14s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/44b282ceb9214cb4b9a311777d8a6018 : SUCCESS in 13m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a891101fb10e48eb8310bac5cdbda2bb : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9b3af390d93d43baa80b47cf4e51cf0a : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/562243a563bc427a84ec9f7ee5da1e85 : SUCCESS in 28m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9921bd4a3870436c99518e96504c3234 : SUCCESS in 6m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6f58495ac0274c74b602d18c4e867032 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b129f3eede9a4cae861cf78787142b4c : SUCCESS in 5m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/901a3833325b41d6ad359fb656b53c6e : SUCCESS in 13m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/101e072aa4664a368525a2210ae82f02 : SUCCESS in 5m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11eb983653a94cfd9a58d9c58ca3d59b : SUCCESS in 1h 04m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ad4a63c4b67490fb5d036a209da5fe8 : SUCCESS in 1h 59m 26s","accounts_in_message":[],"_revision_number":7},{"id":"82cb07e4a2ffb75f2638781b32308f73871cbbc4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 01:21:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"37affc1e10a5b00518059592f86413dca24d2066","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 01:22:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d47be537784a40d89de2d8ef0f0f22cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae360b788ee64be2b63decb05f6d62e4 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0fdf79db7b059c9d47889f80753d43e51b9720ff","revisions":{"4efc5072969dcbe44552725bc018b781d85099ac":{"kind":"REWORK","_number":1,"created":"2025-04-05 16:04:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/946499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/946499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/946499/1"}}},"commit":{"parents":[{"commit":"aea4bdb03846d4b08c0b3decf0ef6dec618a14ad","subject":"update constraint for os-traits to new release 3.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aea4bdb03846d4b08c0b3decf0ef6dec618a14ad"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 15:59:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:04:24.000000000","tz":540},"subject":"Remove ldap3","message":"Remove ldap3\n\nIt was used by barbican but the dependency has been removed.\n\nDepends-on: https://review.opendev.org/937752\nChange-Id: I702394b5f925d6a3e1833109a82f9827cdc3088c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4efc5072969dcbe44552725bc018b781d85099ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4efc5072969dcbe44552725bc018b781d85099ac"}]},"branch":"refs/heads/master"},"1f38f36f91946c060d980d5b96ab987b1a757a36":{"kind":"REWORK","_number":2,"created":"2025-05-07 13:28:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/946499/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/946499/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/946499/2"}}},"commit":{"parents":[{"commit":"b58f4867d2406854e476fa599ca52d270aa2cd19","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b58f4867d2406854e476fa599ca52d270aa2cd19"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 15:59:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-07 13:28:14.000000000","tz":540},"subject":"Remove ldap3","message":"Remove ldap3\n\nIt was used by barbican but the dependency has been removed by [1].\n\n[1] https://review.opendev.org/937752\n\nChange-Id: I702394b5f925d6a3e1833109a82f9827cdc3088c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f38f36f91946c060d980d5b96ab987b1a757a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f38f36f91946c060d980d5b96ab987b1a757a36"}]},"branch":"refs/heads/master"},"14dd476c7365c68f27308f0d5bb0a6834ca41216":{"kind":"REWORK","_number":3,"created":"2025-05-16 11:10:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/946499/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/946499/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/946499/3"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 15:59:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-16 11:10:47.000000000","tz":540},"subject":"Remove ldap3","message":"Remove ldap3\n\nIt was used by barbican but the dependency has been removed by [1].\n\n[1] https://review.opendev.org/937752\n\nChange-Id: I702394b5f925d6a3e1833109a82f9827cdc3088c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14dd476c7365c68f27308f0d5bb0a6834ca41216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14dd476c7365c68f27308f0d5bb0a6834ca41216"}]},"branch":"refs/heads/master"},"1fed507c029de0457cde0938e4517fb667bfa868":{"kind":"REWORK","_number":4,"created":"2025-05-26 02:54:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/946499/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/946499/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/946499/4"}}},"commit":{"parents":[{"commit":"0697f59bec1402466cfc71e33529293ac1739e72","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0697f59bec1402466cfc71e33529293ac1739e72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 15:59:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-26 02:54:04.000000000","tz":540},"subject":"Remove ldap3","message":"Remove ldap3\n\nIt was used by barbican but the dependency has been removed by [1].\n\n[1] https://review.opendev.org/937752\n\nChange-Id: I702394b5f925d6a3e1833109a82f9827cdc3088c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fed507c029de0457cde0938e4517fb667bfa868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fed507c029de0457cde0938e4517fb667bfa868"}]},"branch":"refs/heads/master"},"3e55ff49bdb6eb5305a4aef85894fdccc635544f":{"kind":"REWORK","_number":5,"created":"2025-07-14 10:16:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/946499/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/946499/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/946499/5"}}},"commit":{"parents":[{"commit":"1f2be26ab5a4d601127817b90c794e61cc40bd12","subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f2be26ab5a4d601127817b90c794e61cc40bd12"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 15:59:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-14 10:16:21.000000000","tz":540},"subject":"Remove ldap3","message":"Remove ldap3\n\nIt was used by barbican but the dependency has been removed by [1].\n\n[1] https://review.opendev.org/937752\n\nChange-Id: I702394b5f925d6a3e1833109a82f9827cdc3088c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e55ff49bdb6eb5305a4aef85894fdccc635544f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e55ff49bdb6eb5305a4aef85894fdccc635544f"}]},"branch":"refs/heads/master"},"f070a2195ad3a6af951b894a0f05f8827392b13b":{"kind":"REWORK","_number":6,"created":"2025-07-20 10:21:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/946499/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/946499/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/946499/6"}}},"commit":{"parents":[{"commit":"d78665852a59c1b18559c53ca11fbd9b672a0fdd","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d78665852a59c1b18559c53ca11fbd9b672a0fdd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 15:59:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-20 10:21:43.000000000","tz":540},"subject":"Remove ldap3","message":"Remove ldap3\n\nIt was used by barbican but the dependency has been removed by [1].\n\n[1] https://review.opendev.org/937752\n\nChange-Id: I702394b5f925d6a3e1833109a82f9827cdc3088c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f070a2195ad3a6af951b894a0f05f8827392b13b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f070a2195ad3a6af951b894a0f05f8827392b13b"}]},"branch":"refs/heads/master"},"0fdf79db7b059c9d47889f80753d43e51b9720ff":{"kind":"REWORK","_number":7,"created":"2025-08-07 15:19:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/946499/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/946499/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/946499/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/946499/7"}}},"commit":{"parents":[{"commit":"6324b28b1bfe61eace77bde3978daec14ed3d176","subject":"Merge \"update constraint for futurist to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6324b28b1bfe61eace77bde3978daec14ed3d176"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 15:59:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-07 15:19:30.000000000","tz":540},"subject":"Remove ldap3","message":"Remove ldap3\n\nIt was used by barbican but the dependency has been removed by [1].\n\n[1] https://review.opendev.org/937752\n\nChange-Id: I702394b5f925d6a3e1833109a82f9827cdc3088c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fdf79db7b059c9d47889f80753d43e51b9720ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fdf79db7b059c9d47889f80753d43e51b9720ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
