)]}'
{"id":"openstack%2Frequirements~946500","triplet_id":"openstack%2Frequirements~master~Ie550cca55e7c5d180701b2b8c2c3eb8752694e19","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2025-10-14 00:33:50.000000000","reason":"Change was submitted"},"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-09-11 13:19:36.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-10-13 23:16:22.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-14 00:33:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie550cca55e7c5d180701b2b8c2c3eb8752694e19","subject":"Remove unused pywinrm","status":"MERGED","created":"2025-04-05 16:05:41.000000000","updated":"2025-10-14 00:34:33.000000000","submitted":"2025-10-14 00:33:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946500","meta_rev_id":"8ca86492ca8c17e9986703843beb242dd8bc54d6","_number":946500,"virtual_id_number":946500,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 00:33:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-09-11 13:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-13 23:16:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-10-13 23:16:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-05 16:25:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-04-05 17:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-31 14:34:52.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2025-08-29 08:27:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"e39a30d5719b5688dbd9b59893c73745400bfc83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-05 16:05:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfb783d604df3360f6ac8d7a6782e4b590d749d7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-04-05 16:25:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf0b9d4911fa9ae9c8f806c7831837d1fe789397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 17:08:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4e597777e6e46e0b76542771fdffecf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aeb6bd8366f45588a662a9c142e3340 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4135ab606af64793baf7d6ffa8147f78 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78de824f1b784808a04ce89347b25f57 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb5db5fbb5c54c1a8772e7c5cad40393 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdcd6029d1e54990b5cbe18fca7a410c : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d492e6280664004ba724fc2020e265a : SUCCESS in 1m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d550e8baceb4865a6b11fb20f72d5da : SUCCESS in 3m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8621d0491dd348bda05f321ea7ba7b14 : SUCCESS in 6m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fd5a07ec6f214c48b47094735c98b9c4 : SUCCESS in 7m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ed1b3cfc98404352b80f0b55a0814e80 : SUCCESS in 15m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9b261a7898a74786b1a52ce79967b8c6 : SUCCESS in 24m 40s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8e782a64dcc04b2d8542df4f09b31d97 : SUCCESS in 4m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f28ad668c729459d83b656fc0f1b88e6 : SUCCESS in 6m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/96e8d4300b1d474a81f09c7d5c5c9206 : SUCCESS in 3m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0454fc238a554bfbb257375f36fbbbda : SUCCESS in 11m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0cd9b9bcad674085bd37967217697217 : SUCCESS in 5m 53s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a4ead86664314d32bbc76cd08c703f2d : SUCCESS in 4m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/83c214b0031948ff931f1e62d33f17fd : SUCCESS in 6m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6f9781707d9b4495ad1a84d3b8aed333 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/00211c194d9a459281d89b07548c9403 : SUCCESS in 5m 23s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4d5db5a0e3134e948f1038eab7b4d72a : SUCCESS in 7m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/73cfdb51c40f439c88159a6338adf2f7 : SUCCESS in 11m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4062fd9d88c14dbd96095321b3d7c737 : SUCCESS in 2m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ba7bfe817d244aeeb9a6e99c57fc7bfe : SUCCESS in 8m 59s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/21219050791c49e6b24332bf3a02d42f : SUCCESS in 4m 21s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/439c61bf781e4ca9ad577b4bcc11376c : SUCCESS in 5m 12s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/798b9001ee3d44e192f2010556f5bacf : SUCCESS in 11m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bd7b0dfc8251401fa66e99cdbb1558fd : SUCCESS in 16m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f263fec12744405093e5798b098b84fa : SUCCESS in 6m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9b503141acc74e9cb5583a8b3d29da44 : SUCCESS in 10m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f7e81601b8a64eda9d4419013d8bdfa8 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2c16885e60494a34b0acf79643ac0e6b : SUCCESS in 2m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf08912899124e10ac46be6c61d7db1d : SUCCESS in 22m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/66ef391211724bdc9e896e023028574e : SUCCESS in 7m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c80d8229d5024952a45e519b3bfd7148 : SUCCESS in 2m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f449e00ce404bc3ba55ec226686ae3e : SUCCESS in 5m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ea384d3f14dc4c638c9a3508c6337c04 : SUCCESS in 6m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be74f9ecaca148dc8b44c81587fc6e65 : SUCCESS in 56m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66d9cb1f2deb45bbbf641a348cd532e3 : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":1},{"id":"795efaebcb97b9918e8e9874338e224395a74aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-07 13:29:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f403a1a70c94e3156134f2b2ddd947a770abc073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 15:24:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66cbf2e72a184d26ac543321270efe82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2541e11b3f6342c695b5bd9a666b3efb : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1adad6e333594af9b2716bb77c484a1f : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2f66faf2cc745b4aad06ee3d713550b : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dca884eb42d44943b42422c806f53b4a : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ccfd9bfad30d447e9e736486c6e75d5b : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06e0b7c62c1f4b06aad85586282adf9e : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb4ae99e8cda4e718658037c9fd7fb96 : SUCCESS in 4m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bb9f17dcdb434d18bd8d30d66a5922c9 : SUCCESS in 10m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/009b0c8864ce4e83a5f5ec81229b9e1b : SUCCESS in 14m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d9df5f1af9e64dd1b968139a139d98fa : SUCCESS in 15m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/43902293740144a0aca4dd5529cc00e3 : SUCCESS in 18m 44s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2eabfe88b3df4cbbb3f0d7c3f348f90d : SUCCESS in 4m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1bba43ca45e34cafb46c42ec91292cfd : SUCCESS in 2m 54s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d02ce72bb4d343a091eeb6bd2c7e353b : SUCCESS in 4m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6a72cab221234646b389ce9ab5968b18 : SUCCESS in 9m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9df3991779f248738f782c59883d74a2 : SUCCESS in 5m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3db065208ca840408f4380901d9d2b70 : SUCCESS in 2m 52s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0ab0e11f0ad048d3890b71090b9bfae2 : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/439db4e8d82640b8bc5f67b3a041aae6 : SUCCESS in 13m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b9ee3b03ba90459684ec09985f0ebbc5 : SUCCESS in 5m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bdc921a4b2224c43a79f93429fb8c799 : SUCCESS in 7m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/79d59deab6fc4c168e9118c2faff6f19 : SUCCESS in 11m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ccd614b452e34d0a92429e6d597a7d3c : SUCCESS in 1m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2bb31da1dccc49d382323cc44e450110 : SUCCESS in 10m 41s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/cbf29793860749eb97ee53947eb96f48 : SUCCESS in 3m 36s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/779fcab54f7b4fa89b3e3edb841740cd : SUCCESS in 6m 51s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c12542f2cf1341cb869dda09970566be : SUCCESS in 13m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8714f2a5de894f8a8ec6547994bbff1c : SUCCESS in 18m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fd92749cb23644938f3c1a5abd856b95 : SUCCESS in 9m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/42e862e869ca4d71864879742feda2e8 : SUCCESS in 12m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3e8559fdfe0b4499bf3e8603c63c9adc : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/377ab7ac48694d97a78627873a11c2bb : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38c0e0d1a42245eeb60e23f5f3327218 : SUCCESS in 23m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c818761154b54710858127e7f9b549d8 : SUCCESS in 5m 37s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aebd2ee3a5d74e638b3ac9df11c0309c : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee63bfd2a4374561a3d955fc15a56074 : SUCCESS in 6m 19s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9648d40f78bb4b0585f1ba52f2e8c314 : SUCCESS in 16m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04e68d49e7cc44f49878baee9f1a3986 : SUCCESS in 54m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bca3ccf1a534c78a86e273fc57418b7 : SUCCESS in 1h 49m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c578bda287b3c78a56519dbe08187cc3dc263066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-16 11:11:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c02a0aec5bba255077aac94f67206926513a32c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 12:07:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c98b7b40e1e45c5b1190b630fc16c81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/297b4a6a18af4981b80f2dce73c63110 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b121183eadf4a2797cf7b7b7014d969 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3214d3394c244e3790de06be11a3aac2 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17b20b95f6694e24b0434d1addb256d8 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b2b9e0f0f7242a191fe2991c911afa3 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba6497095a9440149d7b8a244b42481b : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ce64c44fc0b4bfb85d0dba3b88e9d41 : SUCCESS in 1m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/89f4f6c462cd40d1b8337f3af3b376bf : SUCCESS in 16m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bf0efa639c6b45c0a1f2d11c9217753a : SUCCESS in 9m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/65084d5295a2495aa42e39230d2742b8 : SUCCESS in 9m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f7e38d7e176d431e8104463ff8e50efc : SUCCESS in 19m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/965460b2e9614b3e97c761aa3576c7c8 : SUCCESS in 30m 11s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9eaedb4906544c249356458c32f3c5d1 : SUCCESS in 2m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/01405d635d3e46428a864dceaff4d00e : SUCCESS in 6m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5b7582af5896474496495f836185b480 : SUCCESS in 4m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/001432d0198d49fa822522341e841788 : SUCCESS in 12m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e1538e9980994a2ba1cbf35710f60f19 : SUCCESS in 3m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e16935a2cc114d33978ee41bd09522d8 : SUCCESS in 7m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/995839f5b5ba4b54b37c5343250cc58f : SUCCESS in 8m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d442d109becf41a39eefc50a91e7642e : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bcc6e4da83eb4a9383d1f06ac3a70a4b : SUCCESS in 7m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4d5ee8abb2b1452e9ed89798cd5a880d : SUCCESS in 9m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/65710f6234d94915a7a2bfd1262986f5 : SUCCESS in 6m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9bbbc792d5264e01aba859ca0c33cea2 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a002b3e17e4d4062a5a3f8ba30fb0b4b : SUCCESS in 10m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/61230e712b784355a1061d0efbff6130 : SUCCESS in 4m 22s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/22cf9c47f17b4a17981c96f439ce16a6 : SUCCESS in 4m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0b59a4206fb04b60a591316db9614044 : SUCCESS in 12m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/071c67eb4e4440b29a61930592c0263a : SUCCESS in 21m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c02f4b4ac81b475b9d28da827ae56ad8 : SUCCESS in 4m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a0c5bd70c38a4a0da8817d2b70a4da5e : SUCCESS in 11m 42s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/22ce3f5a9e114658ae2ebb9336414252 : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/83414bba0a3147829ce45ff310fc2bdd : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7506e1a299a44061a581229657ca71ab : SUCCESS in 22m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8c6b60d8083146a6ab68fe5159ca3df2 : SUCCESS in 8m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aeb511cc0f464b948e4ffaf9a7476c07 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e488fed34bcb4450a6f46084ece884c5 : SUCCESS in 3m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cd185efe6e594cccb8ee10a0cf909c22 : SUCCESS in 16m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bfa251ea8774ab592c1dfcdb47b7f45 : SUCCESS in 52m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3df8415e515a4c6ca5b5633e1d2bce0a : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":3},{"id":"176ea28f21d56ccf3c9d8a5a194b8b5786058767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-26 02:53:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b8f420c73da5645ee8390f8b8e87203447a86ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 04:54:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff0a488421a040e688f4bf20d865b282\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70aa9c8a1f944eb5a7b8856b6d5558db : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b59b2f4a082d4d49a6fde0ed6e240dc4 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb4fa1f1711e4e26a303c108a9d81760 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6abff62ba4b4e188ed30932d0fdbdf2 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c29b06a830454d188c071e9205c42c45 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82e020ffd9104f4f95f2116468c294b7 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36d7d8758817463a9c64d48ef0aefbfe : SUCCESS in 4m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/67cd24c76f77456e89329d0586065e94 : SUCCESS in 17m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/71c6ac3cf2be41dea67f14a1375c0ec1 : SUCCESS in 11m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4ad327d871a041d4961bbc854decc7b8 : SUCCESS in 19m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2c68bc705a154f2d91231de95fcf67b6 : SUCCESS in 23m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/643686de54bf489895c19be6d24c4a3d : SUCCESS in 25m 28s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b3f6bbd996f74f71aacb023fcc7f1339 : SUCCESS in 6m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/53435b9ec755495a8a7bb9fabc4f51d6 : SUCCESS in 5m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/75edacbae2704db089b34ec436dd3129 : SUCCESS in 4m 34s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b4ccddf512d64bf98f5bb596dcf8f424 : SUCCESS in 12m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cd3abf630c3a4be7954aba222ae42b65 : SUCCESS in 6m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/64cd8ec5a77847e8a24a74d7e71982f8 : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5488bc1e05fa4e4592c7f27000d4aa4d : SUCCESS in 10m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d917db15d5604fd0a37b3c7bcb22d15d : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20fe3ac9bb8b45d88b9589a79c242387 : SUCCESS in 8m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ff25bcbcce60445b901ed5407637aeb1 : SUCCESS in 8m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/795c72de100a47e1877b3819c8074eee : SUCCESS in 15m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/797ff7281840475e97f6c709ab740e84 : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1ed5e88e0f6a4982823e80a060f9573e : SUCCESS in 14m 44s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f2d2f8d55df94d42bc9d3ef52f4bb5e9 : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bedb79bc48df46ffa386567d642fe82c : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9071e9d0940f465a8ab38e9d8875c2cf : SUCCESS in 10m 57s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0cc7122694b34565986700401f5a3be2 : SUCCESS in 20m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/359aee6cd8ef424b8dd9892bd100a6b2 : SUCCESS in 7m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c69efc4f99a84ceb88e1db92789ed78c : SUCCESS in 14m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/180d5d087ec24ace9c7e21a91b952e78 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e1a5315cab304a358e0a087bfb197ab4 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15d3b6f8a0e945199ff5dae1ff559489 : SUCCESS in 27m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/df2d20e754a34fa68ebd6190972fd28d : SUCCESS in 6m 36s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e40d292389fb452b884908bde885980c : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f59235d342a54a39b7e03a3243428154 : SUCCESS in 6m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b51fed90629c43d1838281df020e9bab : SUCCESS in 18m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b1396db02b6c4fb8b00ed7c0bd9d0999 : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ebce9d998db4e99bd561e395a97acac : SUCCESS in 1h 00m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/730fcd60d22d4031ad35b3765561b369 : SUCCESS in 1h 47m 01s","accounts_in_message":[],"_revision_number":4},{"id":"a164d2c0dcf9bac6d23cc13b2747813522f33f8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-27 12:49:10.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"71aef2b93d140d5ce66075c051b8393a795bd8ab","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-05-27 13:20:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"3952aabdedbc9a9a80dce09a728633d2f467736a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 15:01:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4cda3c4964046848270a43b5f10f2ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f3da2997add4bb8a83fcb3ab666ef71 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7b5a86a567e4161a14fdcc7a394d7f1 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c08f59e2e284d95a850f6e9ada03719 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e341958233d1497c8bebf5d7dc8c70f5 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e08852b95052409b8be0ff66c316b86d : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3b88f50980a4ac9a433e0b9c8ec8a12 : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd02da43783e4903a3d82dc5b449d0d0 : SUCCESS in 2m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a2f5d61e97034acdafbfff8545eb123d : SUCCESS in 19m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/94e80c539fdc47df9f8e022a0e5cf1aa : SUCCESS in 11m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/11da5bfae3fa4c7ca9ee3b748e70bcd4 : SUCCESS in 18m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/92cbc08b7e5f4cf4b31ce39fcbe7e07f : SUCCESS in 18m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/685f0a3fb0a94b489c724bffb38b719a : SUCCESS in 23m 38s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2b7b251571b64c5fbab70bc985fdd998 : SUCCESS in 4m 27s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/af468a285caf48178f9eb6d195e015b5 : SUCCESS in 4m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3093f1f00c27419daee7b785df49193c : SUCCESS in 4m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/00396b01578948508b29ea3e45e2deaf : SUCCESS in 11m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ede248a3d20d4492864bb384915892c6 : SUCCESS in 4m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9184f1834fc5499a9dd7dcaf9e9605fd : SUCCESS in 5m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3940a281eafd4cffb61a1bb1937b0cb2 : SUCCESS in 10m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ac18e35623774ccabfa779a8bad03af7 : SUCCESS in 14m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c2f9550fa824af0a8b6d327342f9a2c : SUCCESS in 6m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/990dfb91fe2645e2ae96d511c1cf49c8 : SUCCESS in 7m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2308f68b14f04cce82b3c8b1c5187e82 : SUCCESS in 12m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/268a43430ab6490aade13417972f08b2 : SUCCESS in 3m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d8007433b85042d1852aac5d5d6c7ca1 : SUCCESS in 10m 20s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/da16121ec4704c9abad19a7a2e82bb47 : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e31293c3dabb44ebb5368e40e94c8cf4 : SUCCESS in 5m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/db662c5a1b9f40e2a951db443ae3f499 : SUCCESS in 13m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2d6447e78d4144668c20a183ef8eff33 : SUCCESS in 19m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4ff1e8d367a849c992a13e501dfe2de9 : SUCCESS in 7m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f7a77f768d164639b54799e708968fbe : SUCCESS in 13m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e5ed74002df340edb637ec59f5831dbe : SUCCESS in 4m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/733875da38be480b82c39e9a3fe4434d : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9089c55e44f4f80bd42519f72034f82 : SUCCESS in 24m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6a8a9febee3f461fb0ee39d322f4f7f2 : SUCCESS in 6m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/218211180a0141309cea359107775b31 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/501cf82fa26f49f985c40ca2db87d34b : SUCCESS in 6m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e528ba06402e43b1bd321ceefe876b15 : SUCCESS in 17m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1ba286446c0a49eeaaf295dc2ac7b801 : SUCCESS in 6m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b872d202eb64431ca34b82f18654a394 : SUCCESS in 57m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aaa8065d4e714dcc8f86ff918c1512e0 : SUCCESS in 1h 38m 29s","accounts_in_message":[],"_revision_number":5},{"id":"33f394c333cd2b3b984bdc544a37babd77efdcf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-14 10:15:42.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f09aa9faf8ffb8734f19f18ba1857902f40e0a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 12:06:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d668078e86a41db97343b60ab810bb0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04815e7d9a8f4972a118634cd057ac14 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33caf669467d4caa94c0c27a4e2dee8f : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/720f1ac539a14050bd4ab25dade4bed4 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/575fda80016840a4aa6eb915bf22d4a9 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c38fb5e8ff244b5c82632fd8be3e0ba5 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d3a3b328dc64008a27167f510117c57 : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b603d528eff04552b875520e3036e586 : SUCCESS in 3m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6afc47aaa5b14d109d7433909ef6dc98 : SUCCESS in 11m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/482054e44a83436d91510834c012e4a4 : SUCCESS in 8m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fc54e6f889dd452686bf4b9c2d02bb8c : SUCCESS in 12m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/15f202da0f064c4b9c2894684fd928e7 : SUCCESS in 18m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/14a7b22907da4a8dbbc12c632fab7179 : SUCCESS in 15m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b26ead903b5e461baa6cd37d841ce1e2 : SUCCESS in 4m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d621168cc7034d98ac5a01645598dae7 : SUCCESS in 4m 21s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/bf24167777f643b49f4fcb44b044d3be : SUCCESS in 3m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1643a91e60ee4dc4926f028425d51309 : SUCCESS in 8m 26s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b651d6b981d5408195de9badad18cdb6 : SUCCESS in 4m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9cc43c0301a1423fa3b4e8018afeecc6 : SUCCESS in 4m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/22ca898d741a451f9f8e9b3d46cc6196 : SUCCESS in 8m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1fe390c626f943e6bdf1fb40350eda33 : SUCCESS in 12m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08a6a41ecd5c4e78842eab0bb738e67e : SUCCESS in 4m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/52c71afa633943cf9d12c5a5d9bc9328 : SUCCESS in 5m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d38b22ccb6f348e0b36776b31882f589 : SUCCESS in 13m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/864f337c5d8048e8b3642c630c0ec694 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c988d95ac424445ba975a331a9e1ceb9 : SUCCESS in 10m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/cc32333c8f344a57ba94e44fc6573d40 : SUCCESS in 6m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6cd0fa30f7c14d68aa532982d49999c2 : SUCCESS in 4m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c447da5e01154b3e83a94074d88bca39 : SUCCESS in 13m 38s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ad12f4f0277341959f073842e5daad51 : SUCCESS in 17m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e91502d517b3485ca88f276e5483157f : SUCCESS in 10m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/16808bfd27984e52882ec0b02d189d78 : SUCCESS in 7m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d3d0ea887e7f497787ca8b139d548788 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0289308d704f4584b66bdbb53a0410ca : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2338d829107f4d5191ddd3fff4f9745c : SUCCESS in 25m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7e840efb519041018ab5f2aafa18cffc : SUCCESS in 6m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0e54b53059854f179598b089363361dd : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3d0e444d2f344f13b11cd4ddff914794 : SUCCESS in 7m 32s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/952bba7dcba040ed9975f3854048375f : SUCCESS in 13m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/963a1f93f2ba47db9803926526dc4a43 : SUCCESS in 4m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6385c641cb94493eb13885d317bf5965 : SUCCESS in 53m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/915c7fe0d4ca4933ae28fb383b742492 : SUCCESS in 1h 45m 22s","accounts_in_message":[],"_revision_number":6},{"id":"0d7d2963ac5726d09cc1af9bc76c5d06365b6310","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-07-14 14:28:07.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"048dc2b0a7c4b4ab27a3314f3f33ce112b5e1b5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-20 10:22:27.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"f69ad5d5a7213b122f3c33bc9fc4eeae74c8917d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 11:38:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfea0128f165439a92b07b18ce9fa0d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087a6f4b9e98468688604d238761f723 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d0f28b9f8184af99b74a6c54765902a : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29a98572138044518f0b90d880007a97 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e58722e3f87544c384865e12670c4e33 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e52abe43220544fd9725b101b7dbba5c : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be036d63954846ee9e5ff213a57482a0 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/215ae60c0971401f85ee075471b156ae : SUCCESS in 4m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c6a4e97810bc4230853d9cf1609da2d8 : SUCCESS in 14m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7a4f48f656bf43b7990b94be3f0ba386 : SUCCESS in 14m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/82cc4abf3c2441488513cc9114c5c59e : SUCCESS in 20m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8292459abf90427290eccbf409aa68d9 : SUCCESS in 17m 59s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e04bac7923594c749af61568e467188e : SUCCESS in 26m 24s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a7f74c64488f46f9a964e7070a853e45 : SUCCESS in 4m 22s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/41a213431c984e9da426f5e7c81de30e : SUCCESS in 6m 28s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6c24781985344d9eb8b41e34a58f00fb : SUCCESS in 5m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8f68fa9992d94478bc7f240691df49f0 : SUCCESS in 10m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/088117f709ef4ebe9bc8657955a2a09a : SUCCESS in 3m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/290976b3cd4344d4b85930de7697b583 : SUCCESS in 7m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/611a6d0e340547e99e6cddb1c898d120 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ad83df60f99a468f9936a4778bf14a94 : SUCCESS in 14m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0246afcb6f0e45309eea37824d8a53fa : SUCCESS in 6m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bd0b3b1b39bc45b2a7571dba769d1caa : SUCCESS in 9m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/21e91e2b95794583a200613b5c3eae3c : SUCCESS in 11m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7a2736f496bf4fb995f7e883aecd9459 : SUCCESS in 3m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/93ae449b4ba24d7e865130242c795820 : SUCCESS in 8m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e45147e4a76f4b508daa7ae637b4e1f4 : SUCCESS in 5m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1b3b604f689c4ec28002943d75897141 : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c7ee8b5a6f324aacb24ae0b13ddf096f : SUCCESS in 12m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ee361d5eec17490a922ac4f7bb8c4045 : SUCCESS in 19m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e6ae93dbeaf64e718d9f4604706d7a65 : SUCCESS in 3m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/261dcc24756b4b5ea87a1cf3ef7a5001 : SUCCESS in 13m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5bbb8f5d24eb424b8329f26448ea49f7 : SUCCESS in 7m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0cce095dc43a410c958101e85f600594 : SUCCESS in 5m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1fc071ae662c424086f2edff087f2856 : SUCCESS in 22m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0f5d762b05e945a4a9f5d0f2e1c9efdf : SUCCESS in 7m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/65bc331527464e2dae29b452dd8c2572 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5041463a2fc4b04abcd3dc87b7e1313 : SUCCESS in 5m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/08675b676e3c452d9e7eb3db9ae98033 : SUCCESS in 11m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3598d447c73f4a689d37469796e91a30 : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00bbd2368aba41378a1bb5a86812048c : SUCCESS in 1h 01m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5675b92f3e7445f0a84630c32f02e263 : SUCCESS in 49m 11s","accounts_in_message":[],"_revision_number":7},{"id":"1c957251a9617c8856f1e78bd047849427caa16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-24 04:42:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d3234da96680bed27152d13a290bd3de0daf4b49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 06:46:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3fe0bf3bdd8467aaed35c924f9fc5f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5bbfe94c71348b69dd2f82851bdef38 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23af60d600914926a268fce726a8c7a6 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c4f5a2b617344209331cb58c76b0e6f : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e071f8548b04a158062b676ee167010 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11de4919a74a4b1eb45b46b1583f8001 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c876d9a7a8214e0d8098c41a85244eb2 : SUCCESS in 1m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb0859f6092e450eabead8b15ae87b81 : SUCCESS in 1m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fc14454eb64b436fad17e3dfff21d9d3 : SUCCESS in 15m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f6caa59990e2468a82e2d335857ef064 : SUCCESS in 9m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2edf7e2a2c9b4d8c9a0e2022bf0d7ad6 : SUCCESS in 15m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8c5c9e765a1644a1a7ca2b2970b1b246 : SUCCESS in 19m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1fcfa181712f4b2f8aa7a1c2bb29d872 : SUCCESS in 22m 24s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a892f5817f994b1181661da4d31fd013 : SUCCESS in 3m 49s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/89375497706a4370b51195a9832228a4 : SUCCESS in 6m 17s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d61c0a6d59634a6793360f466dc7a8d3 : SUCCESS in 5m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ccb1ecccb813428793332e73ee86c39d : SUCCESS in 11m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8d098eedc3b24355953a178a4707a642 : SUCCESS in 4m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ff70b056e61840a59c834c8667ed3e82 : SUCCESS in 4m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6535d4352d8b4dc291211afeaf620021 : SUCCESS in 7m 21s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2bc041515c4540b1af0d3a1423e8722b : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f9fa34d81964f5f9f2b379ad960bee3 : SUCCESS in 8m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d1c9828052da44c58b3e677c35f11a98 : SUCCESS in 8m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/79348885a7f54f55ad620e8beff060e1 : SUCCESS in 11m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cc3025948f4f4d75a1e6ee16d0762270 : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/629f0bc7c1c84c9f8eff636b37abe714 : SUCCESS in 12m 47s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8c1fea89a0f84f9491c9cd4eed071c43 : SUCCESS in 4m 22s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/91bef668ded34c618ca9a4b437f6d4e7 : SUCCESS in 3m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e80a1497f9b3416190a5ceaa6aac7fb4 : SUCCESS in 12m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7ad79b47b5c448879e7c84954953592a : SUCCESS in 20m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/10db35be77724918a97a711c72491578 : SUCCESS in 3m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bf96c9f1d8b84643b27b7368e5ebddbd : SUCCESS in 10m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/68e3e9d7a7144ad19677762a1388583a : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f7fb8d9ae004214a41404ed41c1ddb6 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e05fdd2be25e40e0b2eef2896197659d : SUCCESS in 25m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/75f31fcadef64216b0d2700e47c8127f : SUCCESS in 3m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5bfc0f41486542fba944b8b51767069e : SUCCESS in 4m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ef884cf8d72499a8a3eb7b0a2d55bbf : SUCCESS in 3m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c002e87b509a45648957a14b69cad70b : SUCCESS in 17m 03s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4865df5acdda4dffbb48e1b1ba91c3f3 : SUCCESS in 4m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18ac8dba8d5845e2abe27587ba576de6 : SUCCESS in 1h 00m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/628f92f0f7d04f91808d0686b547c0b0 : SUCCESS in 1h 56m 49s","accounts_in_message":[],"_revision_number":8},{"id":"80c388c5aee5cb8d87c05aa807490ea50f93d08c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-31 12:05:43.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"9426d8ba65df12e1137c76c31c2cd6f2adef8ad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 13:36:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1b8edf8ef6a4b1ca0a76f8aed4eb16c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02da7e643aba4000a26585b12633a33a : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d4db4707d5441eba1bea1f3f91cd4c3 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42a0b405882249da957b800eec1665ae : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99b3d71ce93c43c183348813e053d615 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3e31c1be8674cc19a15c627e2796850 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f10d1d50a844f3886a809f10a8d3c9d : SUCCESS in 2m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e949fb611d045bfb727cddfb766b004 : SUCCESS in 2m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/65f26b626e9a450eab629f0535720817 : SUCCESS in 9m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ad94548e94c340b693e477fb623b90ff : SUCCESS in 7m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a7d18ad6390a4352b5bbd8f6682d54af : SUCCESS in 9m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/abbd8c9adfd944dca6218e72b5340b04 : SUCCESS in 11m 04s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d0f9d81ef9b54aafba90c3e7c1610a13 : SUCCESS in 25m 57s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7e5f57a65fc84000aba338029d7ca46b : SUCCESS in 4m 56s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/027732a04a0046baaf67b34c2603b634 : SUCCESS in 4m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1217c6da03594ad8855d73c8b805534b : SUCCESS in 3m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1cabc52205b646dab3047fdde4c53d25 : SUCCESS in 11m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b22efdbb73524514b9bfcab231141756 : SUCCESS in 3m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a3a4035f60e045f0aef86422dcf5744a : SUCCESS in 3m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/dc873155d419457986f1b7d93c17bab0 : SUCCESS in 4m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7a238f7057b9489d958d217303752b8f : SUCCESS in 13m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6148fbd7265743f0857416734932962b : SUCCESS in 4m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bc8dad612b434add88a6174f547c9e0c : SUCCESS in 7m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/940afd1182d44826ac277a5faf69f49a : SUCCESS in 9m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/12b8879662724ad6a62ea5753fb7f746 : SUCCESS in 1m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bdf0fd7ba41c4ecbb3b03d883ccc527b : SUCCESS in 9m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fee3f79434c64aae8e72203b4657fd12 : SUCCESS in 5m 07s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/862fa01968c144689b319df707aabb47 : SUCCESS in 3m 54s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ec544e2b11704257879130e0946d3965 : SUCCESS in 10m 53s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7afc751b93d84498805ffef75b529085 : SUCCESS in 21m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cfe4b76bb2314e45becb913f573a5a2a : SUCCESS in 8m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/92291e33d89d4d90a3e2146f0674f51a : SUCCESS in 13m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/345b13d6e03b48ae9bec9bffce0cfeaf : SUCCESS in 4m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e44a5660a954597abe9bc12515216e0 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7806c582ef9b4dedaade278c9f07d359 : SUCCESS in 21m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/190564e9581943bc9c81785c95a3dfdd : SUCCESS in 4m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8136d2b9a75643849b1ceaaa7352493d : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87e41bbaa1664ef0ac15fed266812f8a : SUCCESS in 4m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/15067ec5abb442ba9ef81597fe353395 : SUCCESS in 8m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a2488083bb104b75971865dcc25be30b : SUCCESS in 4m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae97024948694a4b87cbf8b314f646ab : SUCCESS in 40m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85f421d5f9d340a2bbb2e6aa8c20b250 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":9},{"id":"3aa6aacfe1f16287cb049575763dfd5e5e18e7b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-07 15:19:05.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"94c9037dcb3d55863dcb9254bd353c43cf412a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 16:52:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/631d0a6b35f04fc1bf8c40caa3b47a06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ff3270f9fd45229ee2682c312c97ed : SUCCESS in 2m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71189562645d48779c411699ae8de12a : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1766483a3b14bf18ca7925335a9920d : SUCCESS in 1m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff4a816d31824c67b62e220699a7300b : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f55c158cf604a71b2830cf05d8a6d21 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af432877d1b0439283513f26eaba8f99 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c95f4916ca24e7089e2602cc319c044 : SUCCESS in 3m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ccb79ffa9214436b80575e57b92234c2 : SUCCESS in 20m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/537fea33ef4d404e8fbf5e3bd813e0bf : SUCCESS in 4m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/83051d9291074c1f8275754b829aa4bb : SUCCESS in 19m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/97a5ae861d3144f0ba6e08a6cbf53a13 : SUCCESS in 22m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/bd356ebabff14018a76eb172d31c47be : SUCCESS in 24m 53s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/95b1e9ffe7074a92a428b6e26e4438b0 : SUCCESS in 4m 49s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/566c2c9fba234997a52de57aecf6a216 : SUCCESS in 6m 00s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/59d7d744c47e45eb8b11b822a8f1b120 : SUCCESS in 4m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/80c94a456c25497499e2a80421ba4c38 : SUCCESS in 11m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6d969cfd24de48aeb2835e521fcf74c7 : SUCCESS in 4m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a0479365b9c64cdeb121b232e8c41c72 : SUCCESS in 7m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/311caf9824d84d7a948c35c763dbc013 : SUCCESS in 6m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3c3545b15a0d4177bed02475d00990b6 : SUCCESS in 17m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e8c8132208342b9b79223698305ff47 : SUCCESS in 8m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1802a9d39c854505a836baed10e72b68 : SUCCESS in 8m 27s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0ff7daec5fcb4615824f0c1dbcd8abd9 : SUCCESS in 10m 28s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/97522cd784864675b833a9a98c661e8c : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6c98e588dd2e474c8b8fe0640332fd4d : SUCCESS in 12m 12s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/25b80865f6ed48f89a4c46765d6582c7 : SUCCESS in 5m 41s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/47c8a14b895b45fbbe0b882cc6f40f66 : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7878efdfb14b45f6a085fe13aafc1df4 : SUCCESS in 12m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d8df58e386474e7da9b51f248953be0b : SUCCESS in 21m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4578ed04d7f04671b57d4c75b953853d : SUCCESS in 8m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/eeefa68a853a4bfbaad9c91219550f1c : SUCCESS in 12m 15s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6a1333d184f54cfd8b352d0c78df588e : SUCCESS in 6m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/26e52f1a0ffe4030bed45d892b84990b : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/801e74b843b84a9f81e8aaf92b5f6291 : SUCCESS in 25m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f50006ab4e1a4f3aa529f2ce43647799 : SUCCESS in 6m 52s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e239e166f5314a8ebaf983f90762ad56 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b528715ae354bd8af623cf874d8a6aa : SUCCESS in 6m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/710d319d82ad477d90388379c23f7b9a : SUCCESS in 16m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/827af3edef7b49ddacd95b51007e75ab : SUCCESS in 3m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1b216ad080c42acbf8880339c197c3a : SUCCESS in 1h 00m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a137d721fe0741078d49c907ba7ea7f6 : SUCCESS in 1h 29m 48s","accounts_in_message":[],"_revision_number":10},{"id":"79ff4234d4e53e0cbc1e232cb5dd1566f18cea75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-29 04:27:52.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"ef2b1e18bc3c42723d029a242f03982d21c171c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 06:15:46.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/207ca08e2d2440a29914e1bbb7431197\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b5ea6bd5f949aab2beee3380d75950 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc509c6ebf24d4f939ab80ba39f49ed : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f5ca8a4cbd849cfa0177d853594c0e9 : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/621e31ad55ca4a97aed19a6aa3aace1f : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/425c17e5ea894acf8fb8043a8c0157b2 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c83f51f0f6204b579d9422111a663e36 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53602b1ad7ba43e69be821b5da4df02a : SUCCESS in 3m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b9ada9d12edc4510b9cc303dee4cc61a : SUCCESS in 7m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2f6c2c0f3d1f4b3db68aad35aae9675c : SUCCESS in 19m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/68373d6834484000ad0fe7da9d3f241e : SUCCESS in 18m 46s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/71480c1a19b649ec9778aad96cbbfa24 : SUCCESS in 22m 56s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/83d9af4197d547ec89bf1bbdc445d418 : SUCCESS in 5m 04s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2344a36a97b6485093a96913e84a3950 : SUCCESS in 3m 48s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2b7f1f95b763405d9792f46445ef1d2d : SUCCESS in 3m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/80f1578a1209455289cebaa34eb7ddf4 : SUCCESS in 12m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/065566e254044867b2b29b70340ec546 : SUCCESS in 3m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/680b702cc6e144a2ac77fd505cdb5648 : SUCCESS in 6m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1ef0fcde1f544a7890c7b163700c828e : SUCCESS in 9m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6fa689a3ee44444daea550931f70fa58 : SUCCESS in 17m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e6150325d494681ae3e06d6326dad5b : SUCCESS in 8m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3d67811f87704500829a4226dda0ba6b : SUCCESS in 7m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ae703ea5642e424e8baa5fd1781d1874 : SUCCESS in 10m 21s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/69d4999fb2414dedba4ae19fd4e6ade3 : SUCCESS in 3m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6f568e0b1da94a6b938058238a6f8bf7 : SUCCESS in 11m 37s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/51421125eb8f4bf4a7757df1aab3de8e : SUCCESS in 4m 40s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3b05808690e147a4ae576cc7511a1f72 : SUCCESS in 5m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4dbcd2f7ecb442beb55af0ef194fe1e0 : SUCCESS in 12m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/279cee3ca9804dc5b45c9c4247507ff6 : SUCCESS in 14m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2a4ccc43cc05449f86f48ce0dda7eb9f : SUCCESS in 9m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/965dd3c9181c4f03b28a6e622c49e5b1 : SUCCESS in 8m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/10eaf579a7d54f0585417a4c61d6c924 : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a99f65d37c0b4315a465de59da6e2132 : SUCCESS in 2m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d363dd3158484a6f8da35be398b6ae5d : SUCCESS in 24m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/66948fba66854dbf858dcd54c8047543 : SUCCESS in 3m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f4aabc927a32432ea6921e9f5d9f1ffb : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ce4a64d80c24082b0575f4a18d97bdc : SUCCESS in 6m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8309e19acd4947b39b11bb4b6980e6f0 : SUCCESS in 6m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7c0e39c990f94b82834093c0eae6dbe0 : SUCCESS in 3m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae37a1b12fef4c098f83aed63c9d4d70 : SUCCESS in 31m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c23b9dcf99241ce8fc82283d57ed9a9 : SUCCESS in 1h 43m 50s","accounts_in_message":[],"_revision_number":11},{"id":"8730fda85e1a970743585cbbcb539ba3a341a8e7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-09-11 13:19:36.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f9b4071fa4a1eccc2c4b4f0c2c03e4c6d8abd3f5","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-13 23:16:22.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"f860b249c01cd920649fa5aeffc5c027486aee04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 23:16:55.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"f61f7446c342ae1ed1b59b7dc5f3545a69a5f179","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:33:50.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee64535c2a14b0780a93994fefb91d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/066c7108aebc4ad59da5379895b5683d : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cf3b4d25c1c4267b66255c6b6ed9dbd : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e549f2d0adf94146abc8d4512938f6d3 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd9c31db5fed4c9eb7116392b2300902 : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91a595e862da4db0ad2669954736c08f : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2cec1337c9174171a012b904f218ccae : SUCCESS in 3m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fefa6ce0f8624c7ba3af57bcb74c90c8 : SUCCESS in 13m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5b44be86084845b8a46997600e1ae60a : SUCCESS in 18m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3cbe0e852cd74748855f21fbf52b18cc : SUCCESS in 19m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fb011ca9edf344368905bc22581a8986 : SUCCESS in 11m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f5917fc1fbf54c1286eca893343b1867 : SUCCESS in 4m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b38e70a2b9c9449482e3800624301d80 : SUCCESS in 5m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d260d5cbc0fa474ebb7a61aa83437223 : SUCCESS in 6m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/79ff171b5be04aa3a74326f6e7ebf2e4 : SUCCESS in 17m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/349f9d33e0bb44baa5dfc382875a0368 : SUCCESS in 6m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/15f9dbf0731245c89511803c1a6566e1 : SUCCESS in 8m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5280c7b484ab456ba69341f86336b53b : SUCCESS in 12m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7b6f49d028384855a54f8b25d8b14864 : SUCCESS in 2m 35s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/80b967623981406bb0d5a27c59dae03c : SUCCESS in 9m 55s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/36e8fc3bf2e44643b8d874340b6a463c : SUCCESS in 25m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2ed75345c24846d3b0766ea2eaa4017f : SUCCESS in 7m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e2c6fc892f4d4f62a9180d596a19b879 : SUCCESS in 16m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/60601bb721724b24bf6a12e4b10096a0 : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac180499812c45dc8de892ecd468c5a9 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef4fa31167e04d8e93123f430ad03e64 : SUCCESS in 26m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/09bf05d4022745a7a13c5fa049402444 : SUCCESS in 7m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4ba0555719bd4c5cb7571e8de7a77754 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2dd606edf7e847ea944cfb2f8d73fa4f : SUCCESS in 5m 14s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c23e6c96eeae45adafb4b604ecedf53e : SUCCESS in 19m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f905207260a24d1fa37f11ba8191216f : SUCCESS in 6m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d021937592d4b2da1e093fa9fd2c762 : SUCCESS in 1h 06m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/302d0248bbe44c3982879255bf8543c3 : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":11},{"id":"85029a0513110d0de6e053353de6a6688264c3b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:33:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"8ca86492ca8c17e9986703843beb242dd8bc54d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:34:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb289f91e2d046aea7809e7a9450ae91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da7ede82ade4431846f19e9d6dca99b : SUCCESS in 33s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"574d98e59de6faeab5bd2aee9e676961f6975f38","revisions":{"86880d5b6b72fd108020bfa6d485dbc8669dc9ad":{"kind":"REWORK","_number":1,"created":"2025-04-05 16:05:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/1"}}},"commit":{"parents":[{"commit":"aea4bdb03846d4b08c0b3decf0ef6dec618a14ad","subject":"update constraint for os-traits to new release 3.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aea4bdb03846d4b08c0b3decf0ef6dec618a14ad"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86880d5b6b72fd108020bfa6d485dbc8669dc9ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86880d5b6b72fd108020bfa6d485dbc8669dc9ad"}]},"branch":"refs/heads/master"},"24c1b9e7f1b8a6d964f85f7d4c3223c2507d11c4":{"kind":"REWORK","_number":2,"created":"2025-05-07 13:29:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/2"}}},"commit":{"parents":[{"commit":"b58f4867d2406854e476fa599ca52d270aa2cd19","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b58f4867d2406854e476fa599ca52d270aa2cd19"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-07 13:29:04.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24c1b9e7f1b8a6d964f85f7d4c3223c2507d11c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24c1b9e7f1b8a6d964f85f7d4c3223c2507d11c4"}]},"branch":"refs/heads/master"},"be802f2ba7d2d195ed5413dc143d4bafc24aacc3":{"kind":"REWORK","_number":3,"created":"2025-05-16 11:11:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/3"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-16 11:11:33.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be802f2ba7d2d195ed5413dc143d4bafc24aacc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be802f2ba7d2d195ed5413dc143d4bafc24aacc3"}]},"branch":"refs/heads/master"},"2c38c04af9eeb4872de9d67684696d3fb984d93e":{"kind":"REWORK","_number":4,"created":"2025-05-26 02:53:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/4"}}},"commit":{"parents":[{"commit":"0697f59bec1402466cfc71e33529293ac1739e72","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0697f59bec1402466cfc71e33529293ac1739e72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-26 02:53:27.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c38c04af9eeb4872de9d67684696d3fb984d93e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c38c04af9eeb4872de9d67684696d3fb984d93e"}]},"branch":"refs/heads/master"},"bfe6d1df58c93a09f65962d8d0115f0cc093b9a5":{"kind":"REWORK","_number":5,"created":"2025-05-27 12:49:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/5"}}},"commit":{"parents":[{"commit":"ec76c273008e3317426231cdad782622aec6e265","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec76c273008e3317426231cdad782622aec6e265"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-27 12:49:01.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfe6d1df58c93a09f65962d8d0115f0cc093b9a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfe6d1df58c93a09f65962d8d0115f0cc093b9a5"}]},"branch":"refs/heads/master"},"d2a13c40264d99214c6bca0328464b937e1bb077":{"kind":"REWORK","_number":6,"created":"2025-07-14 10:15:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/6"}}},"commit":{"parents":[{"commit":"1f2be26ab5a4d601127817b90c794e61cc40bd12","subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f2be26ab5a4d601127817b90c794e61cc40bd12"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-14 10:15:34.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2a13c40264d99214c6bca0328464b937e1bb077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2a13c40264d99214c6bca0328464b937e1bb077"}]},"branch":"refs/heads/master"},"a1ebf632d9fb1adc24798abcb0b4d8bc686aabd1":{"kind":"REWORK","_number":7,"created":"2025-07-20 10:22:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/7"}}},"commit":{"parents":[{"commit":"d78665852a59c1b18559c53ca11fbd9b672a0fdd","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d78665852a59c1b18559c53ca11fbd9b672a0fdd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-20 10:22:18.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1ebf632d9fb1adc24798abcb0b4d8bc686aabd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1ebf632d9fb1adc24798abcb0b4d8bc686aabd1"}]},"branch":"refs/heads/master"},"37219eb674f6a29dda029c28f62e9d18a1469148":{"kind":"REWORK","_number":8,"created":"2025-07-24 04:42:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/8"}}},"commit":{"parents":[{"commit":"8d1a0ebe4d9eff55ddca5bfde6b4943f3d4dfd72","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d1a0ebe4d9eff55ddca5bfde6b4943f3d4dfd72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-24 04:42:31.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37219eb674f6a29dda029c28f62e9d18a1469148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37219eb674f6a29dda029c28f62e9d18a1469148"}]},"branch":"refs/heads/master"},"ba8fb02d5b751fc0ddaa9163980fffb3d78afc3a":{"kind":"REWORK","_number":9,"created":"2025-07-31 12:05:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/9"}}},"commit":{"parents":[{"commit":"2b2b12da983a9f3a0ed6559f5d3482e9f85b9a98","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b2b12da983a9f3a0ed6559f5d3482e9f85b9a98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-31 12:05:33.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba8fb02d5b751fc0ddaa9163980fffb3d78afc3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba8fb02d5b751fc0ddaa9163980fffb3d78afc3a"}]},"branch":"refs/heads/master"},"bfbe7cd0983e0f7e8398e5c23d7d156fe76d3be6":{"kind":"REWORK","_number":10,"created":"2025-08-07 15:19:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/10"}}},"commit":{"parents":[{"commit":"6324b28b1bfe61eace77bde3978daec14ed3d176","subject":"Merge \"update constraint for futurist to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6324b28b1bfe61eace77bde3978daec14ed3d176"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-07 15:18:57.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfbe7cd0983e0f7e8398e5c23d7d156fe76d3be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfbe7cd0983e0f7e8398e5c23d7d156fe76d3be6"}]},"branch":"refs/heads/master"},"574d98e59de6faeab5bd2aee9e676961f6975f38":{"kind":"REWORK","_number":11,"created":"2025-08-29 04:27:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/946500/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/946500/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/946500/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/946500/11"}}},"commit":{"parents":[{"commit":"91d4d40cda062c036d4e42390e5d9d032153a6f0","subject":"Merge \"Bump pyroute2 to 0.9.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91d4d40cda062c036d4e42390e5d9d032153a6f0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-05 16:05:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-29 04:27:43.000000000","tz":540},"subject":"Remove unused pywinrm","message":"Remove unused pywinrm\n\nThe library is no longer used by any projects.\n\nChange-Id: Ie550cca55e7c5d180701b2b8c2c3eb8752694e19\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/574d98e59de6faeab5bd2aee9e676961f6975f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/574d98e59de6faeab5bd2aee9e676961f6975f38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
