)]}'
{"id":"openstack%2Frequirements~946543","triplet_id":"openstack%2Frequirements~master~I1c3b9dabcb5ee9c636bc23aba8c9b304895cf0b8","project":"openstack/requirements","branch":"master","topic":"pyroute2_0.9.1","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-04-23 19:05:38.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-04-09 15:41:59.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I1c3b9dabcb5ee9c636bc23aba8c9b304895cf0b8","subject":"Bump pyroute2 to 0.9.1","status":"MERGED","created":"2025-04-07 10:31:16.000000000","updated":"2025-04-26 11:18:08.000000000","submitted":"2025-04-23 19:05:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946543-pyroute2_0.9.1","meta_rev_id":"18e5201973d991978e829f52745d49cce1ff3e7b","_number":946543,"virtual_id_number":946543,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-23 19:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-04-22 10:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-04-11 17:53:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-04-23 17:43:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-07 12:05:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-08 05:31:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-04-10 13:54:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-04-11 17:53:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-04-16 08:55:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-04-16 08:55:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-04-16 08:55:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-04-22 10:35:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"7b3f9bf64fb1eb67de25dfe9e1f27b8dbb365f87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-07 10:31:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae78169c73f4e7b54e9722270f152a6fee7bb575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 12:05:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83f272848be548e0b8e668fb6e833d28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c5535f022894fdf87daafbd3b1e7982 : SUCCESS in 1m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f029d8fd4b804e66b2c6f2b8919ec5bf : SUCCESS in 1m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d746e5a2c1264afab8b99c928daee93c : SUCCESS in 1m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16081415d5c04d4a89476a049c704e02 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d93db64f9d04b9d80189b11f0a2aaf4 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81b42f346dfe4bfd8402da5ef467a645 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/27e7057b9d484ab88f62d1f75509b49f : SUCCESS in 1m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7d67bdd036eb49e7ac01d40374775cfc : SUCCESS in 12m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/08130f66cec14930b035a1903666bfe7 : SUCCESS in 6m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ec369bdb2df04920aa536b21f831532f : SUCCESS in 11m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c297209bee6748e1b688f328aec5b3ed : SUCCESS in 9m 41s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/098fe8b04f484262a5c814fcd53bd357 : SUCCESS in 4m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fc06e06705564fd18e7935d0af29032a : SUCCESS in 4m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/74b3a5dbb99c4da4b3dcc573632bbc46 : SUCCESS in 3m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5e09e584f59a4c299e8764f1e9708e02 : SUCCESS in 9m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6c699895a7e44950b1e759730fb399ad : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f7243397e9a949de8918210cba304a9c : SUCCESS in 5m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7aee41b9acba48958f8f7c62f177eedd : SUCCESS in 8m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0110f43423f94f599cf85e1f4240094b : SUCCESS in 14m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7aef6784de73469c8c58d4dd5cb52971 : SUCCESS in 7m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4060db450cf34807a24c21aeb8d7d73a : SUCCESS in 5m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/aa8dd60a730648eaafe9840778e176eb : SUCCESS in 11m 26s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/695a68bf80104147859c3182f1613dc4 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c77226f5a52b44d396d1f2cf56bfc3d6 : SUCCESS in 16m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ec71345ecbb8454884929e72f3ae4f0a : SUCCESS in 3m 49s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e98e266e6a12439e85e33ee551e9224a : SUCCESS in 4m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5831454131c144a382c5393921c315de : SUCCESS in 17m 19s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0bc04df3d47e4a32ba0d90fe0f5b6e72 : FAILURE in 3m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/102ea42c39c841d0bb59f27e133fe65b : SUCCESS in 3m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0b0eaab1afe24c38b87e0e3a1eb7b921 : SUCCESS in 8m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/00743200c569443f96bff302a2e4ae1c : SUCCESS in 2m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c4b30847628f40f1af87346c1f2cb2b9 : FAILURE in 2m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/666d715740aa46b3a9a36f989233b610 : SUCCESS in 20m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a263f5aea6544c83ad8e15f0d3cbb7e7 : SUCCESS in 3m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e8d116cb03d84533b8fc2fb13340e736 : FAILURE in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9aef733f6c46424caaaa64398bd1ec07 : SUCCESS in 5m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/69afd11559a744cab4a118ac75ad6b6e : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/969644a4a16f40d89027367e43fe38dd : SUCCESS in 54m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7c89edd6c6148c1b2163056d272dfaa : SUCCESS in 1h 33m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8ec77ec4e0616b76f9be8d736fafabd2ed6dc1ce","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-08 05:31:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c4fe53e701f2548a466bb9f8ee4edced92fc60a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-09 09:39:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c851582de2e0f12c7f0b8a2bf60e852f71dba67","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-04-10 13:54:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a30d668f921c3254a5ad574bd627e772e0b3531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 15:19:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c4b5c4475544d5298a025de13bb6a9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df6382302b2f4d11b9f6249a5ac52d80 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36af336ef3d24cd6ad22b585651d83c2 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2daec3e8c4ac4c04ac56df4e1ace789c : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7c468e44a8a4251bb8fe210b2630855 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/99df6cc7277c4979a9192575e777a64a : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86371b8790954909967b9ab776306d7a : SUCCESS in 2m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96745877205c4ab5901b81738279b185 : SUCCESS in 3m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/194d8ade646440ee8dd2fb6695cd5544 : SUCCESS in 6m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/964a96cdbfad478e85df6653a27273a1 : SUCCESS in 8m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a13f267be216400e80c67a3eee5487d0 : SUCCESS in 15m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/eebc170ea9f94dfeb0bec59bb3bda3e5 : SUCCESS in 8m 56s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/43242fe2589040efa81525660fa7e7fe : SUCCESS in 4m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/045b69bb6893485a83114144727b483e : SUCCESS in 3m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fcbea0468fc546e285250a16520be562 : SUCCESS in 4m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/65d0e4bde4c14047a797a8dba20bdacc : SUCCESS in 11m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/90ddccccde5046adb65183eaac0a4020 : SUCCESS in 5m 34s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2de0844a288f475c80a8f33c68c0a749 : SUCCESS in 5m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/728a7339121c4b21a7727128a54a51fe : SUCCESS in 8m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5e81c6572c7e4add916940c119cb3ed5 : SUCCESS in 14m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a80be5205830435dbdd5f05f6d4e7b33 : SUCCESS in 6m 03s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dc6f9b310f3940c38874ed5884c498eb : SUCCESS in 8m 27s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d1beac777202418a88cfab8c44947a77 : SUCCESS in 12m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c04d4b40528141fc9461fc6aafb56a5d : SUCCESS in 2m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4d5e9086f60c44f88c28f7ce931186bf : SUCCESS in 8m 55s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ee859a4d713e4ff78121715bc47fc625 : SUCCESS in 6m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b053a07d71554a58801fd6e261077f6d : SUCCESS in 5m 15s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/94f1c15158164c4a9da8cb56250dcb59 : SUCCESS in 11m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/974dfd6c01194cd380732e47074e8d2e : SUCCESS in 19m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/76ea43d308c747df95c6495c4ee2affa : SUCCESS in 7m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/663d8a1f4798478c9030bbda46516513 : SUCCESS in 10m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ee818938d7c54bf0a0b4a5947afcbf5b : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bfefbe47a20f437995e73b6056942696 : FAILURE in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d4cc51bf1ac547d79539acf599de5bed : SUCCESS in 20m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/55d9ff72b17b4bea9a53ef0df2dcae73 : SUCCESS in 6m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f21036907f6f48fca7bafa1adbcc6a59 : FAILURE in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7db59758111424f8a0c882f4dcec852 : SUCCESS in 6m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/017b0b90889345aba51d880bb9b2f11a : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cdd2643b6b84baeb99f2f5dd40da3ce : SUCCESS in 48m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbf8f615e9084b6f87112f4eefb9f78e : SUCCESS in 1h 22m 49s","accounts_in_message":[],"_revision_number":1},{"id":"907afb248ee76699d95de84ff39f23aeb0147f0b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-11 08:56:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c691c720fb97069a52c786043b9bbd63a10b003b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-04-11 17:53:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72915fde2117754b5ae10a45930df3e4bdb23d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 19:00:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e79bd7bf77594b93b1e475ce6fe138ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c62538d53bb4e4b83e73baf8b92d00e : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe13efbdf4b49c39d99ddc1330de380 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99d4c567d88d45dea8f9a9cfc1867432 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5572f1b8375f4da7a51be42eb1668a39 : SUCCESS in 1m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/149b7a8b8d4d41d0a5d7697425c2d5fd : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d85f1a1a2d54e37abd32c47ad5f8079 : SUCCESS in 1m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68a645cd7b4f4a869a54ed166896f16e : SUCCESS in 2m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b360fc96895642f39589523cfdc4f639 : SUCCESS in 6m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3b8602a47ce64b52a321b813ed4eb689 : SUCCESS in 6m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f69c3a6ec6fb40c2b0429f1c04f03ca6 : SUCCESS in 7m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d2db91baabd24839831266f9710021e5 : SUCCESS in 10m 37s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/dd7148ede0d541d9a09215e053fa8255 : SUCCESS in 3m 12s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e651f6e4c1a448689b0a1d5665640edf : SUCCESS in 3m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/314507cdde9b4bd9b108e8249b71c0e6 : SUCCESS in 2m 42s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a1afe0cae9844b759394f4ac72f0e3b5 : SUCCESS in 9m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9033c71d01c34f8ab8e2d04eff855d8d : SUCCESS in 4m 48s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/53ca6a5d2f45490182d6a21e8cec5911 : SUCCESS in 5m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fbef8f72e225471ca6a494270ce1ec04 : SUCCESS in 7m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e153309014ac44babb237be23a1d2c01 : SUCCESS in 16m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15ee703badeb47809830b81a8fef8772 : SUCCESS in 6m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d0d6a43b27f44bf8b5ba5a49bcb06976 : SUCCESS in 7m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d477d80116f848dcb4d019d95df3b7eb : SUCCESS in 13m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/bcefbbf636cb4e06a68240965e29939c : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b6a83a616dcf404ea21926ddc3e17879 : SUCCESS in 10m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b79ae617337a41a38ecd35e99694a8f2 : SUCCESS in 4m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/18f10e0525f04410b545e3fd6e9c7f9e : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f093885122e34d2ca1e545787bd31562 : SUCCESS in 13m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/746589d8f48a4d42bc0fc9cc208c3087 : SUCCESS in 15m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3414b2e7e52d468a9f17f86bfdc4b13f : SUCCESS in 7m 13s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f93bd9707f0049f3a2fbd5ad8801f525 : SUCCESS in 13m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/077f88e5fdc54fceb6ac7d138e2e7437 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee7497f9c3604111a5d4e14d68163c49 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4123748cd43a4721b5492f9901721da9 : SUCCESS in 31m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a94ba68830fc45d0bc36a56c553d89a1 : SUCCESS in 7m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4ca2f19e008c4f91925b8731563c04c7 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/650de3ac2ef84eda80df7eff8129ae06 : SUCCESS in 5m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a9c2ff7cbbe34907b752a4862892c016 : SUCCESS in 15m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b05546e7b8f417ba8db9a30efc35ba3 : SUCCESS in 53m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d509c010fcd4680ab9a87c75f3f1712 : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":1},{"id":"24998b9ddb268b69f0de0e70c2e96c14ab0e8c32","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-22 10:22:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ba4be180af96261962365e70aa0e8c6fa6d1397","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-22 10:35:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"765882a3dd14d06af37e6636a5aed3167440b39d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-23 17:43:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1963e4ebc404e3e8a30740e1058dd605339f070","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 17:43:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dfb3c79741f74d9423be855c05b7369ac009a33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 19:05:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/177329ee64814858846bb35a1a1a23d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f5c81ff683477f8d75467814dd8f12 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b4cdccef10444ff9ca80ba232e9fa9b : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85dcd1c2be894e5680e7c1fadcb0c19d : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8fd46b3357b4da58235bbf2d02d87d1 : SUCCESS in 3m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c8ecf5ffef64e43bd8e8d23eb2863fa : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d17c9ad48ee64d9db6e8ca9fa2406920 : SUCCESS in 2m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0c75adb641fe4e349c189e5b85ae55e9 : SUCCESS in 18m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/138e94f8096e4210ae54e9e1a5557c18 : SUCCESS in 7m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c890e004fcb94d22a1fddad31b2b4f9e : SUCCESS in 15m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d38293e4c3d043f6ba42a4364cbf4d77 : SUCCESS in 24m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/50ca724eb9204b80826c069532505beb : SUCCESS in 7m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7a2b584f56cb45af9e483b2d807e01d8 : SUCCESS in 4m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/87f6de5b285346f38a41dec1b8de0391 : SUCCESS in 4m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/04ae685be9a543898961daac065e2c27 : SUCCESS in 9m 51s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/72a3f6ad6f6a44a4a5ac4b3fe1455b8f : SUCCESS in 12m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5717a41c2b594846a7e5bd890d1239d3 : SUCCESS in 5m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ddaad27a52cd4729816108c1f5f35d41 : SUCCESS in 7m 31s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1939d8c5d91a4f4a885433fb200b67e3 : SUCCESS in 11m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/742c88c8bd014254a2fdc81f3101996c : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/76e4600fb45e4ac6af100c90746261a2 : SUCCESS in 11m 24s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6fc619430d8a4357a75716ee230d420a : SUCCESS in 13m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e17c90e7be6f4e4bbb1ce45b60ec6101 : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/47c9c8d04ec749baab0b2941ca515b57 : SUCCESS in 11m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/cdf8aa235c7941fdb2e527efe65dc702 : SUCCESS in 2m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd16a450a94945d7ab80ed89ed8a11bf : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b6cb39f83394041aaaca2f27a66868e : SUCCESS in 22m 55s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a029ef56bc3747a59d406b2dba5b2e2d : SUCCESS in 6m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/09bfca53f7404cb0b0110696726e5d18 : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ed3873d2a893469d9090fd3bf9bd44d7 : SUCCESS in 5m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/978af53a0c984b5fb7e85c760001ec58 : SUCCESS in 15m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd777ac94936437aba39f64353102eac : SUCCESS in 58m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ffd262346f94b82955595b09d9d989c : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":1},{"id":"13e7b2ce3bd6adff7c3459f4d9697fa0f2230d70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 19:05:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"339b69a939de452ca4e64c7c9790abfe1903eb24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 19:06:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2964b9974044e7087ce09bb83c337c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e975f0b4c147adb3d3ef123de9420b : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"18e5201973d991978e829f52745d49cce1ff3e7b","tag":"autogenerated:gerrit:revert","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-26 11:18:08.000000000","message":"Created a revert of this change as I072cf76654e7fde57a0e19a60392da7e472bff6c","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d01db4879d64760dde42763166907dd72b3e968e","revisions":{"d01db4879d64760dde42763166907dd72b3e968e":{"kind":"REWORK","_number":1,"created":"2025-04-07 10:31:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/43/946543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/946543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/946543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/946543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/946543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/946543/1"}}},"commit":{"parents":[{"commit":"aea4bdb03846d4b08c0b3decf0ef6dec618a14ad","subject":"update constraint for os-traits to new release 3.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aea4bdb03846d4b08c0b3decf0ef6dec618a14ad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-07 10:29:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-07 10:29:01.000000000","tz":0},"subject":"Bump pyroute2 to 0.9.1","message":"Bump pyroute2 to 0.9.1\n\nThe version 0.9.1 was released on April 2, 2025 [1]. It requires\nPython\u003e\u003d3.9.\n\nThis version has been tested periodically in the Neutron CI, using the\nmaster branch.\n\n[1]https://pypi.org/project/pyroute2/0.9.1/\n\nChange-Id: I1c3b9dabcb5ee9c636bc23aba8c9b304895cf0b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d01db4879d64760dde42763166907dd72b3e968e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d01db4879d64760dde42763166907dd72b3e968e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
