)]}'
{"id":"openstack%2Frequirements~946898","triplet_id":"openstack%2Frequirements~master~I2132d2d08b3ec0e6c3fa1fae4c7edfb607436da2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2132d2d08b3ec0e6c3fa1fae4c7edfb607436da2","subject":"update constraint for openstacksdk to new release 4.5.0","status":"MERGED","created":"2025-04-10 12:36:32.000000000","updated":"2025-04-23 21:56:03.000000000","submitted":"2025-04-22 12:11:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946898-new-release","meta_rev_id":"9571e038afeac72316c379472c7ea885b38ddfab","_number":946898,"virtual_id_number":946898,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-22 12:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-04-22 10:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-04-14 12:20:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-14 20:01:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-04-22 10:44:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-10 14:02:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-14 12:20:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-04-14 20:01:04.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"REVIEWER"},{"updated":"2025-04-22 10:44:38.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"4f35d207d23cdd2393450cfe21a0b819e9ce55ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-04-10 12:36:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"302e8c11743e9395d97cc0d7c6a34d57380810e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 14:02:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b4b985a6c547c5b24766395d346441\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916678755fe149e3bfd02c60b8d0cac8 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2127f53503c34bf4b55297a869725f92 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4aa85094a96249ee83a4ecf587b75e5c : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6733401d6eb4cc28ff4613ce364a8f0 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88ebd33c002b4420a4cb4ca9623f9818 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23490a1f9fa447e8a55ef7a0e8a0db6b : SUCCESS in 1m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21e97cb2341c49858f9098a24be1efad : SUCCESS in 3m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c70e808366bf4b978f60d8ccff019425 : SUCCESS in 15m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dd958dd483eb4271935af803c5176a9c : SUCCESS in 8m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/088ad9f17a0d4a618ae34c60738eb6a9 : SUCCESS in 15m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ec902106870948b4940e392497f0a6fd : SUCCESS in 24m 38s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/baa4c9650dcc4e10a36e06a10a22c43a : SUCCESS in 4m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3064de13a23d4ba19cd5279745d78655 : SUCCESS in 5m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b1a879ba3b0f44b88b02da9a5f48dc3e : SUCCESS in 4m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8e9c739971774cb18ecbfbe2a998b986 : SUCCESS in 10m 44s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7d2ca98759dc4701aefaa72731793a1d : SUCCESS in 4m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a66d1a981cdb4ce8bb8d183a975fa064 : SUCCESS in 4m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6a7b4c896c5745f2874a4a15e9a3091d : SUCCESS in 6m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/151c31d25aa04356a2703fe5d45a6287 : SUCCESS in 13m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c38cfe0fc0074cc3ade4fe8740f353cd : SUCCESS in 6m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/89556e3baf9f4aa39f26d4d069689269 : SUCCESS in 7m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cb5e982facf341da950e193f1f12dfbe : SUCCESS in 12m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/02f153f286f14d518877a53d00addca5 : SUCCESS in 1m 52s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c63518a74b224fa4b806b39abdbc9691 : SUCCESS in 9m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/30f3f162462d4ca0b275b78b8ecaa650 : SUCCESS in 4m 25s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7325fa11be274089bf6ba7cf6d8f21e8 : SUCCESS in 4m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/620503ad45a14d1ba3455409017a65de : SUCCESS in 12m 25s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c355ae270dbc4aa1bf3db4d9acbaba96 : SUCCESS in 18m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/09fdb1cb6c5942938d5b1cc8f4a944e0 : SUCCESS in 7m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/30ee810470e54715954bdf01cee7c578 : SUCCESS in 12m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0c78008a8462429cab32bf914eabb67f : SUCCESS in 6m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/967db89ad12047adb7856d362bf22d81 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c676b7d784e64172883ba9fa2855d6dd : SUCCESS in 25m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7d411fc6949e44499bebe5bd7e463bf2 : SUCCESS in 7m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6b974dfb19da4024a403f2fe44c4f00b : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c2f9c885a1ba4a2a9237ca0a1b074f4b : SUCCESS in 6m 18s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2b20901a6a55443897224ff32e0b1f65 : SUCCESS in 14m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df954adea4194d89a04a5f4b4f9ee291 : SUCCESS in 58m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58d13b061e0c49e7a0eeef458d6a047f : SUCCESS in 1h 22m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4b27076f9d94fb9edadfc07f91baf218f530c585","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-14 12:20:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc4eac117365e08fb417dbf3e540b2662e56015a","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-04-14 20:01:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5db648e6fda9201b57633502044e14ab24dbfe4f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-22 10:44:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7166822a593df976f2de7905e06508f5584ac7bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 10:45:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64b550a7b10c6914b3ea862b7a2f6537eaa5ffcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 12:11:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/886fb22739f14bfa9a9d71af2f94befa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2d159d7636483ca68a2ce0bbe2d5be : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e838a55979574192b93c4b01457dde4e : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06e6f12190664f0681d509085af7eb61 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9dadd5c561345f2b1b6f95e550179e5 : SUCCESS in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5df4017d1cb04001bc09592df01ca548 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d11eb2719f60447faacf7be9a48f0b33 : SUCCESS in 3m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cc3e14c7a2a0465495f9d9856c21299a : SUCCESS in 18m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ef9b149ccd804867966f2d35aae7e4ed : SUCCESS in 8m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/801fd7a2f0ab4542aefc7ef8e48e47b7 : SUCCESS in 16m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/72430266af3e42758fc48ff3070788cc : SUCCESS in 22m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d5735dd6d9f4439e9559d189c08d48b4 : SUCCESS in 10m 25s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c40b516975694379b2f575122cbae153 : SUCCESS in 4m 21s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f3a86c75eab94ec0a55be2f4aebbc8f1 : SUCCESS in 6m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f794273da64e480bb295cb73dcdbb344 : SUCCESS in 7m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d5cf779491734abf8b71621e3ec4267c : SUCCESS in 12m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad6a48deb3f14ff6ba90fe5fb9ce64ba : SUCCESS in 7m 28s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a869e0d963b741f58b7a235a3d221f66 : SUCCESS in 8m 28s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4d8a0cb8421041aeace4c44402b6efa7 : SUCCESS in 12m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3baad73539654acdb39c42cb952f7e84 : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bc40d06aa2154375a21dad7328061a96 : SUCCESS in 11m 23s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d31543de1dbe445181713fc43d9571bb : SUCCESS in 21m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f41fdc8c5abd42e68559e5c0565c5a9c : SUCCESS in 8m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/196f09c8deb745b9a2a470d542628077 : SUCCESS in 15m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/48287fdac3e6410aa14a67fe2badf5a4 : SUCCESS in 6m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f15ca9a49f0f4612af4c94e419b4cbce : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dbdcd86aafae40c897c43824abbdf796 : SUCCESS in 26m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a32925a2694c4affa45dc1382556777b : SUCCESS in 7m 50s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3a64fd0415034ffca12b96de79d8dc55 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62d783fa4d88489a8ccb195ccbafddad : SUCCESS in 5m 19s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a7794b76eaa5442ab6c75c352a00e9b6 : SUCCESS in 19m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/534946d5e57543f48fb85dd2d276bb1f : SUCCESS in 57m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/749189f9970d4b5d88e9e5fa0ea325cd : SUCCESS in 1h 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"87bedb8b55154a3cc9f4a1c60c8151464428229d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 12:11:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"00289fa4a93dcaab723fc64ed0999d389fa31330","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 12:12:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6c523c5a5474b20bae48ddb432af68f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a7f51429ab74defb253cd01e96b2125 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1},{"id":"0cc67bac9db62eaa38f6ca28043c4ff8667921c4","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-04-23 21:55:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9571e038afeac72316c379472c7ea885b38ddfab","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-04-23 21:56:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fe0f67186b8a8bd561c3c9c31491484534a7cfc","revisions":{"6fe0f67186b8a8bd561c3c9c31491484534a7cfc":{"kind":"REWORK","_number":1,"created":"2025-04-10 12:36:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/946898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/946898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/946898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/946898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/946898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/946898/1"}}},"commit":{"parents":[{"commit":"742e4e18efd269d49b22a36e5b963bf0d1b9d361","subject":"update constraint for python-openstackclient to new release 8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/742e4e18efd269d49b22a36e5b963bf0d1b9d361"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-10 12:36:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-10 12:36:29.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 4.5.0","message":"update constraint for openstacksdk to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I617c7417df89e8b1d8e3127778182c4500d2b34a\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I2132d2d08b3ec0e6c3fa1fae4c7edfb607436da2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fe0f67186b8a8bd561c3c9c31491484534a7cfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fe0f67186b8a8bd561c3c9c31491484534a7cfc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
