)]}'
{"id":"openstack%2Frequirements~947556","triplet_id":"openstack%2Frequirements~stable%2F2023.2~I079e2c7e7ecc1d2f7afd9ce4f299b78f3447e7b1","project":"openstack/requirements","branch":"stable/2023.2","topic":"new-release","hashtags":[],"change_id":"I079e2c7e7ecc1d2f7afd9ce4f299b78f3447e7b1","subject":"update constraint for python-cloudkittyclient to new release 4.8.1","status":"MERGED","created":"2025-04-17 09:27:49.000000000","updated":"2025-04-29 14:25:16.000000000","submitted":"2025-04-29 14:25:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947556-new-release","meta_rev_id":"e27e7a4a7c340a252d73f5d3d44d977750a523d9","_number":947556,"virtual_id_number":947556,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-29 14:25:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-04-29 12:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-04-29 12:42:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-17 10:20:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-29 12:42:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"82895041676f64a6f230d819342806db3edb057f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-04-17 09:27:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ff16cb6ba2d169c560be5d5045f30c624e9c4b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 10:20:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b7baca976cc4280b08f27ba2278bce0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1849a1b43578418f8ef0a62e82798825 : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7d3b30b8561420b9a935ea81b8c8d21 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b29b442c44b547ea8adb6a4b1a8ede4d : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6be375f927b24a02a2675ec01ce4643f : SUCCESS in 1m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e466eae5ba946959c1987766c939488 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1059d8d440a942c49091577cc2042cf6 : SUCCESS in 2m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a99cd8ecffbf4f9e8bffafad3302d17e : SUCCESS in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c67feefa23634c149ea5c5d94aca550a : SUCCESS in 15m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d18d12ccdecc4ef09a46b7677481191b : SUCCESS in 12m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6f6c998560974a0f99988f744ae21cf9 : SUCCESS in 9m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d8e5c352062b4969b54b8233529fc0b3 : SUCCESS in 14m 54s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f994cf313d1c4fbfaa94a17987b1dc7c : SUCCESS in 5m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/021309976999411089fac13a7a3ed367 : SUCCESS in 4m 01s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/341df338ce6a43eda0e3cebc645e2672 : SUCCESS in 4m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/453618a0b89c48ba8127f09c2760d9cc : SUCCESS in 10m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e1741904bb9047c19ab794aa3b55a49c : SUCCESS in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ac8c185da2c24152af1a61db8ce3661e : SUCCESS in 4m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9ec6c626d58849659d9098994e3c9c96 : SUCCESS in 7m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d9a9eda6619c459c939f7bc114d6ab26 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac300037d26e4e79ab044329c8fb1f69 : SUCCESS in 6m 21s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f95a3a1e2f584faba5697770b9064ed9 : SUCCESS in 7m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b441d8e94c9440a49c38d14913d8f02f : SUCCESS in 10m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/70a0f25a722a4dc9b2d5cf0ec7c14dc1 : SUCCESS in 2m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3696937dbd2e4016bc60995fdc217354 : SUCCESS in 9m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7e6b62014f65461a97a0361f343d7991 : SUCCESS in 2m 48s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/92629e6f7b014457a5bd58d62f512f1d : SUCCESS in 8m 51s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e67eaaad52724a70a81141a341fbb232 : SUCCESS in 25m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7a9db90ed9fe4445b1763b53f2ccf932 : SUCCESS in 7m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c7f670862f0d4febbbbe5fcd05453d4c : SUCCESS in 9m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f886ff691b664b17b5e37c44c9bb3d7f : SUCCESS in 2m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fede514ded784aafb59f8e47741dde11 : SUCCESS in 2m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ca616391b9d4c57bfe12e2f0d9186d7 : SUCCESS in 18m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7b28926852af4830a2368f222f4f6271 : SUCCESS in 4m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/18d5ff73112b44eb90bc9b23fb4593aa : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0dfd46b70f87404496d57592db5ff83e : SUCCESS in 2m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d5c82c0dc04a41c98c8b128d57571a38 : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d099feda2984ae9b5cd39b83c980df7 : SUCCESS in 38m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9c7360deb59478ebde185f034948d60 : SUCCESS in 51m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d3b7f6783ddd4dab2367e1cfb537319083635e14","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-29 12:42:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4da4408d341012b653541743291fae23dd467459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 12:43:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b2716a3685c14af98b47d26d8300c149667a015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:25:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e783181c19614b629b7cac438242e515\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b7498b793d641e59869676ce6555468 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df50080b6e384a1a8795156748c3b4c0 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a8c6bb3ba9749a08768f55b08d9eff5 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22c66f84f49b4c00be6ef6a4fc1d0f2f : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a582248aa65432e9accaa2fbde46026 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c926636d7d14a0fa914726e163421d0 : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/172b91870182426f94ed20b774a261cd : SUCCESS in 7m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dacacb5309a34bb1afc83e92e4563f12 : SUCCESS in 12m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e8ffe8b869dd4bf1962e58a8717b1c7a : SUCCESS in 6m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9072fae62cae4227ba625b39947fe7a8 : SUCCESS in 10m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/02e198fd7d3c4804ab1118bf8c26916c : SUCCESS in 4m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5a324e9d530d433180f7fc6af62d3e61 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/418512ed18cb40f9b6eadda0c84fb5ab : SUCCESS in 6m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bb77e111fb0647748d79d74d89db4d80 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7cadedd94daa44108ab3b4babacad831 : SUCCESS in 7m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ce7328cbda2e44a6a44fa040c0ac3236 : SUCCESS in 7m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/910319f260b040b481a1aad4ddb07a59 : SUCCESS in 10m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b24a9101c8924f92872c904202f71b74 : SUCCESS in 3m 27s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0c12770253af4af9947419e99fa46d8a : SUCCESS in 9m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c89319047a15446c94d7465883b01029 : SUCCESS in 24m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/63105a6665db4da3b06d61f2f3fee28c : SUCCESS in 11m 08s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d4678d3ac7cd4c779cd76e9928a9d2b7 : SUCCESS in 12m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/539e55ae3fa54c99ac921875efe6f7a1 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/72c8cbba0cc8427fa5d80959a97ef490 : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3edd67c7c8647629b3e98982ef706e1 : SUCCESS in 23m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1dc6f4f95b484bdbbf4032e4e2798480 : SUCCESS in 5m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/68b3914940fb4cf296187695ffa9b517 : SUCCESS in 3m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe05f24a273d468e98f6a92064fd6a33 : SUCCESS in 6m 46s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f67befda7d2248178f0fb4b6d1605fc0 : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6963cd2bc69b4517b5d888a47eadf354 : SUCCESS in 48m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31f02fca84224bbabe8acceaf45d63fd : SUCCESS in 1h 21m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e27e7a4a7c340a252d73f5d3d44d977750a523d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:25:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4b012c45f32e7b578f8e06e1a9d5ae934e85fb7","revisions":{"f4b012c45f32e7b578f8e06e1a9d5ae934e85fb7":{"kind":"REWORK","_number":1,"created":"2025-04-17 09:27:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/947556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/947556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/947556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/947556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/947556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/947556/1"}}},"commit":{"parents":[{"commit":"9042ccffadf3de68b5e2e0ee76e3d085842fd161","subject":"Merge \"update constraint for oslo.db to new release 14.1.1\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9042ccffadf3de68b5e2e0ee76e3d085842fd161"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-17 09:27:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-17 09:27:47.000000000","tz":0},"subject":"update constraint for python-cloudkittyclient to new release 4.8.1","message":"update constraint for python-cloudkittyclient to new release 4.8.1\n\nmeta: version: 4.8.1\nmeta: team: cloudkitty\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: stable/2023.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rafael Weingärtner \u003crafael@apache.org\u003e\nmeta: release:Commit: Rafael Weingärtner \u003crafael@apache.org\u003e\nmeta: release:Change-Id: I1775cd9f8b76f69178bb07edfb06471d3104eda5\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Rafael Weingartner \u003crafael@apache.org\u003e\nmeta: release:Code-Review+1: Matt Crees \u003cmattc@stackhpc.com\u003e\nChange-Id: I079e2c7e7ecc1d2f7afd9ce4f299b78f3447e7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4b012c45f32e7b578f8e06e1a9d5ae934e85fb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4b012c45f32e7b578f8e06e1a9d5ae934e85fb7"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
