)]}'
{"id":"openstack%2Frequirements~947727","triplet_id":"openstack%2Frequirements~master~I6bd678411b36640532001a9c0da1081672e4d3a3","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2025-04-26 22:39:29.000000000","reason":"Change was submitted"},"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2025-04-26 22:39:29.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-04-26 22:39:29.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-04-26 20:56:24.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-04-26 22:39:29.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-04-25 14:42:03.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-04-26 15:37:16.000000000","reason":"\u003cGERRIT_ACCOUNT_10239\u003e replied on the change","reason_account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}},"hashtags":[],"change_id":"I6bd678411b36640532001a9c0da1081672e4d3a3","subject":"Add cheroot","status":"MERGED","created":"2025-04-21 00:23:25.000000000","updated":"2025-04-26 22:40:35.000000000","submitted":"2025-04-26 22:39:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947727","meta_rev_id":"72cae953b0ce3df914e1756ac3ec6042bdbf5d78","_number":947727,"virtual_id_number":947727,"owner":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-26 22:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-04-25 14:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-04-26 20:56:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-04-26 15:32:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-04-25 14:42:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-04-26 20:56:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-21 02:13:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-21 03:56:52.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-04-21 06:13:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2025-04-22 15:20:13.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2025-04-24 08:36:53.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-04-24 09:51:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-04-24 09:51:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2025-04-24 09:51:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-04-25 14:51:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-04-26 20:56:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"008616b7957c15626e0d0cb54d627ea26f586a7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-04-21 00:23:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f9d350b3ba3a365a8dcc6538229ca667615bfdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 02:13:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b05fe726131c4e669a5203343a3cc20c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d96fbde5fe7404397df7db2fe990ae8 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ba13b1640a449887737c68f4e1a7c6 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dba6a2f97b2b4c4a95b1b43c04efeef9 : SUCCESS in 6m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76f077b0e590474b8b15cc474640ac16 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60515ca6f8bf41319ea8410ee79c3785 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f5c19d3fee94369953d0112f907fad8 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a18ec629cddd4d32aa3eb5a5e6328dd3 : SUCCESS in 4m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5aac116da01e4a559c5bf0ea52d437ab : SUCCESS in 15m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6caca5543daf4832831b7ab1b2ea32fb : SUCCESS in 8m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7a6ce85c4e184ac681920e0dc003ba8d : SUCCESS in 16m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/694ec781d4f04dbc90b6e0b362c2fe1d : SUCCESS in 10m 04s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8e9ec02707fa4b06b1a77c208b2d1152 : SUCCESS in 5m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ddff12ff4e0e4ca7ad51e0d3383423a8 : SUCCESS in 4m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/46574976528b4000954ddb09cc1d19ba : SUCCESS in 2m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/96fade58090b4837800c2238dd80ee2f : SUCCESS in 11m 22s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5150f18aef644d64986da672b1b45e74 : SUCCESS in 3m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4b7b7ed709e14b23a00c9c074881a962 : SUCCESS in 5m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2dcf86a0fb4440d5b5fb29a10c8b4aa9 : SUCCESS in 4m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8b8ced6a77a2405c8452471656351906 : SUCCESS in 13m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ced68b638e67476984885fd59162dd13 : SUCCESS in 5m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/14ebb30719e04f70ba4c917630dd1682 : SUCCESS in 7m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4a05fd24f7fd4f6d91423f7e8c8cac68 : SUCCESS in 6m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b91c66112aa74749aa58e2cf307137ac : SUCCESS in 2m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2e50ce36d3f04841994b7324714bb158 : SUCCESS in 9m 47s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b2561097598d4774bae39ca199d93be5 : SUCCESS in 3m 30s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4d62ba3190de4214bf975378aab85e28 : SUCCESS in 3m 58s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7304522796bc464eb1a68dd5e0ff9f9e : SUCCESS in 8m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0939d45d9a85419c92a0afeab47c7b42 : SUCCESS in 14m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9368325a1cc14055b91d87011ec4b1c8 : SUCCESS in 7m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6eeb0e11c84f4da5a481bb1e1718b632 : SUCCESS in 11m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d8af1484cf3b4f2a9cd7cc970a9e4756 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab4c8a2b38024fc3be075d622af57cf9 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88306fec622d417abce9abf8179081bc : SUCCESS in 21m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f30ded88401a4ac197ec6a06f1846d41 : SUCCESS in 6m 14s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/eebb6e1207734b5ca1b72e610fb5cf44 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/482800cc4d2c4b77bf9e8ccc4f565351 : SUCCESS in 5m 19s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b89f6708b18646e89e1987e472531e4e : SUCCESS in 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51b87a00be6446358d329a6213fc6a68 : SUCCESS in 1h 02m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/168a02610a1942fd9649ae5ba845f633 : SUCCESS in 1h 49m 22s","accounts_in_message":[],"_revision_number":1},{"id":"adf5bee96dc29b9e8d08dbe4e13aa0282b94bde4","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-04-21 03:56:52.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b6095a820ece865038bb0c285e91d538f0d6043","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-04-21 06:13:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a017cabd1c11a9d64be12115df821b17f5f1932a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-04-22 14:25:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6d68dc2ae14e293b62d3077c677ef420fac819d0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-04-22 15:20:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74e17d3f49e33b7657e267f02cb6eed81723592a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 16:25:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3aae9c1fa0c43e7ad23640d817da577\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc9c96186fa64a13abeffd2d577da5ab : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a232281ea5d481391d3b41f4a17cf88 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32eaa88b35ee42f5a0ffac42382baa80 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d548848141004cf98190116fe462fa22 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d61549708db74a299ff4c7254c947122 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1231388971234d29a592f56ea782ed17 : SUCCESS in 3m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b57d8685fd4429bb637cdb1b9426de0 : SUCCESS in 46m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6150952ed4ab4cc2ab9deeeaff833fdc : SUCCESS in 1h 53m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ca3629dea2ae98c20f340afbf65cf1cfb5b413c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-04-22 16:36:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3a179e6bd6e5a7a445f4066d0e24b868065c7db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:13:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd2d4f55c4f54f9698fd4508d9a3c1d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1258d7194045cfa02e5f7423af2ef4 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00223b8633ae4f20af748dae01825deb : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9fd15df315740cfa6c05ab7dd6d4e81 : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de034baf7d38403784426955410cf6f5 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e48531a32384937920cb242a6643190 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/22e9d41170934ef7a8f81752879c0970 : SUCCESS in 2m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c98fe49bd5664c34bce860768b91b3c0 : SUCCESS in 1h 03m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2272b27d8b264c6184f379ef7d20bd55 : SUCCESS in 1h 34m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d71873eaba976b96008b2eb6e00c51fb0a2cb0a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-04-23 21:00:29.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1c74b9bcda26498ce6260f8148ad40e655369480","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-04-23 21:04:40.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c5c0bea6a2939564c5a11edaff6ef97e61a7d58d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 22:04:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9415227e00ea4cff96b12ce546436c85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0199572febaa4b21b7da70f189ad2e9e : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92a7da056778441a872c2cde112df1d0 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45f86e06e3df4a4c933254b798dfceff : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13b3c2d76ba14937a703ce32c7575765 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a8ed4f4f637475d9fc7f5f4ff1d69be : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba3a75aef3fd4280aa3a9d4762c41230 : SUCCESS in 2m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3f997d11ebc40c0add1f52c2beecbeb : SUCCESS in 32m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4ff446c07ce4a9684a51440c4649058 : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":4},{"id":"9ab37764bd2e546ba48c5ee9b725a09562a7268e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-24 08:36:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8e3fdc3d83f3db22957c7e9dff6c0ef303179967","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-24 09:51:43.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2299aa955f36bc6e28d307d4104240794dfba833","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-04-24 12:24:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"98d71a9ec69dba79ed37c1462665ec3e30af3953","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-24 16:36:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"76658c0421405a1a0461a2335ae8f4ca2353fc7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-04-24 16:38:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"81fb3f4f6a031713adc2a958af9b000916d7dedd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 17:54:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/454f8ecfbd494453a739e40c27259972\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2c87013d1d457c8c3918e80c2ba809 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e0966668c66470baefca01f8afc8323 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/beddf207d5c24c5aa3ed961099ed057f : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3f74963307b4cdbb27af6fc83ddea9e : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad919223939541149107abd658df6d64 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51caaceacb364f27830ede1f27ead3b5 : SUCCESS in 2m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01f3c13fea42430faeccbdd185e95bf7 : SUCCESS in 1h 00m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/453663f137694f3f84501f391d2a2e51 : SUCCESS in 1h 15m 03s","accounts_in_message":[],"_revision_number":5},{"id":"d3ade141c6da2bc501d578762c154672dbd6e4e6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-25 14:42:03.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0a6b26831f110a4b93f502457a67c88ed23d3e33","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-04-25 14:51:19.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bee073963d1449896becf0f5bf016d7792a9291d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-26 13:41:09.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0c16ca1cbfefff87ce16f9446bbee567d121efb0","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-04-26 15:32:39.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d6c4c69dd2c62a3d2bade9e6a1dd39cf4e2c8468","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-04-26 15:37:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"146b49cd2be5d730111ffb96eef2820c47ac53b3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-26 20:56:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b363e4b6d9dcacf24462632a6f9edfac7f2a90d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 20:56:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"02a62cd06926ff265e4246ae019cc3d0e924f5ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 22:39:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc209558681f4d5ca5eab57a1856e9ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1241e20ac42a48d3bc1f9bc1c3282b0b : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec53b431dde427a863388f7445d4c70 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d6ebd370beb4014ba66725c18ee3670 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0214db85bb644c3d88743376b5c5b2ee : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28525d017b1d412eb3e7bca042c4ab9c : SUCCESS in 3m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6dacfd29bde4cdf83cd95b5af56550e : SUCCESS in 55m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd15650b0e294efc941ebf3dcc9c4f74 : SUCCESS in 1h 38m 04s","accounts_in_message":[],"_revision_number":5},{"id":"02e063aaa7c6077da986f33560f1aabfe5d3e802","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 22:39:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"72cae953b0ce3df914e1756ac3ec6042bdbf5d78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 22:40:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c004e4811da34a5abeefd2ad641c9c13\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a910ba440dd491e861d3bdfe187eb06 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0165429148bae695e62013c7ffd38b6310fac489","revisions":{"860228dea49e112155591331cad6bc9004b97aea":{"kind":"REWORK","_number":1,"created":"2025-04-21 00:23:25.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/27/947727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/947727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/947727/1"}}},"commit":{"parents":[{"commit":"1cecf886b2df680590d5eca285fae3ffe899ce38","subject":"Merge \"generate-constraints: Make use of ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cecf886b2df680590d5eca285fae3ffe899ce38"}]}],"author":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-04-21 00:22:34.000000000","tz":60},"committer":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-04-21 00:23:21.000000000","tz":60},"subject":"Add cherrypy to requirements","message":"Add cherrypy to requirements\n\nChange-Id: I6bd678411b36640532001a9c0da1081672e4d3a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/860228dea49e112155591331cad6bc9004b97aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/860228dea49e112155591331cad6bc9004b97aea"}]},"branch":"refs/heads/master"},"5607e79c2c2176019317f9c37f5d2124e9f77daa":{"kind":"REWORK","_number":2,"created":"2025-04-22 14:25:14.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/27/947727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/947727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/947727/2"}}},"commit":{"parents":[{"commit":"1cecf886b2df680590d5eca285fae3ffe899ce38","subject":"Merge \"generate-constraints: Make use of ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cecf886b2df680590d5eca285fae3ffe899ce38"}]}],"author":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-04-21 00:22:34.000000000","tz":60},"committer":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-04-22 14:24:21.000000000","tz":60},"subject":"Add cheroot to requirements","message":"Add cheroot to requirements\n\nChange-Id: I6bd678411b36640532001a9c0da1081672e4d3a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5607e79c2c2176019317f9c37f5d2124e9f77daa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5607e79c2c2176019317f9c37f5d2124e9f77daa"}]},"branch":"refs/heads/master"},"a43e8b9dcde3ff9d2db2c3bfcdd02f8a8417e60a":{"kind":"REWORK","_number":3,"created":"2025-04-22 16:36:10.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/27/947727/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/947727/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/947727/3"}}},"commit":{"parents":[{"commit":"1cecf886b2df680590d5eca285fae3ffe899ce38","subject":"Merge \"generate-constraints: Make use of ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cecf886b2df680590d5eca285fae3ffe899ce38"}]}],"author":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-04-21 00:22:34.000000000","tz":60},"committer":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-04-22 16:35:24.000000000","tz":60},"subject":"Add cheroot to requirements","message":"Add cheroot to requirements\n\nChange-Id: I6bd678411b36640532001a9c0da1081672e4d3a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a43e8b9dcde3ff9d2db2c3bfcdd02f8a8417e60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a43e8b9dcde3ff9d2db2c3bfcdd02f8a8417e60a"}]},"branch":"refs/heads/master"},"7a775cb922144200d699f78e5e6d7309f6509250":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-04-23 21:00:29.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/27/947727/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/947727/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/947727/4"}}},"commit":{"parents":[{"commit":"1cecf886b2df680590d5eca285fae3ffe899ce38","subject":"Merge \"generate-constraints: Make use of ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cecf886b2df680590d5eca285fae3ffe899ce38"}]}],"author":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-04-21 00:22:34.000000000","tz":60},"committer":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-04-23 19:57:04.000000000","tz":60},"subject":"Add cheroot","message":"Add cheroot\n\nAs part of the OpenStack-wide effort to remove eventlet dependencies\nbefore its removal in version 2026.2, the Ironic Python Agent project\nis replacing its eventlet-based WSGI server with cheroot.\n\npython3-cheroot\n\n* Is the library actively maintained?\n  Yes, cheroot is actively maintained.\n  https://github.com/cherrypy/cheroot/commits/main/\n\n* Is the library good code?\n  Yes, it is a mature and well-tested high-performance, pure-Python HTTP\n  server used by CherryPy.\n\n* Is the library license compatible?\n  Yes, it\u0027s BSD licensed.\n\n* Is the library already packaged in the distros we target (Ubuntu\nlatest / Fedora latest)?\n  Yes.\n  https://packages.ubuntu.com/oracular/python3-cheroot\n  https://packages.fedoraproject.org/pkgs/python-cheroot/\n  python3-cheroot/\n\n* Is the function of this library already covered by other libraries in\nglobal-requirements.txt?\n  We\u0027re specifically replacing eventlet\u0027s WSGI server functionality with\n  cheroot, which provides a similar thread-based server model\n  but without eventlet\u0027s dependencies. Also, cheroot is already part of\n  the upper-constraints file as a transitive dependency.\n\n* Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n  Yes, it\u0027s needed for ironic-python-agent (IPA).\n\n* If the library release is managed by the Openstack release process does it use the cycle-with-intermediary release type?\n  It\u0027s not managed by OpenStack.\n\nChange-Id: I6bd678411b36640532001a9c0da1081672e4d3a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a775cb922144200d699f78e5e6d7309f6509250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a775cb922144200d699f78e5e6d7309f6509250"}]},"branch":"refs/heads/master"},"0165429148bae695e62013c7ffd38b6310fac489":{"kind":"REWORK","_number":5,"created":"2025-04-24 16:38:20.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/27/947727/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/947727/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/947727/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/947727/5"}}},"commit":{"parents":[{"commit":"1cecf886b2df680590d5eca285fae3ffe899ce38","subject":"Merge \"generate-constraints: Make use of ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cecf886b2df680590d5eca285fae3ffe899ce38"}]}],"author":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-04-21 00:22:34.000000000","tz":60},"committer":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-04-24 13:15:47.000000000","tz":60},"subject":"Add cheroot","message":"Add cheroot\n\nAs part of the OpenStack-wide effort to remove eventlet dependencies\nbefore its removal in version 2026.2, the Ironic Python Agent project\nis replacing its eventlet-based WSGI server with cheroot.\n\npython3-cheroot\n\n* Is the library actively maintained?\n  Yes, cheroot is actively maintained.\n  https://github.com/cherrypy/cheroot/commits/main/\n\n* Is the library good code?\n  Yes, it is a mature and well-tested high-performance, pure-Python HTTP\n  server used by CherryPy.\n\n* Is the library license compatible?\n  Yes, it\u0027s BSD licensed.\n\n* Is the library already packaged in the distros we target (Ubuntu\nlatest / Fedora latest)?\n  Yes.\n  https://packages.ubuntu.com/oracular/python3-cheroot\n  https://packages.fedoraproject.org/pkgs/python-cheroot/\n  python3-cheroot/\n\n* Is the function of this library already covered by other libraries in\nglobal-requirements.txt?\n  We\u0027re specifically replacing eventlet\u0027s WSGI server functionality with\n  cheroot, which provides a similar thread-based server model\n  but without eventlet\u0027s dependencies. Also, cheroot is already part of\n  the upper-constraints file as a transitive dependency.\n\n* Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n  Yes, it\u0027s needed for ironic-python-agent (IPA).\n\n* If the library release is managed by the Openstack release process does it use the cycle-with-intermediary release type?\n  It\u0027s not managed by OpenStack.\n\nChange-Id: I6bd678411b36640532001a9c0da1081672e4d3a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0165429148bae695e62013c7ffd38b6310fac489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0165429148bae695e62013c7ffd38b6310fac489"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
