)]}'
{"id":"openstack%2Frequirements~948337","triplet_id":"openstack%2Frequirements~stable%2F2023.2~I352695754242d721b6e9829ca7a2db38a246b475","project":"openstack/requirements","branch":"stable/2023.2","topic":"new-release","hashtags":[],"change_id":"I352695754242d721b6e9829ca7a2db38a246b475","subject":"update constraint for ceilometer to new release 21.0.1","status":"MERGED","created":"2025-04-28 15:57:18.000000000","updated":"2025-04-29 14:25:04.000000000","submitted":"2025-04-29 14:25:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948337-new-release","meta_rev_id":"f1a73c60f2c744e1c91cc2b40e7fa8bb8d47ff10","_number":948337,"virtual_id_number":948337,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-29 14:25:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-04-29 12:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-04-29 12:39:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-28 17:21:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-29 12:39:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3f89514f7526ae29f99850fd0faac41fd5c5bc75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-04-28 15:57:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ddc202e51a4b241f82144352799b275f5755edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 17:21:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaf43d003cf7453c87e75be51c87ec89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8285b5a7974f46bea45691e1fb545e8a : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5b0e9ec7ca0485eb9c9f637aafaff8e : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/883c0dc272f84b93adaef8f5e8dea2d1 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03864141f4f34033a27dfb12bc908248 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16978753041041f79425e29ffb583bc5 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40684f415cc845fdbeae00d6ce1d56c6 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0135e19590134081b7655b6f7ddbc04b : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4e31f4a555a0452c9c3b739f6cdccde9 : SUCCESS in 6m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/335d840420be48ab95f64efd7551f4c3 : SUCCESS in 14m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5785efe430324ca4bd86d9a07f3575c7 : SUCCESS in 6m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/adefc0fdb3a846d8b81b8c3a0c451ae1 : SUCCESS in 12m 14s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ff2ea134e38c46f2a770448ca2861842 : SUCCESS in 8m 06s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8fc7166e1d784f31924c3d40a8131068 : SUCCESS in 4m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/77fc26c141f34e3cadba77578cae0b32 : SUCCESS in 4m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d05d68e155d24b5082f140d3f3e03779 : SUCCESS in 10m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/07818f303c554bbb81537c271020ac31 : SUCCESS in 4m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/37fc40981c814cfaa47e64d5a2c40671 : SUCCESS in 5m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b763209c5fa84cf9b8193fb0d30cb853 : SUCCESS in 7m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/da2e87a433774e43938ccdfbefe24627 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f26538da126491692879d5dd9463d54 : SUCCESS in 7m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/13ddb5dcbcf443d6b5df00f12116c274 : SUCCESS in 7m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2bc2008578be47ed87ce491962470763 : SUCCESS in 9m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2b7bca803a83487bbaf0953248c35980 : SUCCESS in 3m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0fe515f595f4442cbdb1e3c8e7628cb0 : SUCCESS in 9m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/00f3425d7eed46c8bcf1c6a1f3cb1827 : SUCCESS in 4m 05s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/96eb0a5bac424a33afcd2d404f458c82 : SUCCESS in 10m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c0d6ded33c1a4f1097fcb6bf72422ee3 : SUCCESS in 26m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/efad19c8d89e4de6a6789d0ca92430eb : SUCCESS in 11m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0d69e55757394cd6bb42320510f1351e : SUCCESS in 13m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/410b266664b7473bb2938fcefc8285ff : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cb07c748a479499fb0eb3df08c22cb2c : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54941e015a4345ca9be7b2074ca53656 : SUCCESS in 24m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e404c86bfe5f467094ec0923da708f05 : SUCCESS in 6m 34s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/92dd6342b6e149ef804ca6c818e97de9 : SUCCESS in 4m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c3ad704e1890432cae0ac294a02fe356 : SUCCESS in 5m 22s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/886bad967913435ea4c60f43a2b8ded4 : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cc318785e9740c386ee2e149e128bf5 : SUCCESS in 51m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4226a6bd20af4230bc69b476fd8400c8 : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":1},{"id":"45173a2426a4a8d2c298fe311c4a3269f95ceb95","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-29 12:39:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2619d5e45ab50223c9c9559df1563f243d5754a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 12:39:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8867731b2903e52b43a9e57620c632bb290a4854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:25:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edc98132030f44d79edb32df0a8e05b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc009f233be4ec7bdc56a749fab6ee6 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3afab1a533ae41a58a01b4dafd1b1775 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3ad3e9674af445d87c3cc9e5479c0bd : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e37609e84e33485b9ee1aaa56dc31769 : SUCCESS in 1m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00fe026192c24330a8bb18a972fed9cd : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a15aa2b21df40119e8ad53e4782c80c : SUCCESS in 2m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9979bc8764d447b780d7631555035942 : SUCCESS in 3m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1b27a46ca69e48aba55b92103f76cb60 : SUCCESS in 12m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/337918382c854bc7ae0920f7823e284c : SUCCESS in 8m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c30a9b4bbb944774af373c2974a3d9b9 : SUCCESS in 8m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/130fd6710e974c05bf263344e72c9bdc : SUCCESS in 5m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9eead4984ab34a59b046acb8446bdae8 : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a40f2cb4a2714f15842bb15bafca04c3 : SUCCESS in 6m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/01557fe83ff74735806005c8d6c9670a : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eca83e83ed654d9db5be747443f4c2f4 : SUCCESS in 6m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/009a96ee094d44849e92a61b7989e2a3 : SUCCESS in 7m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/eadeac119ea3441583e640689c6abc68 : SUCCESS in 11m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1319a2a6693a4ceeb09ea09c35161115 : SUCCESS in 3m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/146c44a4d2fb4b179078a8bd5d81a965 : SUCCESS in 7m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0b0154285ac047499206d2f77b5d26f8 : SUCCESS in 27m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c880922fce6042dc985783857f8d613c : SUCCESS in 10m 41s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6ae5231b5f4a4073985768202061781f : SUCCESS in 13m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d1f1ecdeee9d43d98eeb83a18eb042ee : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08b5873711414fd59a7e5c3cc7511a87 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d8b2cd19d9744bdb64bb9432a926f4e : SUCCESS in 23m 06s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a9ae012c96b547649eeafd4adbd74aea : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8f44f9628ffd433380d33c3bbdaacc6f : SUCCESS in 4m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/871896e9fa5546afb482fd91694633ba : SUCCESS in 5m 51s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1362c867e18b4d608efefa3684e369ff : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8920d3f06ef341eaac2b3f9dfec33405 : SUCCESS in 1h 00m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/580719b9b49246748619b77dd3253a35 : SUCCESS in 1h 35m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f1a73c60f2c744e1c91cc2b40e7fa8bb8d47ff10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:25:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23d29c55c65cc5e6296bcee4d41867b4cbdcae32","revisions":{"23d29c55c65cc5e6296bcee4d41867b4cbdcae32":{"kind":"REWORK","_number":1,"created":"2025-04-28 15:57:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/948337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/948337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/948337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/948337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/948337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/948337/1"}}},"commit":{"parents":[{"commit":"9042ccffadf3de68b5e2e0ee76e3d085842fd161","subject":"Merge \"update constraint for oslo.db to new release 14.1.1\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9042ccffadf3de68b5e2e0ee76e3d085842fd161"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-28 15:47:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-28 15:47:17.000000000","tz":0},"subject":"update constraint for ceilometer to new release 21.0.1","message":"update constraint for ceilometer to new release 21.0.1\n\nmeta: version: 21.0.1\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: stable/2023.2\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia5501dcecbab494a7bf92cff2dafd130d7c04fbe\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Juan Larriba \u003cjlarriba@redhat.com\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I352695754242d721b6e9829ca7a2db38a246b475\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23d29c55c65cc5e6296bcee4d41867b4cbdcae32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23d29c55c65cc5e6296bcee4d41867b4cbdcae32"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
