)]}'
{"id":"openstack%2Frequirements~948424","triplet_id":"openstack%2Frequirements~stable%2F2023.2~I629c59ed0c78c9a1bd5011dc20f58aff996f5f37","project":"openstack/requirements","branch":"stable/2023.2","topic":"new-release","hashtags":[],"change_id":"I629c59ed0c78c9a1bd5011dc20f58aff996f5f37","subject":"update constraint for os-brick to new release 6.4.3","status":"MERGED","created":"2025-04-29 11:46:05.000000000","updated":"2025-04-29 15:11:43.000000000","submitted":"2025-04-29 15:11:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948424-new-release","meta_rev_id":"e2677dac0b97e0980c3025074016bce7a7f0169f","_number":948424,"virtual_id_number":948424,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-29 15:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-04-29 13:41:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-04-29 12:43:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-04-29 12:43:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-29 12:43:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-04-29 13:37:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-29 13:41:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"5c718ec5cb606e5b3dc10247f2bd99f7d2ed5ed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-04-29 11:46:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2f671933f1116f38308a647a1279dabd4080c56","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-29 12:43:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"24e57499e3e744b4eeaf34dec884cdf4808443f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 13:37:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3319c7c2935b4128a644de4c3799a19b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e866c0ee9464d9d857559948bb57baf : SUCCESS in 1m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef010ce7505f47279065f013fe95a3ed : SUCCESS in 1m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5ff8ae0c8af43388f08a6f8c0d38dfc : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/220f7a4046404cb7a207a2f955c02706 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e185e500fd94616b7f6302f42c87176 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b065e9ed1e62415b9d05cc3ab4162eec : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e94f9d66803746a682d14cd4b3c66fce : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ad8d1e508b6d492d84c7115982f412ad : SUCCESS in 8m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/07def5e6816f4d58b565831f8527895e : SUCCESS in 11m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3305ae5543ad4756a760bbf303670249 : SUCCESS in 6m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1bd062b038954e9daa36ef26f4c77329 : SUCCESS in 10m 35s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e703a248d2c843199d9dba0a65773d70 : SUCCESS in 6m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7ad563483c87415a822052ab41aa5f49 : SUCCESS in 4m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f280558aa5964b1bbbd39417bfab8201 : SUCCESS in 4m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a8aaee60070f413f87897678c14840d0 : SUCCESS in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/aba0af81292a459c8b1999387e33785f : SUCCESS in 4m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/65de3e26cc0a497baebbe92048139b9d : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dd64e45085724145a6700ddc2305dae2 : SUCCESS in 4m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/025494ceccc84fa5b64aeda43ddd9053 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c80e7059c82a482b8280112f26e3b067 : SUCCESS in 4m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/23d439c4b1c34374897abe35d14353ba : SUCCESS in 7m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9d7258fa8c7c41c1b94bae0ecca51d1b : SUCCESS in 10m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/03c1325ce6ff4618a6193d9cd18a517f : SUCCESS in 3m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/82982e876dc147a189139217127f0406 : SUCCESS in 6m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/00a69fec82184d5b96300bf173283f52 : SUCCESS in 2m 26s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0048ee816eaa4a2d93835dce15b66980 : SUCCESS in 9m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9adb6ff9f45f4523a89c2b7b76b93c08 : SUCCESS in 27m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9766c6835c6c4313b3701b26e8cb3942 : SUCCESS in 5m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5d0963f4a4c54bdbb585ebd61e5caae8 : SUCCESS in 13m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3cd63c89e4e640ae9124e00df1f23acc : SUCCESS in 2m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/88061f3472f941cfa38a32843aef2f33 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82b226b801ed4f689701b5b319f5b59b : SUCCESS in 25m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/882a24f927ff429ea415cdc04a8259c8 : SUCCESS in 3m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ea08bd0802d74b8e8c92e922ad56b102 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57b9230e4d424ae5a159a680d1fe8b36 : SUCCESS in 2m 51s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/918d5010daec4dc7b0bded31628f9035 : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc8401755e28413b9d0648617778de08 : SUCCESS in 53m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22712946cffb47df8ab8dd567c331000 : SUCCESS in 1h 38m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f29a8820ebadb487491422dea35441dd7e287cce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 13:37:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4d73c183dd5a4b2db252ecba8efe4941b02d8c0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-04-29 13:41:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc68b28b59b1e591b4d2e547c8617aa4ceeee0f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 15:11:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/573dece7ec9b4728899db1eb465bfcbb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c69a675fc794799af583372a9e79d79 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3498f074f544b16aaa33b8cfa36c26c : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56674426c88342ce91adbce08b55a41d : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d73ef37c8872496d95873100640c1431 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12996faec59c4f5793e69061354469fd : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/43e16fe7ac7948e3b24949c32ac33138 : SUCCESS in 2m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c49ffe1ba8dd4b2a8b91003d1c83f75f : SUCCESS in 10m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b4abca3928a248689ccce97461693c02 : SUCCESS in 12m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/00b5fa65951b4c15a01051683b3e0a34 : SUCCESS in 7m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/42350c318c4a49e88f5c2a9f37a5ac34 : SUCCESS in 10m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/683fc9c2caf7411291b1c5ca7d21bab4 : SUCCESS in 4m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/57fae171d2d0463ab0ab70d1ac01c281 : SUCCESS in 4m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8f543bbdd57a4868b644c9654e49dfa3 : SUCCESS in 7m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f9be4a4d943344deb26e787e40a6a754 : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1cd2f808bcf5433ebebe5b9f321777b0 : SUCCESS in 8m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/628704d0634e477ab605ef85d6976274 : SUCCESS in 7m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/171846e126754d93aebf6088a09f821c : SUCCESS in 11m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/eb052693c1ce4daea7a5c831639e7b3b : SUCCESS in 3m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fde4a60f4e5f4d1f9972b6ab7b500619 : SUCCESS in 15m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3a234aef04e049aeb150e606b72ffa0a : SUCCESS in 29m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/73f672a419b848289c8e73d7790d0f0f : SUCCESS in 13m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b36ffdcabb9d4ca5851627c954aa3d0b : SUCCESS in 13m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e1991d33798349d3b06d7fbe328f9723 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3dc2cf76737e42be9e2dbbb15272cb58 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1eecf4cf03a04e1683a94531571bda83 : SUCCESS in 24m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1833606d380645fc923ef413dafeb65d : SUCCESS in 6m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8b397154040d46ecb7fa084b613e241c : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c79a4f84cfa4c8abd430fa3c67e93ef : SUCCESS in 5m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1b63b3b7eeb74a54b32144be8b6db538 : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab8137b687124d2b861fb871c8e8888f : SUCCESS in 56m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/319d16547ade4738a01d11360d3d97f9 : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e2677dac0b97e0980c3025074016bce7a7f0169f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 15:11:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a976b95d5028b2c2102390e40fc7f89cbf0cc0d","revisions":{"8a976b95d5028b2c2102390e40fc7f89cbf0cc0d":{"kind":"REWORK","_number":1,"created":"2025-04-29 11:46:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/948424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/948424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/948424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/948424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/948424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/948424/1"}}},"commit":{"parents":[{"commit":"9042ccffadf3de68b5e2e0ee76e3d085842fd161","subject":"Merge \"update constraint for oslo.db to new release 14.1.1\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9042ccffadf3de68b5e2e0ee76e3d085842fd161"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-29 11:36:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-04-29 11:36:04.000000000","tz":0},"subject":"update constraint for os-brick to new release 6.4.3","message":"update constraint for os-brick to new release 6.4.3\n\nmeta: version: 6.4.3\nmeta: team: cinder\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: stable/2023.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Change-Id: I351dfa90f9b9f3a848020ac7d103e984eb9b9170\nmeta: release:Code-Review+1: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I629c59ed0c78c9a1bd5011dc20f58aff996f5f37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a976b95d5028b2c2102390e40fc7f89cbf0cc0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a976b95d5028b2c2102390e40fc7f89cbf0cc0d"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
