)]}'
{"id":"openstack%2Frequirements~948791","triplet_id":"openstack%2Frequirements~master~I40c7486bbdc7a54dc7d376a4be1f3fa035d9de4e","project":"openstack/requirements","branch":"master","topic":"numpy-py39","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-06 10:43:22.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-05-06 09:26:21.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}},"hashtags":[],"change_id":"I40c7486bbdc7a54dc7d376a4be1f3fa035d9de4e","subject":"Support numpy in python3.9","status":"ABANDONED","created":"2025-05-05 11:20:50.000000000","updated":"2025-05-07 06:45:29.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a7a73f82dad11f67e0b103830c819201cd19cbe","_number":948791,"virtual_id_number":948791,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-05 13:09:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2025-05-05 19:34:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-05 13:09:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-05 19:34:02.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"6c6bdc7da799d8482affa8fe5c33aa372ea3d4cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-05 11:20:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebb3a8802d3c56e50c056bd127e5e37c0d3e93b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 13:09:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/770990d8919f403d8d4cc8760a9315b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8604d22c33d44b29e86269aec3d3643 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6adfc4b3862f4918a88b3c18790e2404 : SUCCESS in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c584081d57d4e6f8debd31797b101e6 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/92e1e9dcb30a41ab821c412bdbecb1b2 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a3b6ed392fc4b7d84e8af611cc9eaf6 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d95c26c68af04b51915b49e001a8becb : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b5085d22986462ba09cb08117212b0b : SUCCESS in 1m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4549c8ef9f0a42d69c0f1c93e459c1f7 : SUCCESS in 7m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4772eaec620f41379d042efe81de53c5 : SUCCESS in 19m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/17eed9ea2068493692f77686fca24614 : SUCCESS in 21m 12s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/dc97875942f047979bd07ddd0112cb0c : SUCCESS in 16m 40s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0b7b455a3c7e4978ad46b16fbe66172d : SUCCESS in 4m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6af343b064aa4467861ee39d5b6a940e : SUCCESS in 5m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/58b4c9899ddd4b3a8d71f38c4954a325 : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/55545ae03d6641c8a11b6e7b1ffa6f6a : SUCCESS in 11m 25s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/30d7fda06a8c4483bebb905e94814b1b : SUCCESS in 2m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7291bf2f58db46cd8edc3ac24d64f71a : SUCCESS in 4m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f93634fc8af34bc0860e3a820ca85de1 : SUCCESS in 5m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/136635c615f843ef85a3e2e8dab8ff84 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04dad8377158440c923922808b0774cd : SUCCESS in 6m 10s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/949301ccc1684f27ac2f347d7fd5898e : SUCCESS in 7m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0c44019d2fca4180862ffe71bc1e94f5 : SUCCESS in 11m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1708cab8dadb420099e42c700188278f : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/97f2ede02f4740c69280e93350993bb0 : SUCCESS in 9m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b57107d05ae448dc92b387c424636c5e : SUCCESS in 2m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7de6be56d3e74792a505c2671f73565e : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/046cacf03e094b048599fa988ca7e5d1 : SUCCESS in 9m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/27026573842c49d0aa576ead50534f4c : SUCCESS in 20m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ee37da1befd24710925e57f2e7dceabf : SUCCESS in 6m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/50a9305f83cc41a782bfb0e46d31c024 : SUCCESS in 9m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e87e9c0ee7e543a6b4856051ed468d63 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d7822693bcf3440ca6dc21391b526a6e : SUCCESS in 2m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4215f2a47f9b47d19968b4db1007aa54 : SUCCESS in 19m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/dfa8e10480ab4048a264cba70fdc1971 : SUCCESS in 6m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0e33d21cf87741d490e6124a4cd02b4d : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cbae4b095d15460a941869475887ed28 : SUCCESS in 5m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/12726c583d54416ea0dc5008bd82a937 : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53d1d99917c3455d8b189ffab6cfad33 : SUCCESS in 31m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d036a0e3640c4ab7b77c30486cda2b96 : SUCCESS in 1h 43m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d39e9b7b40d6590c4190714e5e2f9f66918e0444","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-05 19:34:02.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b29c38dc29b667e2cad2a0576eb8bd25128b0d8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-06 09:26:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a7a73f82dad11f67e0b103830c819201cd19cbe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-07 06:45:29.000000000","message":"Abandoned\n\noverridden by https://review.opendev.org/c/openstack/requirements/+/948876","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ec0602d30a92f83021ac9262451bc47152469ae","revisions":{"6ec0602d30a92f83021ac9262451bc47152469ae":{"kind":"REWORK","_number":1,"created":"2025-05-05 11:20:50.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/91/948791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/948791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/948791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/948791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/948791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/948791/1"}}},"commit":{"parents":[{"commit":"5ad45c0c1d4150b0bada72ab8bd1f0fdd2c6fc01","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 5.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ad45c0c1d4150b0bada72ab8bd1f0fdd2c6fc01"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-05-05 11:20:05.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-05-05 11:20:05.000000000","tz":120},"subject":"Support numpy in python3.9","message":"Support numpy in python3.9\n\nThe numpy library dropped python3.9 support requiring python 3.10\nor higher starting from version 2.1.0\n\nChange-Id: I40c7486bbdc7a54dc7d376a4be1f3fa035d9de4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ec0602d30a92f83021ac9262451bc47152469ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ec0602d30a92f83021ac9262451bc47152469ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
