)]}'
{"id":"openstack%2Frequirements~948809","triplet_id":"openstack%2Frequirements~master~Iaedcabd84eabac8baa28bf0fa92afe489b9803d1","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"1313":{"account":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"last_update":"2025-10-24 00:53:35.000000000","reason":"Change was abandoned"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-10-24 00:53:35.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-09-12 19:29:17.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-10-24 00:53:35.000000000","reason":"Change was abandoned"},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-10-24 00:53:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iaedcabd84eabac8baa28bf0fa92afe489b9803d1","subject":"Add uc check alternative method","status":"ABANDONED","created":"2025-05-05 14:55:22.000000000","updated":"2025-10-24 00:53:35.000000000","total_comment_count":10,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"b34c7f5fc91075b64d89d891a476f05322e700fc","_number":948809,"virtual_id_number":948809,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-19 13:46:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":-1,"date":"2025-09-12 16:04:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-05 16:31:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-06 13:23:00.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-05-06 13:23:00.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-05-06 13:23:00.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"state":"REVIEWER"},{"updated":"2025-05-06 13:23:00.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-09-12 16:04:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b2dd8479a1d24dcd38e1f2fad3d67c5e14fa6f99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-05 14:55:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d36177e3517ed271018699c4f751b6dd4944530c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 16:31:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea8707da63504f74823b82fdab8d73a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55567998438d4595b46e8f72d83bb4b3 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/880315cec00a443e93f4fc0cfeebdf90 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77e9edf21090459e9cf2616c1c90ad4d : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ad018f181f8406488936677d9499f3e : FAILURE in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c35f06377a9e42d189436d40c978f5ef : FAILURE in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/405563830a004fa9bc331eb277e65244 : SUCCESS in 3m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b19436cc22f2435aa11d479959112f74 : SUCCESS in 55m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d119d5ff40424bcbafa9688f92cd9ea4 : SUCCESS in 49m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5525ed346464ecc71e627f8a3c2f3cef8b8e17f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-06 07:17:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9c97c73240639e0c0b9c69874ba6f780d64ac497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 09:11:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00f23603e6004dbcae5f3721741b097e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313d2fb61cb34e7e9f0fdd54c53a45f1 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfcaed364f1645aaac6fea1e52118da2 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a32709ca16143f2aabf592ee80c6e1d : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/968a4c8b42cf4ba7b7fbebc348b8b781 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb0d81407e1b421d9aeb9825dbb91522 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3810c6d357844035986c1811b8685f0b : SUCCESS in 3m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54d858af1c0b4297b5f4d3bf2033f08e : SUCCESS in 50m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ca27b62c5ac4fb79de0d6f3162f48a2 : SUCCESS in 1h 48m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f9f4dddcdbf41dd32b866783b467775fb1a440f7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-10 10:08:42.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"445fb7562d4d148fc87a5fcf6875c78df2b0c96d","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-14 12:13:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abfdc34ccfb84c3a7e4d290f892cccf47926d634","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-14 15:32:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"afa68c0b077e56b68155cc82064542a9b839d641","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-16 11:40:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19ac4b1d67362eca4e0bb7309829706c448ef6f1","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-19 09:04:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b776dcaa5842c25740ab4d5c114949f54f8a2efe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-19 10:07:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35222046cad8df4f9c7822765fc0d5b73afcc5ab","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-19 11:52:01.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"487a0a1b11889ed9f04c2b9c6ef36d035db12d6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-19 11:55:59.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b917e7ddb8ef41cfe0a6c26125633cc636528e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 13:46:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bbf7c5f69704c26989987d2bb544660\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f6e390a9bb4c689f5324f5d1cdb6e4 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94428569b68344caa559f6dee95208ee : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ad6592f19f14da39eca7a500b176744 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0ca6c090ec04b7eb3eea392b8a2110c : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42bcb8e4ccb5494a8fd89ca2e505312f : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f04ed9db64e4e538c0ef503ecfd46ad : SUCCESS in 4m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/552ebf3dab9249dc964119d3e022135e : SUCCESS in 58m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a85a446f5c74d5bbdfd621b9d0616d9 : SUCCESS in 1h 44m 40s","accounts_in_message":[],"_revision_number":3},{"id":"8a10fe05defb5eb0b991bf53fab49937b75025ad","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-12 16:04:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b34c7f5fc91075b64d89d891a476f05322e700fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-24 00:53:35.000000000","message":"Abandoned\n\nNo longer needed given: https://review.opendev.org/c/openstack/requirements/+/960844","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f96fd83c4ebb07447165c36b3eaf17edd8f0b725","revisions":{"1e954db685e5c597b8f751b94a98e7a9f784a388":{"kind":"REWORK","_number":1,"created":"2025-05-05 14:55:22.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/09/948809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/948809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/948809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/948809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/948809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/948809/1"}}},"commit":{"parents":[{"commit":"5ad45c0c1d4150b0bada72ab8bd1f0fdd2c6fc01","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 5.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ad45c0c1d4150b0bada72ab8bd1f0fdd2c6fc01"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-04-24 11:59:57.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-05-05 14:54:09.000000000","tz":120},"subject":"Add uc check alternative method","message":"Add uc check alternative method\n\nIt also bring the following fix:\npkg_resources.require(name) can sometimes fail due to issues with\npackage name normalization.\nFor example, a package published as oslo.utils may be installed as\noslo-utils, and pkg_resources may not resolve it correctly.\n\nIf this occurs, you can fall back to a more modern approach using\nimportlib.metadata, which tends to handle normalized package names\nmore reliably.\nThis method is also more lightweight and better aligned with current\nPython packaging standards.\n\nChange-Id: Iaedcabd84eabac8baa28bf0fa92afe489b9803d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e954db685e5c597b8f751b94a98e7a9f784a388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e954db685e5c597b8f751b94a98e7a9f784a388"}]},"branch":"refs/heads/master"},"a2485472b5203e8c7d7ba555ee393649ad9ed33a":{"kind":"REWORK","_number":2,"created":"2025-05-06 07:17:50.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/09/948809/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/948809/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/948809/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/948809/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/948809/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/948809/2"}}},"commit":{"parents":[{"commit":"5ad45c0c1d4150b0bada72ab8bd1f0fdd2c6fc01","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 5.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ad45c0c1d4150b0bada72ab8bd1f0fdd2c6fc01"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-05-06 07:17:30.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-05-06 07:17:30.000000000","tz":120},"subject":"Add uc check alternative method","message":"Add uc check alternative method\n\nIt also bring the following fix:\npkg_resources.require(name) can sometimes fail due to issues with\npackage name normalization.\nFor example, a package published as oslo.utils may be installed as\noslo-utils, and pkg_resources may not resolve it correctly.\n\nIf this occurs, you can fall back to a more modern approach using\nimportlib.metadata, which tends to handle normalized package names\nmore reliably.\nThis method is also more lightweight and better aligned with current\nPython packaging standards.\n\nChange-Id: Iaedcabd84eabac8baa28bf0fa92afe489b9803d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2485472b5203e8c7d7ba555ee393649ad9ed33a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2485472b5203e8c7d7ba555ee393649ad9ed33a"}]},"branch":"refs/heads/master"},"f96fd83c4ebb07447165c36b3eaf17edd8f0b725":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-19 11:55:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/09/948809/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/948809/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/948809/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/948809/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/948809/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/948809/3"}}},"commit":{"parents":[{"commit":"36ef8d455a8925d186fe6711c94d4e590e6ede64","subject":"Merge \"update constraint for cliff to new release 4.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36ef8d455a8925d186fe6711c94d4e590e6ede64"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-05-06 07:17:30.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-19 11:55:26.000000000","tz":120},"subject":"Add uc check alternative method","message":"Add uc check alternative method\n\nIt also bring the following fix:\npkg_resources.require(name) can sometimes fail due to issues with\npackage name normalization.\nFor example, a package published as oslo.utils may be installed as\noslo-utils, and pkg_resources may not resolve it correctly.\n\nIf this occurs, you can fall back to a more modern approach using\nimportlib.metadata, which tends to handle normalized package names\nmore reliably.\nThis method is also more lightweight and better aligned with current\nPython packaging standards.\n\nChange-Id: Iaedcabd84eabac8baa28bf0fa92afe489b9803d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f96fd83c4ebb07447165c36b3eaf17edd8f0b725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f96fd83c4ebb07447165c36b3eaf17edd8f0b725"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
