)]}'
{"id":"openstack%2Frequirements~949342","triplet_id":"openstack%2Frequirements~master~Iefcdbc6e050dfe13f54097e665f9a91dec02a63a","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-10-27 15:10:48.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-10-28 01:54:55.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2025-10-28 06:00:41.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-28 06:00:41.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2025-10-28 06:00:41.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-10-28 06:00:41.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-10-28 06:00:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iefcdbc6e050dfe13f54097e665f9a91dec02a63a","subject":"Release cap for bcrypt","status":"MERGED","created":"2025-05-09 13:20:19.000000000","updated":"2025-10-28 06:01:32.000000000","submitted":"2025-10-28 06:00:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949342","meta_rev_id":"ae7c60bc107fbb567ad3006b3f223d5e22ecbdf6","_number":949342,"virtual_id_number":949342,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 06:00:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-27 15:02:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-27 05:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-10-27 15:10:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-09 14:17:23.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-05-09 15:17:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-12 07:30:22.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-05-12 07:51:35.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-07-14 14:30:27.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-10-14 02:52:57.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-10-27 15:02:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"564c9c9c368732b6fcd04b19690812dbdc0d4d77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-09 13:20:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59429a6e70c96d2c03f486a06ddc6acea6d8cb82","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-05-09 14:17:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ce54da45216eec79ff238e466f05d3bf8a137040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 15:17:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ef39bcb4ffd42fb83fc6d8dc4c93eb9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1429267b2674442a8087eac3d058e36 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7183636aa0274310a030947ada697bc6 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efe4ee922d264419a8539817f52d1596 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5054481e188341368db165f5c31b9f9a : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb0ec8a67b0742beb34d4f72c236a345 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fcedcff39a0c45cbb0cf62e3d737992b : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3615d63a8fd8484f9494ecb8cab80ddb : SUCCESS in 2m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b15fd42c0f1f4765bccf1b7d343bbb20 : SUCCESS in 12m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bda498e872e046d7a37666a944d7473a : SUCCESS in 9m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ebf8ad062fd24edaa28d90145661dcd2 : SUCCESS in 19m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/239e88b3fd674e6a9ed8b93c8f690299 : SUCCESS in 19m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cbcd073628f147e29a06c67aec154704 : SUCCESS in 22m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/cf2fa7286a4d402f88f2d1ac99a0ba5b : SUCCESS in 4m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/143616542a12406f87e6ce1ad9163459 : SUCCESS in 5m 14s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c035a3855da7431cb98f28f0cdf7727e : SUCCESS in 4m 12s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ce772004b9f14a80a3bc2d654d92e577 : SUCCESS in 11m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2e8a26fcd6c64cf9912503a8e5fb1016 : SUCCESS in 6m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/90845dce566c44c097fd51db4b345494 : SUCCESS in 6m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b5b89a824a3e47a1866909a67eeb469f : SUCCESS in 7m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/89b601598fb74b2eabe302737ac85dc8 : SUCCESS in 12m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a91e951f78474ea2ab13142e52df1723 : SUCCESS in 9m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/06584683aee64f088d2a96b97f16c09a : SUCCESS in 6m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2f89e758bcec46028d69f04592ffc756 : SUCCESS in 7m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f0ad7bf47fd245c785cbcdd23b555d13 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f0fa3313d344450c918fa8cd004d775b : SUCCESS in 10m 01s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0598601713eb4d7389babe86b5f722a6 : SUCCESS in 3m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bb79c0d2ad014299b2204c11ddfbe1ba : SUCCESS in 5m 03s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/616608ec798c465790d9909e52c14ca8 : SUCCESS in 11m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1f6307d060974a04b9e2f3db8a8aef0a : SUCCESS in 20m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4f60ca26e3c9469f9ac5959897d1de51 : SUCCESS in 7m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/860f4488140f4c8fba217fbd54843b26 : SUCCESS in 13m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7a5b651f9b64481795c7522c888823eb : SUCCESS in 2m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3f1dc05cdda34fa18c5fb075768529da : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2752b2b9b25344d8a0516130c5cc5a37 : SUCCESS in 26m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3e2474bb494844adbb417e45d4199e3e : SUCCESS in 4m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/08b855612149477aa883dc3cb7ee89a8 : SUCCESS in 3m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48320430dc1b4afa92348c20af9e4c95 : SUCCESS in 5m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3abe0f255e3a40f3995de607910459ce : SUCCESS in 19m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cf8dd68ed354cb2bea29b1dea210171 : SUCCESS in 38m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e1fc68ab55741f891070ad6da589349 : SUCCESS in 1h 45m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d9bc3ccd4166a0540bdbbe9335f569546d8a99ed","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-12 07:30:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dd0d991f54bb89cfffbc253ba59dcd195af4c4ab","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-05-12 07:51:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f484f71e3f79ff2e530facb868eb5e97dc880567","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-14 14:30:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"88b3d760374083b3bd1386cea81ba67e57fbc75b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 15:49:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bcdb4857a4549a4918a23bae13fe11f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4bdd08ef79b452eaf4f429e590d88b5 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95449dd491a84a2299a1984223f0379a : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6574d7c4ad04db8887d4aa1eae0460e : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4a775d588cf4fa39f08a8c04a3ae64a : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86c10da95087435d9f05058fe80c2ddc : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8717d3aa51b64efaac2c1df1959bd165 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c399c207714841d89085fb0f010a2640 : SUCCESS in 3m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7aa8a47917544a5780f686d0889f40eb : SUCCESS in 16m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3a71efa37186428482feb925cb0d3dd4 : SUCCESS in 9m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b06547a35f4b4d85941e4be9ddb752a0 : SUCCESS in 17m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/074fb4ad5dba4b4a89fcb5580f226e98 : SUCCESS in 9m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/706267b05e4e4fdeba280e786cf70c96 : SUCCESS in 16m 58s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5783e562921f4516b7bf1074bd26bdd6 : SUCCESS in 6m 10s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/79425abc0dc34561ac729ae853a16100 : SUCCESS in 5m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2f08ee2fb71847e78376ae56ce27862c : SUCCESS in 5m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/213ecb895d8049dcbde6ec6b08d23d14 : SUCCESS in 8m 11s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7ec9b4693e8042cabe49776b6de91001 : SUCCESS in 4m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/46832c7ddc6b4bbe80515d3a92094b27 : SUCCESS in 7m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0e704628f203496792ba89b44b1cc4e2 : SUCCESS in 6m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1ba74e1174c1427da5d869cc421406fa : SUCCESS in 14m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d7154eed3174247891123bd74c9b978 : SUCCESS in 6m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8b07d65ea6ba4c828b526ad145cd2578 : SUCCESS in 5m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/04c94c7806cd47e4b5e9be7b543b482f : SUCCESS in 13m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6f11c066c97b4f518d01013ca472d7a4 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d335400e54974683a0d87634c00f4ae3 : SUCCESS in 10m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/263967c0b9e24f518dc46f4cfa630690 : SUCCESS in 2m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/64705db46b354f198cbfd7396c1fca02 : SUCCESS in 5m 03s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/965134e769204a93b95c27b91d34c157 : SUCCESS in 9m 38s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4aa290cc5e5340cc83bcebb3131c44dc : SUCCESS in 16m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/78ecdfd28a174f86bf1a28cad7066ea1 : SUCCESS in 7m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/28a926ec21db48788e6210dc0c253851 : SUCCESS in 8m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/86de9771d06843aeba3ef825bef1e896 : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d52319c9fb1b482f8bd3a06515b039ab : SUCCESS in 2m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d35bb72456bf49b89bfba9ca4238babf : SUCCESS in 23m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1933532175ed45e0bdea9b9e8bbebaec : SUCCESS in 6m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/07e3d7cb6a644dc2bf1ac775bc313c06 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4af7a635ca4341aaa33580a76d11e360 : SUCCESS in 5m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/da94d445fad445de98086de967812589 : SUCCESS in 12m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/14ab791a4fc544df8e3c61286026e4ab : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b20c7422ca747c69c5a84601955accb : SUCCESS in 1h 02m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d8e5fb48e7c4390b5cd4d70834c565a : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":2},{"id":"69963f1152988c264425c0a7ea8fa60344a67fc6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-14 02:52:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7492921e709b14ffa22457c0ee6e65e97a63b8a7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-14 03:59:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"07215b43986ca43130dcd0eaf69b6b3b992199d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-27 05:12:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9d6341a33f9cee0f005c57d07c40b5ec0cdd0fdd","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-27 05:12:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9d0e427ab48ab03f107906136a42a46dcdd62adf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 06:45:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e91fc24a9ccc41faaa3df9e9c89a85de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe48225e10140f39268fd212b1a9527 : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a729d1a9461e4b0b9938de57076ada2a : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bea0373c6576422e818416c7962472be : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95b4b34b88604a3a9dab50d71749b3db : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79a4ff1a6fb24ae7bf0af8e8a18bdf80 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e611d1aa8b61494f8f2c5e33f767167e : SUCCESS in 4m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/94df1392b95e435bb2412705d9b05684 : SUCCESS in 5m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/566e7d5eea7c44c7b58e5fcb13f55622 : SUCCESS in 9m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/01316874a80e468da59621709009275b : SUCCESS in 5m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2c57c29ac1954269bb73ada2d22b67b0 : SUCCESS in 8m 03s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ef19b6a0040340e593b42aaea5cf44a4 : SUCCESS in 3m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d2950de53dfe4c1a9267f325199c3eae : SUCCESS in 8m 20s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b941c378e36e452bb4643563d389dcfb : SUCCESS in 4m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ac56f60191f74431b7db7730b384a6df : SUCCESS in 10m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/285b0eb768434c04a31fcf3aaf153340 : SUCCESS in 5m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/af68b108d0ae4a52b3605e56a9a7b983 : SUCCESS in 3m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/62bd4cf6ddb24909bb24f79fc79aeccd : SUCCESS in 9m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6b547a537b404b83ac10677571cf7369 : SUCCESS in 16m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8577b226069b40bba9ef29fcf4963c9f : SUCCESS in 7m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/22081e4482564cfdb22a44a9a2445a32 : SUCCESS in 9m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e5b89c68be024ac480bb7154b40f8cab : SUCCESS in 12m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4cf3d82cb5e44a698982784aebe742e8 : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b3c4e08baaed4004a5e14eba13c34bee : SUCCESS in 9m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c091db76e70442eda009f7b443dbaf4e : SUCCESS in 4m 55s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3936d574af4f4894b2bac72686ffc911 : SUCCESS in 4m 58s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d8e1288f08164023a79a26d26820e23c : SUCCESS in 8m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a95fb63c7ffa4745a60c5f103a17ee58 : SUCCESS in 20m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b8f674a5ad7e435ca69b56c6a4b92604 : SUCCESS in 4m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5dc8dd365f5646089581e7555becf7eb : SUCCESS in 12m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/802c3a1b409748ebb867c67023c43500 : SUCCESS in 8m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7943d61f0d9542829ef6aad8ebf34a71 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/147dcd553318400eaabfb16bf87eac92 : SUCCESS in 26m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c08b910831b1432f94033f554b08e14f : SUCCESS in 6m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/40924ca950de4cefa4adf2a07af97d77 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74e6a15e370c4683966c6901b4b020e0 : SUCCESS in 6m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/10edc659b0654fe6ab6d0931f67283c3 : SUCCESS in 19m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c0794202ee53490b8c99df5274352506 : SUCCESS in 5m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/036a680fc4704eab8dde769f1cfb5f7e : SUCCESS in 35m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/465febce13034248b638cb370bba1be5 : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":3},{"id":"f9fa797c64ae0d06c545818e52eeba9d387f2819","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-27 15:02:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0f733052740e9470eca2d2b88612e2dc25282e99","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-10-27 15:04:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f0eb88af2a1c440ead9ba51af4d31766501450aa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-27 15:10:48.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"93b811db5ea0a1bdd4ff9bad05d05b9a7433e497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:11:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b111494a303bb335de92db381053e32c7c5240b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:01:10.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52474d34520746bba532c77bb84ebce4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c8f179368664c56b314ca96416c419d : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9431018ae44902828d5d2cd5dffe60 : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cd73d8ec77041e48e76f5c9104b0cf4 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c41729d7b04549bcabaf2d4412f024f8 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04eda081a2d34d20b1b5f5bdf68fa97c : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22384340c0b74662a832ec159d255d1a : SUCCESS in 2m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/54883132c7e644bb9eb030e4ec9c698b : SUCCESS in 6m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7e8b549188ab40879f8dadea700c1a7a : SUCCESS in 7m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/87ed32d96e8b4a1eba7b6404b91a83d1 : SUCCESS in 7m 27s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bbb912171f5249208de979609d2101ca : SUCCESS in 11m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/55f40d91934b4c9baf656d78edf874bd : SUCCESS in 5m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e241df3bc05f45968e0d31c51ce1c677 : SUCCESS in 7m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2306bfa7f78049059b7af719453388ee : SUCCESS in 8m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d3dfbeced38b4ed8ad66ffaf55b49388 : SUCCESS in 13m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16698c0498ee48c88cb13db748751fe5 : SUCCESS in 7m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1d89d93ba6974779a40c716cd20f9498 : SUCCESS in 10m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9400b9a3c3ae40c1b20904f58303c2d3 : SUCCESS in 16m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c3a1d836d3b340a4a469c7f711aeb5fa : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3235cd4e616d4fb5bdbec54a5de679dc : SUCCESS in 10m 45s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c7223e9dee63488ea8b71fbae6c699fa : FAILURE in 33m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1cda908503e54cb48cc011834543634d : SUCCESS in 4m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b63f3f8df4ee4f9287ee8618e4692602 : SUCCESS in 15m 16s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/dad90d19afe14c018463408b9c0dfe86 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/73ef7faa07dd49d6923db7a8c8127e4b : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a8dd94ad9334235b850fbcf8837ff92 : SUCCESS in 24m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a2f47c7cab634f1791f717ba30823426 : SUCCESS in 6m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f840c645e70d49948eee7fc2b36dee54 : FAILURE in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e00884be9d864643a83322eba1e38efc : SUCCESS in 5m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c3328d8c78334e85ab00729de4438e2b : SUCCESS in 13m 40s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/baaef2a80a2c4519940bab6ece19d5e3 : SUCCESS in 3m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b92004f4e2b14b0684444d555cd25921 : SUCCESS in 44m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e061057b684467cb8544aed35edb6a8 : SUCCESS in 1h 45m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c8ddba58cf4026145ae5b2ce7388d04506b03fe6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-28 01:54:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6d2d69f492496345cedc36699799fc678fea945e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 04:03:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d93b0d535b6433cb934e20073c84f39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11f08a1c1f04fcf9eb855307357ca88 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6e030934334454088b3c462da69fe3d : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cb85f09dae445d5af50cf07dd9be9dd : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24786f1d695346048eee714b3a2c8912 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d20856be6d8f44c78914cb425180e308 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ff5b964a3094e419af4a34077e7a56a : SUCCESS in 2m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/29020996cbe14bb69a9f88c360bf55e2 : SUCCESS in 6m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cf01524d7eff44a7a43f6f26f154ce2c : SUCCESS in 7m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cb717fd2a6134cd7b1486ac414d4bbdb : SUCCESS in 7m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3f38bd38afb74160a5783f7e47c8f19f : SUCCESS in 10m 35s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1c4ea7643c74453794fa9c7e097d6c35 : SUCCESS in 5m 11s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/086aa01447674a69b70e3871dc484246 : SUCCESS in 6m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/571f221b24814af1b560eff235155da2 : SUCCESS in 5m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b29ce6bd8674425ea911335d0dcf5c27 : SUCCESS in 8m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8f39a0398f484c67aba7680dc7bf10a4 : SUCCESS in 4m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/661eccbbcab04013a82839d4908d5f52 : SUCCESS in 4m 24s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d6c879006eb94c238f1f1e94902b20f8 : SUCCESS in 8m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/aaaacacf9f7444718aa1e1ebad5cee66 : SUCCESS in 16m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d5cb810b476b4985afd6089bfcffed77 : SUCCESS in 6m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ff3c2ad388104c2e997b9a9e60395462 : SUCCESS in 10m 36s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ed3da9d0a41d4bcaaa582445c9fd0d11 : SUCCESS in 10m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2e446102636b4e26908cb457227e1a63 : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cfa98a2f14fc42dfb2e19f4022b90497 : SUCCESS in 6m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0ef24444dfb14921ac04394a73dc40dc : SUCCESS in 5m 04s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/42737b8a05e04349bde313a51795439f : SUCCESS in 5m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7fe83beb12f845348239d3f15e8946df : SUCCESS in 8m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4a0bd3bc892c40828aaca02d6c8a89f0 : SUCCESS in 20m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/18c5b225f855427da7ea8f453698981e : SUCCESS in 7m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/40462906667f496db5d9d24adb353262 : SUCCESS in 14m 03s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9648b120e4c54f359b3a6a0b6d265f65 : SUCCESS in 6m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/027a3f76ea774136a3753768db743b3d : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03ee1fdfb01b40258f5c5fe66539eacd : SUCCESS in 23m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c26d1e6748014871bdf9a060fb35e926 : SUCCESS in 9m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/def7396b0f17458a86a4c027d0c61dcf : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0ccff20a13a14493b12156f75b8d38c1 : SUCCESS in 5m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dc062746c11d4bf3b96d64a6800198b8 : SUCCESS in 13m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3a54c011fd8841929dcfe78948e4db06 : SUCCESS in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bc00c877109449ea385f30edfef8b9f : SUCCESS in 1h 02m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aaeef7fc55da4c28861a5745b9f9cdaa : SUCCESS in 2h 02m 51s","accounts_in_message":[],"_revision_number":3},{"id":"2fca7b2b8f1fd9c13b72257de1790d83515b19af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 04:03:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"efcca788c0766e2ceb47a8e926dbf13e4a8c29f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 06:00:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e0e3d8503004b1181f985e726264448\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd3e17f8794441882165626261a881c : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1108aa498c134790832affb5ddadc12d : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7e925295b41428485c9ae1e401c03a4 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d876b58ff9c4aee8c744ccca13f1016 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1096b391b4fc4dc1a9fc813b732db4f8 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6038203259b4825ad685b2d67ffe516 : SUCCESS in 3m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6ea18b860d784875aaf2e8e1ae61a37f : SUCCESS in 4m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/84abcfbf60fa4b9ebc4f49545924e6f7 : SUCCESS in 11m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c2222bd9b1d344b8a39cb21989d5dee2 : SUCCESS in 3m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/192e65c20353436a8c7c1be7dc1316dd : SUCCESS in 11m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b37e6b0c38654c26aeff8d26fe198d4d : SUCCESS in 3m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/da883fa615554abfbf9a889d5646736a : SUCCESS in 5m 05s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a1d84be692e9416dae8b10709dade469 : SUCCESS in 4m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bc5806594c5b44f7acd167a3d4c762c0 : SUCCESS in 18m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fcb40bddf5e042df9c073b70ad2acd70 : SUCCESS in 7m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/57b8f271e11f4830835bed6363170fde : SUCCESS in 9m 23s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/79c5a09f00f749a1bb543beab5614097 : SUCCESS in 13m 07s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/12f0d8fbbc254e1ba18ee06db4b46da2 : SUCCESS in 1m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fc36aa218d0a41fd8643c76c70019f72 : SUCCESS in 11m 21s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/91c7440415b04339a6a6122dae60b030 : SUCCESS in 30m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/80b736e299cd4b839c7f64a3e15854ee : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/77dd6e6455794944ae55934ddd3a51c0 : SUCCESS in 7m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/63f406b7ecdd4672b9969f57e4ec1770 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac0e9eca5e4e44ac81f99050206c76d9 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07dcedd3dd9c40779053005352a8b78b : SUCCESS in 28m 04s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7db94e3a5eab42a8872d6ceba04ab404 : SUCCESS in 7m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/de4c02f47ef64be29c346e8d2c2f5eea : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf59d4855deb49419e5919dfb2d646ac : SUCCESS in 6m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/552dbb20554e486880318ca650bcf5c3 : SUCCESS in 12m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d404720ee785465e8e77b0ccc6da5ca2 : SUCCESS in 5m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/def4dfa957c94dab991a34fa94c71265 : SUCCESS in 48m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51b1fc835ac347b1bfcc54283263f841 : SUCCESS in 1h 45m 56s","accounts_in_message":[],"_revision_number":3},{"id":"13c9a19ecfc5653d3f043c7e5450f3cd29a42320","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 06:00:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ae7c60bc107fbb567ad3006b3f223d5e22ecbdf6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 06:01:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00d9635a1a2f47f9b0db424ca865b02e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28fe0c8daf6a4c3ead3147b4437e5e77 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ced14b82274bf0faa87e4f5a2362c389a8ce2cea","revisions":{"5d045324e698a3fae9636af151cf45983e0162be":{"kind":"REWORK","_number":1,"created":"2025-05-09 13:20:19.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/42/949342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/949342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/949342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/949342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/949342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/949342/1"}}},"commit":{"parents":[{"commit":"eb3441eb910af54ae86e9b9211a8e1f625e9c179","subject":"Merge \"update constraint for oslo.middleware to new release 6.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb3441eb910af54ae86e9b9211a8e1f625e9c179"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-05-09 13:18:17.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-05-09 13:19:43.000000000","tz":120},"subject":"Release cap for bcrypt","message":"Release cap for bcrypt\n\nIt was necessary to cap version of bcrypt while Keystone was getting rid\nof this unmaintained dependency. It was completed in last cycle and it\nis time to remove the limitation of the bcrypt version.\n\nChange-Id: Iefcdbc6e050dfe13f54097e665f9a91dec02a63a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d045324e698a3fae9636af151cf45983e0162be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d045324e698a3fae9636af151cf45983e0162be"}]},"branch":"refs/heads/master"},"a39012e9efb0930a18be1c2173bd417af7158506":{"kind":"REWORK","_number":2,"created":"2025-07-14 14:30:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/949342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/949342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/949342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/949342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/949342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/949342/2"}}},"commit":{"parents":[{"commit":"1f2be26ab5a4d601127817b90c794e61cc40bd12","subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f2be26ab5a4d601127817b90c794e61cc40bd12"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-05-09 13:18:17.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-14 14:30:18.000000000","tz":540},"subject":"Release cap for bcrypt","message":"Release cap for bcrypt\n\nIt was necessary to cap version of bcrypt while Keystone was getting rid\nof this unmaintained dependency. It was completed in last cycle and it\nis time to remove the limitation of the bcrypt version.\n\nChange-Id: Iefcdbc6e050dfe13f54097e665f9a91dec02a63a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a39012e9efb0930a18be1c2173bd417af7158506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a39012e9efb0930a18be1c2173bd417af7158506"}]},"branch":"refs/heads/master"},"ced14b82274bf0faa87e4f5a2362c389a8ce2cea":{"kind":"REWORK","_number":3,"created":"2025-10-27 05:12:03.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/42/949342/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/949342/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/949342/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/949342/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/949342/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/949342/3"}}},"commit":{"parents":[{"commit":"6759ad0f43ed1cdd3e30e01c98ba756ac4589559","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6759ad0f43ed1cdd3e30e01c98ba756ac4589559"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-05-09 13:18:17.000000000","tz":120},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-27 05:11:47.000000000","tz":660},"subject":"Release cap for bcrypt","message":"Release cap for bcrypt\n\nIt was necessary to cap version of bcrypt while Keystone was getting rid\nof this unmaintained dependency. It was completed in last cycle and it\nis time to remove the limitation of the bcrypt version.\n\nChange-Id: Iefcdbc6e050dfe13f54097e665f9a91dec02a63a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ced14b82274bf0faa87e4f5a2362c389a8ce2cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ced14b82274bf0faa87e4f5a2362c389a8ce2cea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
