)]}'
{"id":"openstack%2Frequirements~949862","triplet_id":"openstack%2Frequirements~master~Ifdc3a4203bf706b479801c2da27446e5dc31f5f5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifdc3a4203bf706b479801c2da27446e5dc31f5f5","subject":"update constraint for python-ironicclient to new release 5.11.0","status":"MERGED","created":"2025-05-15 10:04:07.000000000","updated":"2025-05-15 14:29:46.000000000","submitted":"2025-05-15 14:28:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949862-new-release","meta_rev_id":"105ad3d8d7ff560dc3ed00c29bdecf00d130e839","_number":949862,"virtual_id_number":949862,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-15 14:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-15 12:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-15 12:17:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 11:17:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-15 12:17:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"746abb219948fd6827546ee57139c1992e40ad70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-15 10:04:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0b2ceb0a6232279ca98b127f500385e8e9baa53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 11:17:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c15ccd6fe61f470fb85378bd7bf8ca0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef466633a2d4883b1ba5dea705a8fa6 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add6480c5ea0416dac664a916d8f7d15 : SUCCESS in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86a641b56ba6482187863a0ede6e353b : SUCCESS in 2m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64f705ebdffd4401b459a9f9b4703a56 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f264a08d38f24ecb82eb6a0f52506c74 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94f84a3df9cf492889235c53a59bb4f6 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/adefbff866484de7b8b96f6e77183aa1 : SUCCESS in 2m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/829008a62e46491fa7364e85339663be : SUCCESS in 20m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dde6362db51d4bc9a9ff6845471073b9 : SUCCESS in 8m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/82bee132964c482fae91bb1c88188b45 : SUCCESS in 20m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8ef7537e1129413e85111aa34b811f84 : SUCCESS in 11m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ad403701ba3c4ceeabe34e625cec7caa : SUCCESS in 22m 05s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/038a1634deca42eb8b6847826792057a : SUCCESS in 4m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/27f73072e2034fe9a7c3adcc27900846 : SUCCESS in 5m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/38813bed2f144aad970d0a4a0875cc82 : SUCCESS in 3m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0d3f2171e89d4907a9c3dfba1ff81759 : SUCCESS in 10m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fd9ac31a28d14f3c87103edb7eb21b16 : SUCCESS in 5m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/491fb7eec7354e10b80752761a3a978f : SUCCESS in 5m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ca406dadc3e24d2e812108c8549960f4 : SUCCESS in 9m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f48c3f3d3daf47308416ff6796f9d7e7 : SUCCESS in 14m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fafeec5e192542a09a2e93966dc4d730 : SUCCESS in 6m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b97d93c358c244a69cbddafac04f7b81 : SUCCESS in 11m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/effce90d0e5443d8a8e99e9c0e5888a6 : SUCCESS in 10m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b70e6bdb400b45d6b510edb7068c5ed3 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3274d9fdd58540ab861319fa63029c70 : SUCCESS in 11m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b7b29008a26446fca5a4820c4a47792a : SUCCESS in 6m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b846d36e9d2a4659b64e75126263ba6c : SUCCESS in 5m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2418138d2bae437881141ef1fcc23bb1 : SUCCESS in 10m 19s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/df9c9232d7ef4763a43bd859c6c6fa35 : SUCCESS in 16m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6cc2099fd987409caf5b9a96aa5b73ab : SUCCESS in 6m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/99d219c030ac420a93d5f10a38a7f61f : SUCCESS in 8m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/be88cc59fc8741ea9dbd5d8a35f14c7b : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dbed8e2dbcc54f75b5553f3d7b811efc : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e28f631ef5644502afabab522e1c0a90 : SUCCESS in 24m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5f1623716d5c4fa6be4353cb2dbbf8a4 : SUCCESS in 6m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5849bbb6fc204e4a9da75bde73cbfd59 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f56bf2fceee54f72bbe6444d7ebdc98b : SUCCESS in 5m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2491d90681c34bc299ef350042e42e06 : SUCCESS in 6m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29a3a3b8e14740259b324d06343d325c : SUCCESS in 54m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb0f88a3e001469eb7919c2a2b515adf : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2e367307de5f1d702c9886006d22404297af9413","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-15 12:17:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae11d80cec1c2cb8cdc5d99542ff3352d9676c06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 12:18:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb5e0adf7b7d1db4d5b743a545c435b11cc0668e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 14:28:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9e7cf002a3446fa87ebf24fc1345db1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67ff76d4465d43249353df916c8f3022 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517d0a2d363b44d685f42bbbc8511f57 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2aba6d829c7247469d3f9edd79abec3b : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f9218a1e0e54ba1bd3ebc41224ced20 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d731e29b85b46409e9a25aad9ac3b2d : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5c60a0f30f34618b1590d5930ed05e9 : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8be3d704d12a4ff19b748f85a72e304c : SUCCESS in 15m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b23d909bb29e41cc8c4e268fe7c7ba60 : SUCCESS in 5m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/00848afd0cc44004b2ba82cdb74b1c09 : SUCCESS in 7m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b9f2e628c69447d387287759e96e73b1 : SUCCESS in 20m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/508ece9f28374ec9ae7080550f2340ad : SUCCESS in 11m 03s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8b1c28b276f3412fa87887d0e581e0d8 : SUCCESS in 6m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/033ea1c2d6604b70bd9df6d86d962221 : SUCCESS in 3m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4d1c46d8c43341c4ab3ae4fe3afa464b : SUCCESS in 8m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f6338e78089a44138cf8a885656f6812 : SUCCESS in 14m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8fa0a6965f5948fe97aa9b3b435fa9a6 : SUCCESS in 5m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/33294310fc88485180a3d89117235115 : SUCCESS in 7m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1376e8b5098a4dcc888382c8c4d85447 : SUCCESS in 12m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1a9b313ece474b22b70dec79b87fdd45 : SUCCESS in 1m 36s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1b3e75191a414ab1bf3de9ff91a4047f : SUCCESS in 11m 56s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c22eb8bf76034e09881fb473d5f08b23 : SUCCESS in 17m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/684857c5f3e34e91801f9ddc626790b4 : SUCCESS in 7m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5efa51cb99244c2bb28f18b8abed5d5d : SUCCESS in 12m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/02da8af971ba4ea4b242ac8c43ad816c : SUCCESS in 6m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/87a1c6f3a8d743eea74b5cacac57f636 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3209a0f1619146889e9d5336a77b033c : SUCCESS in 22m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8ddb8b8a82d14c73abb6e0116971fe57 : SUCCESS in 5m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1302632e35f446fc98612ad112b7b18d : SUCCESS in 3m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61fdc184a2d341aebb1f2d8e6d2d4ad2 : SUCCESS in 3m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/06b80dfa8a9c4bf2bf2e7e02f4c3e1df : SUCCESS in 20m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be7552d4951341d2b1f314d55b461cc6 : SUCCESS in 59m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2738c74a31d4571b05234d218c1e894 : SUCCESS in 1h 38m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6a7d8043315eb624a40be31262c7df9344c3df15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 14:28:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"105ad3d8d7ff560dc3ed00c29bdecf00d130e839","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 14:29:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58bf8fcd4b0444b99c92662e0da2e7dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495e68835a8843799a8702688482d193 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e1562ba8863f7a8e194e4da3841854d90d4956c","revisions":{"6e1562ba8863f7a8e194e4da3841854d90d4956c":{"kind":"REWORK","_number":1,"created":"2025-05-15 10:04:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/949862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/949862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/949862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/949862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/949862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/949862/1"}}},"commit":{"parents":[{"commit":"ead5099bd7377d98d809363af2f4b619f54ce21a","subject":"Merge \"update constraint for os-traits to new release 3.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ead5099bd7377d98d809363af2f4b619f54ce21a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-15 09:54:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-15 09:54:06.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 5.11.0","message":"update constraint for python-ironicclient to new release 5.11.0\n\nmeta: version: 5.11.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: I3d150a0ba95db3349eaca55ae11906508cf1c26e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nChange-Id: Ifdc3a4203bf706b479801c2da27446e5dc31f5f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e1562ba8863f7a8e194e4da3841854d90d4956c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e1562ba8863f7a8e194e4da3841854d90d4956c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
